)]}'
{"id":"openstack%2Fvitrage~726723","triplet_id":"openstack%2Fvitrage~master~I2f92b6409593733f1de628f74a758160bc65efe0","project":"openstack/vitrage","branch":"master","topic":"bp/doc-upgrade","hashtags":[],"change_id":"I2f92b6409593733f1de628f74a758160bc65efe0","subject":"Update doc: Using version 3 to update templates-loading.rst","status":"MERGED","created":"2020-05-11 03:14:22.000000000","updated":"2020-05-17 16:05:06.000000000","submitted":"2020-05-17 16:03:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726723-1589731430434-d262c4b1","meta_rev_id":"9901fb322d51cdfb70ccb03b3c7dcecdcbdbcd3d","_number":726723,"virtual_id_number":726723,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-17 16:03:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-13 18:34:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-17 15:55:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-17 15:55:03.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-05-17 16:03:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7d806a4c426b0176a5764b43cbfe1e5521e0ef2","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-05-11 03:14:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fde10721e51463a53ebab16b8998a7ef548e52b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 04:16:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84054891c96d4170acc06c41288c6f7e : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a7d38946de645e0ae78fba524d30418 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7aa5a0199c7e4d7d8603256318f869fb : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efcdc39516384feaa9d00fbc01a6039c : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a64aa2f77cf4522aacaeb5b6697bca6 : SUCCESS in 4m 35s\n- vitrage-dsvm-api-py3 https://zuul.opendev.org/t/openstack/build/c3543df2964747a397e879b22dd9ffa9 : SUCCESS in 50m 09s\n- vitrage-dsvm-datasources-py3 https://zuul.opendev.org/t/openstack/build/77e10112cd2e437f89fea8d1685af3c5 : SUCCESS in 1h 01m 01s\n- vitrage-dsvm-mock-py3 https://zuul.opendev.org/t/openstack/build/681527d7ca4c482eabf6b44b342e700a : SUCCESS in 50m 28s (non-voting)\n- grenade-vitrage https://zuul.opendev.org/t/openstack/build/f2bf7d76c2ac4d418610f5647007bac8 : SUCCESS in 47m 28s\n- vitrage-dsvm-api-ipv6-only https://zuul.opendev.org/t/openstack/build/7502216d166941bab586934e141bd808 : SUCCESS in 51m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f6081441c52cc50185ebeabd008abcdc804dd65f","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-05-13 04:55:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"69949d822e2a9ebbbb7a9d3c2480aa90b1d7c848","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 04:56:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02de1b01aab463becc643d4b00438b21ccd217cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 06:13:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/113541bdaa84456aa8499b319f498af5 : FAILURE in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/548e962a8d5040eda3df990cbc6bfee4 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15995fadd8f84f6786d50732067db52d : FAILURE in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e184162e2cd94a798eb3c449d0ed321e : FAILURE in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b8b7f4aab6b4f09b0e4a907f54b4c53 : SUCCESS in 4m 29s\n- vitrage-dsvm-api-py3 https://zuul.opendev.org/t/openstack/build/e1a566d909694fcc83e7c798c8eee4fe : SUCCESS in 1h 01m 24s\n- vitrage-dsvm-datasources-py3 https://zuul.opendev.org/t/openstack/build/c4087e8f197e45ceba22fe94a843f65a : SUCCESS in 1h 16m 09s\n- grenade-vitrage https://zuul.opendev.org/t/openstack/build/35899ec848bc47c49fb34619e2e90d44 : SUCCESS in 38m 59s\n- vitrage-dsvm-api-ipv6-only https://zuul.opendev.org/t/openstack/build/0db5c64321354104bcd3709c7e329420 : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a535ef0037a9ae5168ef4f08fe2a3ecce67f99fc","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-05-13 09:15:05.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fcb3fe0db254cf99a8c51f4a18a4ffeb7ed29714","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-05-13 09:54:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5b92a832bb2d60cc071b712ae26fcbdef51b8bce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 10:06:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fda9cba026445caab9a944b3ee1c7ec : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91c09588b5754daabac77908d3b0ba54 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":3},{"id":"7c369a469fed1058d43b2c39a6e5c0248ad6cf97","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-05-13 18:34:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1d19c98b3d369f77b427b47e9fa7571865947d91","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-05-17 15:55:03.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"109be5139f059c20fdff0a0bbfe08dbfc1e53e7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 15:55:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4d6827c80881c74da8c23f9fbdcc4d1aef3f55c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 16:03:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"77ba3fa0f7c7b83b52f2d7ae8071d458dc8c9d80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 16:03:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c52d48e338034d13abfc48887537739a : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c211abc335ae4ab19a1c02b7908da67e : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":3},{"id":"1ec85ad764e4562d998e43273599e49232b0b667","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 16:05:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e68b8b4b8284836b7ebb8a2fee934d3 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"856a64f25130234e9d929c49bdd643b550b25820","revisions":{"0eff6bbaae18270dfb11c4e3a313903303a8a3c4":{"kind":"REWORK","_number":1,"created":"2020-05-11 03:14:22.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/23/726723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/23/726723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/23/726723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/23/726723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/23/726723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/23/726723/1"}}},"commit":{"parents":[{"commit":"2856f88fcfbb26ab007ab0f7c31b54c50ff7faa7","subject":"Make sure depth is an int in get topology","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/2856f88fcfbb26ab007ab0f7c31b54c50ff7faa7"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-05-11 03:11:24.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-05-11 03:11:41.000000000","tz":480},"subject":"Update doc: Using version 3 to update templates-loading.rst","message":"Update doc: Using version 3 to update templates-loading.rst\n\nNow part of the document is version 2 syntax, and part of it is version 3.\nSo the document needs to be updated to Syntax 3 to replace Syntax 2.\n\nImplements: blueprint doc-upgrade\nChange-Id: I2f92b6409593733f1de628f74a758160bc65efe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/0eff6bbaae18270dfb11c4e3a313903303a8a3c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/0eff6bbaae18270dfb11c4e3a313903303a8a3c4"}]},"branch":"refs/heads/master"},"8b107e152a7ff42a8ffc2474871f1ad9588d0618":{"kind":"REWORK","_number":2,"created":"2020-05-13 09:15:05.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/23/726723/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/23/726723/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/23/726723/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/23/726723/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/23/726723/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/23/726723/2"}}},"commit":{"parents":[{"commit":"2856f88fcfbb26ab007ab0f7c31b54c50ff7faa7","subject":"Make sure depth is an int in get topology","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/2856f88fcfbb26ab007ab0f7c31b54c50ff7faa7"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-05-11 03:11:24.000000000","tz":480},"committer":{"name":"Qitao","email":"qihongtao@inspur.com","date":"2020-05-13 09:14:43.000000000","tz":0},"subject":"Update doc: Using version 3 to update templates-loading.rst","message":"Update doc: Using version 3 to update templates-loading.rst\n\nNow part of the document is version 2 syntax, and part of it is version 3.\nSo the document needs to be updated to Syntax 3 to replace Syntax 2.\n\nImplements: blueprint doc-upgrade\nChange-Id: I2f92b6409593733f1de628f74a758160bc65efe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/8b107e152a7ff42a8ffc2474871f1ad9588d0618"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/8b107e152a7ff42a8ffc2474871f1ad9588d0618"}]},"branch":"refs/heads/master"},"856a64f25130234e9d929c49bdd643b550b25820":{"kind":"REWORK","_number":3,"created":"2020-05-13 09:54:35.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/23/726723/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/23/726723/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/23/726723/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/23/726723/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/23/726723/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/23/726723/3"}}},"commit":{"parents":[{"commit":"2856f88fcfbb26ab007ab0f7c31b54c50ff7faa7","subject":"Make sure depth is an int in get topology","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/2856f88fcfbb26ab007ab0f7c31b54c50ff7faa7"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-05-11 03:11:24.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-05-13 09:53:10.000000000","tz":480},"subject":"Update doc: Using version 3 to update templates-loading.rst","message":"Update doc: Using version 3 to update templates-loading.rst\n\nNow part of the document is version 2 syntax, and part of it is version 3.\nSo the document needs to be updated to Syntax 3 to replace Syntax 2.\n\nImplements: blueprint doc-upgrade\nChange-Id: I2f92b6409593733f1de628f74a758160bc65efe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/856a64f25130234e9d929c49bdd643b550b25820"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/856a64f25130234e9d929c49bdd643b550b25820"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
