)]}'
{"id":"openstack%2Fvitrage~750114","triplet_id":"openstack%2Fvitrage~master~I52df6f16e73cb87f5af9c15af25de3afa2b5dafd","project":"openstack/vitrage","branch":"master","topic":"kapacitor","hashtags":[],"change_id":"I52df6f16e73cb87f5af9c15af25de3afa2b5dafd","subject":"add info to kapacitor datasource state","status":"MERGED","created":"2020-09-07 00:56:43.000000000","updated":"2020-09-07 10:12:29.000000000","submitted":"2020-09-07 10:11:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750114-1599473480409-941d6ed0","meta_rev_id":"42b9dbac10fe34fedf3eeb538f4239681953f7e6","_number":750114,"virtual_id_number":750114,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-07 10:11:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-07 08:34:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-07 08:34:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-07 08:34:51.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-09-07 10:11:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"63731c470bf3a6dbabc73cb207f29b910a47e193","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-07 00:56:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5e139e23d3652334a2941e35d483cb7271bed6a","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-07 00:57:50.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fbb6aae3b16394078efa62ed13d60240474d90df","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-07 01:04:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3ba03ff7c15cdaceb68323472b15a29d0e90b968","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-07 01:47:30.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"627cc505f94682ce3c5f7037a5327117676567ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 03:04:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c575e0e0353f4733b2d3f694298e50ec : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6888cb15a99847519ced863c3a1c910c : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d1d482d6c07455eac779c7ed501f75e : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d920c61b111a4666b2669ddf2efa2e0f : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d61181a78f4049788214d504055afaf0 : SUCCESS in 3m 23s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/76e91dca87c442f481fd925bc40defd2 : SUCCESS in 1h 01m 53s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/86383f46be284738af1db824beef2692 : SUCCESS in 57m 20s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/6060dca278f345969f3ed885892fd27f : SUCCESS in 1h 10m 48s\n- vitrage-tempest-plugin-mock https://zuul.opendev.org/t/openstack/build/9cace9b2e2c740f3b07563ffffbd3342 : SUCCESS in 1h 03m 54s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/a9b1a751959d434a8a386dd88d063c99 : SUCCESS in 59m 56s","accounts_in_message":[],"_revision_number":4},{"id":"66c1858bc07df999b66a3c78864d1fa08b16b968","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-09-07 08:34:51.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"69d96724fdc6a8971608abdad1726497db1d30b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 08:35:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"581c35001660c91faff5babfdbf40a473224f4bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 10:11:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"215ad020d152a22ec34d235025f13d5cc786aff2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 10:11:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2cb1197e02e49e3a81bdcc6ef22ed59 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6d82e80394d4f15a9799aee33361170 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1001ed68c13e46bfb489042135b9386f : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13d18ab9e31547568a1583fb95b0f9a6 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09e7839d74784df982916da569018d36 : SUCCESS in 4m 39s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2b1e9e51396142f481ea14e136550a03 : SUCCESS in 53m 54s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/0eab50669878457e968145272e6b8e10 : SUCCESS in 1h 24m 03s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/98d4a3b6a18c4e68beac9902cf622773 : SUCCESS in 1h 03m 19s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/6df3368e24c54ba9bf6e5756dc4f970c : SUCCESS in 1h 34m 31s","accounts_in_message":[],"_revision_number":4},{"id":"859ce63ca7426880d95a96c187c57c2d0b5b74e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 10:12:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c43e11a9992e486ca689d99e53d8507d : SUCCESS in 49s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c0511ee83453987ebf925ca8029301502a52065a","revisions":{"5e89794497ea051b2c45f6c34f79d7641e022ca4":{"kind":"REWORK","_number":1,"created":"2020-09-07 00:56:43.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/14/750114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/14/750114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/14/750114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/14/750114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/14/750114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/14/750114/1"}}},"commit":{"parents":[{"commit":"99682219f31ba67ae7404c4d9de56289fdf08ef7","subject":"Update api: Event POST","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/99682219f31ba67ae7404c4d9de56289fdf08ef7"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-07 00:56:28.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-07 00:56:28.000000000","tz":480},"subject":"add into to kapacitor datasource state","message":"add into to kapacitor datasource state\n\nConsidering the low level of info, so it is placed in the OK state for now.\n\nChange-Id: I52df6f16e73cb87f5af9c15af25de3afa2b5dafd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/5e89794497ea051b2c45f6c34f79d7641e022ca4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/5e89794497ea051b2c45f6c34f79d7641e022ca4"}]},"branch":"refs/heads/master"},"ccc15b6b8228e250d172e6d3b13520a66266c531":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-07 00:57:50.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/14/750114/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/14/750114/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/14/750114/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/14/750114/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/14/750114/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/14/750114/2"}}},"commit":{"parents":[{"commit":"99682219f31ba67ae7404c4d9de56289fdf08ef7","subject":"Update api: Event POST","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/99682219f31ba67ae7404c4d9de56289fdf08ef7"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-07 00:56:28.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-07 00:57:35.000000000","tz":480},"subject":"add into to kapacitor datasource state","message":"add into to kapacitor datasource state\n\nConsidering the low level of info, it is placed in the OK state for now.\n\nChange-Id: I52df6f16e73cb87f5af9c15af25de3afa2b5dafd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/ccc15b6b8228e250d172e6d3b13520a66266c531"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/ccc15b6b8228e250d172e6d3b13520a66266c531"}]},"branch":"refs/heads/master"},"63da56a8fc4ab3dbe2188c61ced032fc9ca36fc8":{"kind":"REWORK","_number":3,"created":"2020-09-07 01:04:09.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/14/750114/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/14/750114/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/14/750114/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/14/750114/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/14/750114/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/14/750114/3"}}},"commit":{"parents":[{"commit":"99682219f31ba67ae7404c4d9de56289fdf08ef7","subject":"Update api: Event POST","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/99682219f31ba67ae7404c4d9de56289fdf08ef7"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-07 00:56:28.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-07 01:01:52.000000000","tz":480},"subject":"add into to kapacitor datasource state","message":"add into to kapacitor datasource state\n\nAlthough info is a very low level, it should still\ncorrespond to the warning state.\n\nChange-Id: I52df6f16e73cb87f5af9c15af25de3afa2b5dafd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/63da56a8fc4ab3dbe2188c61ced032fc9ca36fc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/63da56a8fc4ab3dbe2188c61ced032fc9ca36fc8"}]},"branch":"refs/heads/master"},"c0511ee83453987ebf925ca8029301502a52065a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-09-07 01:47:30.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/14/750114/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/14/750114/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/14/750114/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/14/750114/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/14/750114/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/14/750114/4"}}},"commit":{"parents":[{"commit":"99682219f31ba67ae7404c4d9de56289fdf08ef7","subject":"Update api: Event POST","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/99682219f31ba67ae7404c4d9de56289fdf08ef7"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-07 00:56:28.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-07 01:47:09.000000000","tz":480},"subject":"add info to kapacitor datasource state","message":"add info to kapacitor datasource state\n\nAlthough info is a very low level, it should still\ncorrespond to the warning state.\n\nChange-Id: I52df6f16e73cb87f5af9c15af25de3afa2b5dafd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/c0511ee83453987ebf925ca8029301502a52065a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/c0511ee83453987ebf925ca8029301502a52065a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
