)]}'
{"id":"openstack%2Fvitrage~751258","triplet_id":"openstack%2Fvitrage~master~I2057c901c8e28e1d61c506dcd8790d404817aaa8","project":"openstack/vitrage","branch":"master","topic":"bp/add-cetus-datasource","hashtags":[],"change_id":"I2057c901c8e28e1d61c506dcd8790d404817aaa8","subject":"Add Cetus Datasource","status":"MERGED","created":"2020-09-11 12:17:39.000000000","updated":"2020-09-24 11:53:39.000000000","submitted":"2020-09-24 11:51:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751258-1600948300328-365cc730","meta_rev_id":"bd87bbb2f79dc28cb3abb68d054301a1e2efe18a","_number":751258,"virtual_id_number":751258,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-17 11:12:37.000000000","_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-24 11:51:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-24 06:02:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-24 10:40:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 10:40:34.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-09-24 11:51:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd4b005356e8fe45928f85a868dd3df2d65db611","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-11 12:17:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e41e4212f9c857a47afa19616c26ed3f1be954a","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-11 12:19:00.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"561462f767806a7824be78d3e477089a1cafce69","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-11 12:20:29.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"dc59cce10fcd6a1c4d78d6080285fec824d64270","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-11 12:27:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5921c190d7fe7255f2477202665d09a4bdb6258c","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-11 12:32:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f0f68d7ac719710ccba8453ac4467667bb334050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 13:37:49.000000000","message":"Patch Set 5: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b6faacf091549d8a7cfd5be2b332bac : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02b040645fc6462bbedf816b399f0137 : FAILURE in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bff5df00e484301952d9a10eb1ea6c4 : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/674c0bb3869d46adbc5f289507ea0c73 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/267384c7a1ea4c3da7d7b70fadc88f76 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bb1d8179fe148ed91fcee510b847153 : SUCCESS in 4m 03s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/39c3877ab1c3458688948d94ea39ce8c : POST_FAILURE in 46m 41s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/42e4382e76004213b4e2151cc8e74735 : POST_FAILURE in 50m 22s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/9a0052c2b59e470e974f8349ad64b7f6 : POST_FAILURE in 48m 54s\n- vitrage-tempest-plugin-mock https://zuul.opendev.org/t/openstack/build/c0a122a4a92040558e020f4102cf4ae5 : SUCCESS in 57m 19s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/1f4d0084028e4a5dad4b1f089e13bd29 : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":5},{"id":"ed05d897bba5544c45df1eaa6c15bfb8679c82a0","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-11 14:12:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4ab742e02119f50ed7e084cabcbe1e014b5fd391","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-11 14:30:35.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8f16dc8d2d7cf58e29af00fe42613109e66b7eea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 15:35:02.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c21a3476d234a089e5a036c556feda9 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f13674bddf6a4adfa9a6f6955024fbb1 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1614235bd2114c488ff6b426f4933bd3 : SUCCESS in 8m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb807f39e04f4e43add1c2bf7c421599 : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66c0e21bedec427589d5aa4906e85cc1 : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad6e95e7ba5444beab5d6e575755577e : SUCCESS in 5m 08s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9139d9140caa45988f0c64f619ab83d0 : POST_FAILURE in 45m 54s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/cb1508015c824bdc978357a4d84e5790 : POST_FAILURE in 48m 13s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/4a91278979184df681b88b7193d28a17 : POST_FAILURE in 1h 00m 08s\n- vitrage-tempest-plugin-mock https://zuul.opendev.org/t/openstack/build/26f6fb35cf6c42cba1f4b5045e3bdece : SUCCESS in 51m 25s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/98dfc74e6a604f8a996982683bedbe96 : SUCCESS in 51m 02s","accounts_in_message":[],"_revision_number":7},{"id":"bec09796e40dc1249f32a4085383424ed3fe7ab6","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-11 18:05:52.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3c158d53257d4bce1af3b199db6308a6ad4089a1","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-11 18:23:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"aa90713b0b4532ce8b5252e0d7528df359fe8cb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 19:16:20.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60a49262730a494e8b36349187249ff0 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eea99dec284d4549bbe8cb43ab49b631 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a901820ccd3a41e9bc46310494645c9a : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2936222920224d37983f31fb79d039c4 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/641ff5092d99485a9e3c859a3faf9240 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70d2dafefa2b4dc186f4781310641d8d : SUCCESS in 3m 25s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/25a38ba71b774e5aaace0eace0ddbfd4 : POST_FAILURE in 40m 16s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/e1040d1682114b77bbe43a755ad605fe : POST_FAILURE in 39m 51s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/72e47323170642e1b996e8df801f1323 : POST_FAILURE in 50m 17s\n- vitrage-tempest-plugin-mock https://zuul.opendev.org/t/openstack/build/0d6eba194a26449c89b53188e7bd0f35 : SUCCESS in 48m 44s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/f5f23b37c94e4b96adc38a504cfde896 : SUCCESS in 49m 49s","accounts_in_message":[],"_revision_number":9},{"id":"962d018cb1a0fdc57ae72b2626cae61920b41007","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-12 05:20:33.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"95d71d2ddb6047a303c681f859285c8fc27df584","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-12 06:22:47.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ff2346209314cb197d955966949ffbd : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6070506ec8074202a751052bd8837f2f : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/476f00b17cc64530b0784057cd4428d9 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/204654a3e2a342cfac425feb456dcdbd : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf94c7b57ac54dd5b95e103570161ca4 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ea81f11fade4dd2a78f441a7b9ccd4d : SUCCESS in 3m 12s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/65eba54dca3c44c3b47997218ac808a6 : POST_FAILURE in 49m 22s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/88611ae0db8947a7a034d3d7120d6136 : POST_FAILURE in 44m 29s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/52ddb0e1126d49939f8cb398badd2b02 : POST_FAILURE in 57m 39s\n- vitrage-tempest-plugin-mock https://zuul.opendev.org/t/openstack/build/6285e529a8b14a3a91858326c416a85f : SUCCESS in 57m 16s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/f584b8bd6d2648b6b64d0e7b52181a20 : SUCCESS in 53m 15s","accounts_in_message":[],"_revision_number":10},{"id":"1fa33e7e9c90d895c2f8dc6c23ea3b529db4ebb9","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-17 11:12:37.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"8f4e196af00ecd551542aa4055747aba05d9bc68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 12:23:48.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed2f4771d22440b59afbef7d9e7e1085 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f483a2bb79b94c53be8bf45bfa2a712f : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b0fa47aaa264bfaaa595f723286cff2 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e7409b0a2b54660bf724ba98cd961d0 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/244269774d3a4afab163f8d1c2622450 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa1e9f3100cb46e0905317d183e286bf : SUCCESS in 2m 55s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2afa05eda6f44685a5d235bbd9c11c9f : SUCCESS in 43m 41s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/80d9c5e15ed44059baf098cf99c5e20a : SUCCESS in 46m 42s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/84d9d951b1df4f589ca1bb1ffdbec0d8 : SUCCESS in 55m 59s\n- vitrage-tempest-plugin-mock https://zuul.opendev.org/t/openstack/build/1e04887d7cdc4472825ad96c0477e94f : SUCCESS in 51m 21s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/42abfedb361d4b3c8c284f9c278b35d9 : SUCCESS in 1h 04m 02s","accounts_in_message":[],"_revision_number":10},{"id":"a6a1f07d09570fded68a4248a979c6861fc99e28","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-09-24 06:02:08.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"af2d8443678eec18e4d3edac444d7004b798ffc4","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-09-24 10:40:34.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"52e1797ccf3a1be3eaa4c77a8fac610d500b15be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 10:40:46.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"4c55887e74940a89c8598b0767535759c7490b98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 11:51:40.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/345b547754af4f12ba186d9e13f758d1 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06d4376dea2f442eb063bebabed39a86 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/256996bef3c4420ba21dcd0c9aa395d0 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dfff368e1634bf0b1d7f6c775b92f3f : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73cea0ac191f4629b9e4dffe308141a7 : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3188d429eb64861a910174bc0bdb369 : SUCCESS in 4m 02s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7f736155c53b480899b22cdc3eb86aab : SUCCESS in 52m 32s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/8dde407d8b3c4b4cad2379d6eda19ebc : SUCCESS in 51m 46s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/7aa95123daca4b65b2adaa34f4fa0d68 : SUCCESS in 1h 05m 11s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/8fbe59d300184475a981fa2d2d155876 : SUCCESS in 47m 23s","accounts_in_message":[],"_revision_number":10},{"id":"d5ad8ee9dd1b6103f5f3179e81537c1a3437a96d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 11:51:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"50f17f0ae95e601eef6d7f0ecca891309b7f17de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 11:53:39.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eee3b7b3e0b3461494ec047a4a7eaaf6 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f58fc02b39454b4492494a29ab1028da : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"9638aa34878a1b4ab6d5fe3bf0a6db4a65e63091","revisions":{"f9c126360cd2bf101aaa20b570d975dd59e0f3a6":{"kind":"REWORK","_number":1,"created":"2020-09-11 12:17:39.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/58/751258/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/58/751258/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/58/751258/1"}}},"commit":{"parents":[{"commit":"870b53ee8ee66920fe0b68a31fadf1ee5431ab74","subject":"fix api-service url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/870b53ee8ee66920fe0b68a31fadf1ee5431ab74"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 12:10:32.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 12:11:22.000000000","tz":480},"subject":"Add Cetus Datasource","message":"Add Cetus Datasource\n\nCetus is a based on openstack solution of k8s on openstack, and it can be\noperated through kesytone to automatically create multiple instances, and\nautomatically deploy multiple k8s clusters on instances. Cetus Datasource\nonly includes Cetus entities(cetus.cluster, cetus.pod).\n\nImplements: blueprint add-cetus-datasource\nChange-Id: I2057c901c8e28e1d61c506dcd8790d404817aaa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/f9c126360cd2bf101aaa20b570d975dd59e0f3a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/f9c126360cd2bf101aaa20b570d975dd59e0f3a6"}]},"branch":"refs/heads/master"},"2e6b0773194e66d5fa9d17620a3023dac3fc9ba9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-11 12:19:00.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/58/751258/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/58/751258/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/58/751258/2"}}},"commit":{"parents":[{"commit":"870b53ee8ee66920fe0b68a31fadf1ee5431ab74","subject":"fix api-service url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/870b53ee8ee66920fe0b68a31fadf1ee5431ab74"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 12:10:32.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 12:18:24.000000000","tz":480},"subject":"Add Cetus Datasource","message":"Add Cetus Datasource\n\nCetus is a based on openstack solution of k8s on openstack, and it can be\noperated through kesytone to automatically create multiple instances, and\nautomatically deploy multiple k8s clusters on instances. Cetus Datasource\ninclude Cetus entities(cetus.cluster, cetus.pod).\n\nImplements: blueprint add-cetus-datasource\nChange-Id: I2057c901c8e28e1d61c506dcd8790d404817aaa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/2e6b0773194e66d5fa9d17620a3023dac3fc9ba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/2e6b0773194e66d5fa9d17620a3023dac3fc9ba9"}]},"branch":"refs/heads/master"},"5d0d053b848d9e60cd800a09efa3541bd0bb148a":{"kind":"NO_CHANGE","_number":3,"created":"2020-09-11 12:20:29.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/58/751258/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/58/751258/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/58/751258/3"}}},"commit":{"parents":[{"commit":"870b53ee8ee66920fe0b68a31fadf1ee5431ab74","subject":"fix api-service url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/870b53ee8ee66920fe0b68a31fadf1ee5431ab74"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 12:10:32.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 12:19:12.000000000","tz":480},"subject":"Add Cetus Datasource","message":"Add Cetus Datasource\n\nCetus is a based on openstack solution of k8s on openstack, and it can be\noperated through kesytone to automatically create multiple instances, and\nautomatically deploy multiple k8s clusters on instances. Cetus Datasource\ninclude Cetus entities(cetus.cluster, cetus.pod).\n\nImplements: blueprint add-cetus-datasource\nChange-Id: I2057c901c8e28e1d61c506dcd8790d404817aaa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/5d0d053b848d9e60cd800a09efa3541bd0bb148a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/5d0d053b848d9e60cd800a09efa3541bd0bb148a"}]},"branch":"refs/heads/master"},"7dc6bf41f2cf1c8a2db326ce121a7d0fc4f2f3d6":{"kind":"REWORK","_number":4,"created":"2020-09-11 12:27:22.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/58/751258/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/58/751258/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/58/751258/4"}}},"commit":{"parents":[{"commit":"870b53ee8ee66920fe0b68a31fadf1ee5431ab74","subject":"fix api-service url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/870b53ee8ee66920fe0b68a31fadf1ee5431ab74"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 12:10:32.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 12:26:50.000000000","tz":480},"subject":"Add Cetus Datasource","message":"Add Cetus Datasource\n\nCetus is a based on openstack solution of k8s on openstack, and it can be\noperated through kesytone to automatically create multiple instances, and\nautomatically deploy multiple k8s clusters on instances. Cetus Datasource\ninclude Cetus entities(cetus.cluster, cetus.pod).\n\nImplements: blueprint add-cetus-datasource\nChange-Id: I2057c901c8e28e1d61c506dcd8790d404817aaa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/7dc6bf41f2cf1c8a2db326ce121a7d0fc4f2f3d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/7dc6bf41f2cf1c8a2db326ce121a7d0fc4f2f3d6"}]},"branch":"refs/heads/master"},"ff138ab8935e3efe91168c86612c76c3ae74a4ff":{"kind":"REWORK","_number":5,"created":"2020-09-11 12:32:01.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/58/751258/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/58/751258/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/58/751258/5"}}},"commit":{"parents":[{"commit":"870b53ee8ee66920fe0b68a31fadf1ee5431ab74","subject":"fix api-service url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/870b53ee8ee66920fe0b68a31fadf1ee5431ab74"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 12:10:32.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 12:31:33.000000000","tz":480},"subject":"Add Cetus Datasource","message":"Add Cetus Datasource\n\nCetus is a based on openstack solution of k8s on openstack, and it can be\noperated through kesytone to automatically create multiple instances, and\nautomatically deploy multiple k8s clusters on instances. Cetus Datasource\ninclude Cetus entities(cetus.cluster, cetus.pod).\n\nImplements: blueprint add-cetus-datasource\nChange-Id: I2057c901c8e28e1d61c506dcd8790d404817aaa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/ff138ab8935e3efe91168c86612c76c3ae74a4ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/ff138ab8935e3efe91168c86612c76c3ae74a4ff"}]},"branch":"refs/heads/master"},"79bdedea034746a5e3ac1b663540b4ee346c8b77":{"kind":"REWORK","_number":6,"created":"2020-09-11 14:12:06.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/58/751258/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/58/751258/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/58/751258/6"}}},"commit":{"parents":[{"commit":"870b53ee8ee66920fe0b68a31fadf1ee5431ab74","subject":"fix api-service url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/870b53ee8ee66920fe0b68a31fadf1ee5431ab74"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 12:10:32.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 14:10:57.000000000","tz":480},"subject":"Add Cetus Datasource","message":"Add Cetus Datasource\n\nCetus is a based on openstack solution of k8s on openstack, and it can be\noperated through kesytone to automatically create multiple instances, and\nautomatically deploy multiple k8s clusters on instances. Cetus Datasource\ninclude Cetus entities(cetus.cluster, cetus.pod).\n\nImplements: blueprint add-cetus-datasource\nChange-Id: I2057c901c8e28e1d61c506dcd8790d404817aaa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/79bdedea034746a5e3ac1b663540b4ee346c8b77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/79bdedea034746a5e3ac1b663540b4ee346c8b77"}]},"branch":"refs/heads/master"},"65efc4fb5de1d4587f1860cd7a986b7a345b5731":{"kind":"REWORK","_number":7,"created":"2020-09-11 14:30:35.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/58/751258/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/58/751258/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/58/751258/7"}}},"commit":{"parents":[{"commit":"870b53ee8ee66920fe0b68a31fadf1ee5431ab74","subject":"fix api-service url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/870b53ee8ee66920fe0b68a31fadf1ee5431ab74"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 12:10:32.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 14:23:22.000000000","tz":480},"subject":"Add Cetus Datasource","message":"Add Cetus Datasource\n\nCetus is a based on openstack solution of k8s on openstack, and it can be\noperated through kesytone to automatically create multiple instances, and\nautomatically deploy multiple k8s clusters on instances. Cetus Datasource\ninclude Cetus entities(cetus.cluster, cetus.pod).\n\nImplements: blueprint add-cetus-datasource\nChange-Id: I2057c901c8e28e1d61c506dcd8790d404817aaa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/65efc4fb5de1d4587f1860cd7a986b7a345b5731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/65efc4fb5de1d4587f1860cd7a986b7a345b5731"}]},"branch":"refs/heads/master"},"a2756c217146f46b11c6335c46552e0f3052c11d":{"kind":"REWORK","_number":8,"created":"2020-09-11 18:05:52.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/58/751258/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/58/751258/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/58/751258/8"}}},"commit":{"parents":[{"commit":"870b53ee8ee66920fe0b68a31fadf1ee5431ab74","subject":"fix api-service url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/870b53ee8ee66920fe0b68a31fadf1ee5431ab74"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 12:10:32.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 18:04:59.000000000","tz":480},"subject":"Add Cetus Datasource","message":"Add Cetus Datasource\n\nCetus is a based on openstack solution of k8s on openstack, and it can be\noperated through kesytone to automatically create multiple instances, and\nautomatically deploy multiple k8s clusters on instances. Cetus Datasource\ninclude Cetus entities(cetus.cluster, cetus.pod).\n\nImplements: blueprint add-cetus-datasource\nChange-Id: I2057c901c8e28e1d61c506dcd8790d404817aaa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/a2756c217146f46b11c6335c46552e0f3052c11d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/a2756c217146f46b11c6335c46552e0f3052c11d"}]},"branch":"refs/heads/master"},"ca465bb2a0971dfed6c77b48022fade86c66f6d7":{"kind":"REWORK","_number":9,"created":"2020-09-11 18:23:31.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/58/751258/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/58/751258/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/58/751258/9"}}},"commit":{"parents":[{"commit":"870b53ee8ee66920fe0b68a31fadf1ee5431ab74","subject":"fix api-service url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/870b53ee8ee66920fe0b68a31fadf1ee5431ab74"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 12:10:32.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 18:23:09.000000000","tz":480},"subject":"Add Cetus Datasource","message":"Add Cetus Datasource\n\nCetus is a based on openstack solution of k8s on openstack, and it can be\noperated through kesytone to automatically create multiple instances, and\nautomatically deploy multiple k8s clusters on instances. Cetus Datasource\ninclude Cetus entities(cetus.cluster, cetus.pod).\n\nImplements: blueprint add-cetus-datasource\nChange-Id: I2057c901c8e28e1d61c506dcd8790d404817aaa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/ca465bb2a0971dfed6c77b48022fade86c66f6d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/ca465bb2a0971dfed6c77b48022fade86c66f6d7"}]},"branch":"refs/heads/master"},"9638aa34878a1b4ab6d5fe3bf0a6db4a65e63091":{"kind":"REWORK","_number":10,"created":"2020-09-12 05:20:33.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/58/751258/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/58/751258/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/751258/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/58/751258/10"}}},"commit":{"parents":[{"commit":"870b53ee8ee66920fe0b68a31fadf1ee5431ab74","subject":"fix api-service url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/870b53ee8ee66920fe0b68a31fadf1ee5431ab74"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-11 12:10:32.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-12 05:18:45.000000000","tz":480},"subject":"Add Cetus Datasource","message":"Add Cetus Datasource\n\nCetus is a based on openstack solution of k8s on openstack, and it can be\noperated through kesytone to automatically create multiple instances, and\nautomatically deploy multiple k8s clusters on instances. Cetus Datasource\ninclude Cetus entities(cetus.cluster, cetus.pod).\n\nImplements: blueprint add-cetus-datasource\nChange-Id: I2057c901c8e28e1d61c506dcd8790d404817aaa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/9638aa34878a1b4ab6d5fe3bf0a6db4a65e63091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/9638aa34878a1b4ab6d5fe3bf0a6db4a65e63091"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
