)]}'
{"id":"openstack%2Fvitrage~752726","triplet_id":"openstack%2Fvitrage~master~I02b6fbbdc70b228e1734c6085ae4e06a484a556e","project":"openstack/vitrage","branch":"master","topic":"remove-six","hashtags":[],"change_id":"I02b6fbbdc70b228e1734c6085ae4e06a484a556e","subject":"Remove six.add_metaclass","status":"MERGED","created":"2020-09-18 16:26:47.000000000","updated":"2020-09-21 08:04:20.000000000","submitted":"2020-09-21 08:02:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752726-1600675327878-0e5654ba","meta_rev_id":"7ad1aeb8a9b182af906f609b0b87c70e0a73382a","_number":752726,"virtual_id_number":752726,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-21 08:02:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-21 06:50:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-21 06:50:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 06:50:34.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-09-21 08:02:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a2785ded1a780f92a2769f1845ba07eeabdc9d5b","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-09-18 16:26:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94b4b48bf212700055f0eb3f13b00e249eea48a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 17:35:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6359267aeb345cbaf88e412018f65fa : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bb41c65bdc54dd189797cce22ebccff : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/572da623adab4e5fa7ec0c1caa4b8f06 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/632e25dbeb154969b5863fad7d0d187c : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9dd20796f1b4b8a80e693a8cb1655ff : SUCCESS in 4m 43s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cbfd6d8831d546a7a410eb78ac229ec4 : SUCCESS in 48m 09s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/f019eae7e2d94716ae2e7d14389c9da4 : SUCCESS in 39m 38s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/46b9afa6efdc4a1aa27d8226e626052f : SUCCESS in 59m 04s\n- vitrage-tempest-plugin-mock https://zuul.opendev.org/t/openstack/build/544bd02fe3ad47f08336c89174504b0e : SUCCESS in 54m 53s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/cdf8e5f8bcf94fbea0d9f79d49b677fe : SUCCESS in 52m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ca18357d64d4c22ed8025eb364d078a3465b268c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-09-21 06:50:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ed1672e6c988437601db74270c0b261fd7a22c50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 06:50:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"462d34d34ce0fbe59bece7786698db1a524bc452","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:02:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2058c71b3dfc4f649046a078de085690 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05699635744f4995823ba827fa89885c : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d3be3da5d534f85af246fb5fbf6645c : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fd4c38f896a43658764672ae25b76ba : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f970b9115bbc49aab214390ca43751ea : SUCCESS in 5m 22s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cf9e61e1a8184743a0d9877eda7f8aeb : SUCCESS in 51m 13s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/1d951e0f7f4849d09ffcbf6fe27148ae : SUCCESS in 49m 49s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/ea9ebb9c6186465198426fddad9c1261 : SUCCESS in 54m 44s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/8dca0dd21422411a996dbdf3bde54963 : SUCCESS in 50m 24s","accounts_in_message":[],"_revision_number":1},{"id":"021f39afe0acbad2523a4422f3011844b2c0fcb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:02:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"84addd7fbc5a78e9d0127421bad2d64eef858c3a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:04:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49527013c4f14adaaf8785e63c90aeea : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a80f3d4f0e6f37a7a3634350b6a07185725c515a","revisions":{"a80f3d4f0e6f37a7a3634350b6a07185725c515a":{"kind":"REWORK","_number":1,"created":"2020-09-18 16:26:47.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/26/752726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/26/752726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/26/752726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/26/752726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/26/752726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/26/752726/1"}}},"commit":{"parents":[{"commit":"71d4993edc7a95c3553c901d4df034bdb9d33954","subject":"Remove six.moves","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/71d4993edc7a95c3553c901d4df034bdb9d33954"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-18 14:19:02.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-09-18 14:19:02.000000000","tz":480},"subject":"Remove six.add_metaclass","message":"Remove six.add_metaclass\n\nWith python 3.x, classes can use the metaclass\u003d logic\nto not require usage of the six library.\n\nSubsequent patches will replace other six usages.\n\nChange-Id: I02b6fbbdc70b228e1734c6085ae4e06a484a556e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/a80f3d4f0e6f37a7a3634350b6a07185725c515a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/a80f3d4f0e6f37a7a3634350b6a07185725c515a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
