)]}'
{"id":"openstack%2Fvitrage~764899","triplet_id":"openstack%2Fvitrage~master~Ie8426b50595b87539bef42520c78a138db32ac7f","project":"openstack/vitrage","branch":"master","topic":"update-py","hashtags":[],"change_id":"Ie8426b50595b87539bef42520c78a138db32ac7f","subject":"Remove unicode  from datasource","status":"MERGED","created":"2020-12-01 05:56:30.000000000","updated":"2020-12-01 16:13:26.000000000","submitted":"2020-12-01 16:11:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764899-update-py","meta_rev_id":"fffb1d1b5f95864936b7917e20c2195d2fe45d1d","_number":764899,"virtual_id_number":764899,"owner":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-01 16:11:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-01 14:08:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-01 14:08:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-01 07:14:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-01 13:16:35.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"CC"},{"updated":"2020-12-01 14:08:25.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"}],"messages":[{"id":"e0cefd615552cf24250549fff7ad277a9a650b51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"date":"2020-12-01 05:56:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"502c7a337ded557cdf1e585c401ccda32ccbe7c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 07:14:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c747536d1af64cd7b78f41d518b9c76b : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b353c64ecdb4469a2b96c09ad9ac4fa : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c63c7b6606dd4858bd20d7c2043dd72b : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cb6287bc5004279ac1d7fd846c4b85b : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82d18ac03dfb47eba75016bc95db1036 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c675f6110ea94d3bb4bf361b64947c9f : SUCCESS in 3m 56s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f924f280921e4795accaa7c33c801aa5 : SUCCESS in 54m 06s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/47ef905ce39c46ee80f3d593249e2be3 : SUCCESS in 53m 59s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/310fef35502e4d819527bcaf20ecd376 : SUCCESS in 1h 11m 07s\n- vitrage-tempest-plugin-mock https://zuul.opendev.org/t/openstack/build/ad8d030d6b3343f8859dd3eeb419f368 : SUCCESS in 1h 01m 40s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/2e1045f8131e41f88b1131b2a38c4b16 : FAILURE in 17m 03s","accounts_in_message":[],"_revision_number":1},{"id":"244f09d331225f8cf0135e3e48b885042f204635","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-12-01 13:16:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"daae18bc2808f23016304f1192d835aa26d83022","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-12-01 14:08:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3a2e9af283006cd47cb3eb9ab7857e1999bcb8de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 15:00:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67183d8389bd47649f8b7e64d396be4e : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eae7ceccd564356a01528da884bcb4a : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b4f81f09b97437fa1e7e239e9a46a6d : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7379118b1864f89832e06bc03b5b98c : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8bf22373038b4b42b43fd5e68333c587 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f81c55130347428f9ee9d81093e63046 : SUCCESS in 4m 38s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fc32998ab95849749299ba980388ab07 : SUCCESS in 44m 37s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/7c6d4662df2c4068a15ed360e4f7c6b3 : SUCCESS in 45m 16s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/15eade68b76e4c9b94389d93b92f07de : SUCCESS in 1h 36m 14s\n- vitrage-tempest-plugin-mock https://zuul.opendev.org/t/openstack/build/bbfa9ef1ce544d1eba5446e1fa28a111 : SUCCESS in 1h 01m 45s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/952bd461d38b4f0fa46c01e79579b8a9 : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":1},{"id":"86c7b5b97ccc5d9bc83660975e97ef02a07e1529","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 15:01:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c94234e1678c00869d674ee6baef1e289e0c0595","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 16:11:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/721a8878c5c649f3ad7f787c07da7443 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40bc2d58fbd24386b2bac9f5252f1bee : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ea57cda53fe4815bcd3d8745839069b : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3535f8957a504a679bdffe84512a301e : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ebc550761f845b2ae68a5a99ef33c05 : SUCCESS in 4m 16s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/14a8419187314a04bd117c60783d43d6 : SUCCESS in 52m 31s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/bf3518f98c634581aeade16f2639b599 : SUCCESS in 46m 09s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/ddd21df6e905422e9df20d7d63f7c0c8 : SUCCESS in 1h 04m 05s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/ae6591d3fdf64ef3a072da35ec2cf09f : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3b1d9ba78b73de32f4651a2a30c75629232735c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 16:11:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fffb1d1b5f95864936b7917e20c2195d2fe45d1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 16:13:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fdb186ee9364427b548a174f3c177bd : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9047c458bc4059e21b5e360131967072e806b832","revisions":{"9047c458bc4059e21b5e360131967072e806b832":{"kind":"REWORK","_number":1,"created":"2020-12-01 05:56:30.000000000","uploader":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"ref":"refs/changes/99/764899/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/99/764899/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/99/764899/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/99/764899/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/99/764899/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/99/764899/1"}}},"commit":{"parents":[{"commit":"aa3524f7a39c0da1166eb4afae655074d651cdfc","subject":"update kubernetes driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/aa3524f7a39c0da1166eb4afae655074d651cdfc"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@aliyun.com","date":"2020-12-01 05:54:48.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@aliyun.com","date":"2020-12-01 05:54:48.000000000","tz":480},"subject":"Remove unicode  from datasource","message":"Remove unicode  from datasource\n\nChange-Id: Ie8426b50595b87539bef42520c78a138db32ac7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/9047c458bc4059e21b5e360131967072e806b832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/9047c458bc4059e21b5e360131967072e806b832"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
