)]}'
{"id":"openstack%2Fvitrage~833665","triplet_id":"openstack%2Fvitrage~master~I7d85f03a07ccfa1dece068ddf907c0b6500ee9a6","project":"openstack/vitrage","branch":"master","topic":"add-zed-python-jobtemplates","hashtags":[],"change_id":"I7d85f03a07ccfa1dece068ddf907c0b6500ee9a6","subject":"Add Python3 zed unit tests","status":"MERGED","created":"2022-03-14 17:00:48.000000000","updated":"2022-03-15 05:37:10.000000000","submitted":"2022-03-15 05:33:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833665-add-zed-python-jobtemplates","meta_rev_id":"e870c14f0c3ae9a7e3b8d31ff4e25a5c0ef2adc6","_number":833665,"virtual_id_number":833665,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-15 05:33:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-14 17:06:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-14 17:06:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-14 17:06:23.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2022-03-14 18:02:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc25f8c9e897035027f9a64aa950a7e437a69bc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-03-14 17:00:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28b211dcd3622e2f2674350d511b6bc26ef99947","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-03-14 17:06:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1274fefd702f6e6c93e4264039c2b039e0d93a63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 18:02:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/050724d49187402e87dca26b5146b455 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9ebfd0864574d5997fa2abbdf17e555 : FAILURE in 7m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a51d48fc85d64537a290e2e70340b666 : FAILURE in 7m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/751fb552e40646439234eabff4f48efb : FAILURE in 9m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c1eaa1c23c0494b883d26672834e2bf : SUCCESS in 6m 22s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ddc01aec71564c8cb79f0ace8c918923 : FAILURE in 34m 57s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/7232de5a2a394822a3cc5994863cffee : FAILURE in 34m 55s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/17e8b4df11504083849ae95453d020fd : FAILURE in 41m 03s\n- vitrage-tempest-plugin-mock https://zuul.opendev.org/t/openstack/build/08e124343ab641ddb68fd8f86d2a9cfb : FAILURE in 38m 08s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/660b897493154d2eba30d93419983e68 : SUCCESS in 47m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6e1783f7a6c8a782ba5efdcbea817623a815500b","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-03-15 03:08:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a53f42da5112e8dcf0f68fc7831e8eba98b877b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 04:16:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f7e57b4526a4533af33052a7617601f : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9be19aa5c9234576941083b9baaa312a : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f409f2d03bd24834b9d4d948bcc06d35 : SUCCESS in 7m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67d5d130e63c46cc82d6219c29d00721 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acec1477c96c49d1b46c1dd364ed0e85 : SUCCESS in 6m 02s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d882b8c42a904a0fa0a6279d337678da : SUCCESS in 48m 00s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/b30012000e114e4899f74c684b182220 : SUCCESS in 50m 08s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/a15726325c3e493f94f6ddca152fca3c : SUCCESS in 1h 02m 06s\n- vitrage-tempest-plugin-mock https://zuul.opendev.org/t/openstack/build/a2cb8fc0d84b4dda930a95ce70b4372a : SUCCESS in 58m 07s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/f761b06a1efc405d9ec444542751426c : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":1},{"id":"4385269e63314d44514bbf54e96d8766c10a87c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 04:17:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e30ad8cab14471dd8fd5fcac55d40629748f1116","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 05:33:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff7168914a4c4255bf1376b877813933 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12354f54a43242a3bdf4caba4ce1e32d : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c0f128be5be4a8982540856042d8e87 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b87edc0fd345afb11975fb59c83897 : SUCCESS in 6m 21s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5440544b399b48edbd074c85369c83c2 : SUCCESS in 52m 50s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/5364534c23ed4290a2dc5ec8e8f49b35 : SUCCESS in 52m 12s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/d7fb066b425c452ead19dbe3d33ecb9d : SUCCESS in 1h 09m 47s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/b4307fdc5d1e4a66990980c5ae68031a : SUCCESS in 1h 03m 57s","accounts_in_message":[],"_revision_number":1},{"id":"5eb346d6b8883198c52751677348a534cc3c356b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 05:33:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e870c14f0c3ae9a7e3b8d31ff4e25a5c0ef2adc6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 05:37:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be2bd8426db2408b904802433ad094aa : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31b2ac7ccb536e7c0627aa05cf5f8f0d3210037c","revisions":{"31b2ac7ccb536e7c0627aa05cf5f8f0d3210037c":{"kind":"REWORK","_number":1,"created":"2022-03-14 17:00:48.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/65/833665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/65/833665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/65/833665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/65/833665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/65/833665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/65/833665/1"}}},"commit":{"parents":[{"commit":"baaa92f4ff2bf29a4cd91f313a2c90046d95dd85","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/baaa92f4ff2bf29a4cd91f313a2c90046d95dd85"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-14 17:00:46.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-14 17:00:46.000000000","tz":0},"subject":"Add Python3 zed unit tests","message":"Add Python3 zed unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for zed.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I7d85f03a07ccfa1dece068ddf907c0b6500ee9a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/31b2ac7ccb536e7c0627aa05cf5f8f0d3210037c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/31b2ac7ccb536e7c0627aa05cf5f8f0d3210037c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
