)]}'
{"id":"openstack%2Fvitrage~846354","triplet_id":"openstack%2Fvitrage~master~I2a8d09aaef4621d3f99d1d7025db9b357c44398e","project":"openstack/vitrage","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-09 14:00:04.000000000","reason":"Change was submitted"},"19134":{"account":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"last_update":"2022-08-09 12:45:42.000000000","reason":"\u003cGERRIT_ACCOUNT_19134\u003e replied on the change","reason_account":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}},"hashtags":[],"change_id":"I2a8d09aaef4621d3f99d1d7025db9b357c44398e","subject":"Get rid of warning caused by [service_credentials] region_name","status":"MERGED","created":"2022-06-17 12:50:42.000000000","updated":"2022-08-09 14:02:22.000000000","submitted":"2022-08-09 14:00:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846354","meta_rev_id":"052d7fec758f35ee1215a4ed06539b8e40d9f219","_number":846354,"virtual_id_number":846354,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-09 14:00:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-09 12:45:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-09 12:45:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-17 14:08:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-07 10:22:12.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"}],"messages":[{"id":"6d7cbaeba515235a8db294d0484fd4d9f55ea233","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-17 12:50:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72fd3a14eefe673b66940da07e7d0c294c83316e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 14:08:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/181cc10a5bf2430ca9bf67dcd057a22d : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ec94718229d470ab1a226c6f3e24e0f : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66abf6da6c3246cc9c7d3d3028345d4c : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4d899445fef473682369fb333f034d8 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51aacaba7edc48e4be046193537324d1 : SUCCESS in 3m 52s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2de9436beb3544ff943e7a02515ab979 : SUCCESS in 40m 46s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/c92c9e96c0ac4cf68d73c852df084d83 : SUCCESS in 41m 57s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/d3e145ea00064a03bcc52034e4fd94d3 : SUCCESS in 1h 10m 18s\n- vitrage-tempest-plugin-mock https://zuul.opendev.org/t/openstack/build/077fcfa6a6ae4fe7b84c3f539b9112ad : SUCCESS in 47m 21s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/a1e43e3e28af40cc9fa55e175ca4732e : SUCCESS in 44m 32s","accounts_in_message":[],"_revision_number":1},{"id":"0f6b848bc763c8d8692f0c4c06440b8f98a88944","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-07-07 10:22:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7216310c67fbcdaf572b2171bd438624cf77368b","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-08 16:22:31.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"90d49d72604999a8378941ed8fdb85ee085ce566","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 16:23:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"584c2f8e05d3508cf570e96e061636be78a6d0a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-08 16:44:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"241829f0a0f04c70f44bc119407ba4a532d901d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-08 18:49:59.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"93762767b748bb1c67db5c531b98a662c3fdd8cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-08 19:52:48.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"92407ccfa631ebe7257077d18a6e131a6798a1ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 20:56:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef255652f1bb47108ae27e446690c59f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c30f45b8e1504b4b9a8628e32d4cf96c : SUCCESS in 1m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7af8dd1bd4a94839ae69b08c81e1f8fa : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e10f984aea3c4295b255c14c6794b678 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86b648da82194a54bc870e2e461a0327 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b1abb0c061141edadebf8d8b1304d7f : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a463a11c82fa4ade83f9f2acb11e77f0 : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7f72bf3d2ff45449c778ef1ae9f3213 : SUCCESS in 3m 41s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/554fe5dd677342c8a49cd927a36e28d0 : SUCCESS in 49m 54s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/2d4262dde2a0452fb2caa38f8ea0201f : SUCCESS in 53m 46s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/83e4ba235fdc4ca29a7d4cd82d376e0c : FAILURE in 36m 03s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/e8f76ed5e11c42c2a9cc82c10b7a2cfd : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":4},{"id":"3aca092e00b4e2dab79afde50d1772e860c044b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-09 11:24:16.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"53c727857594b82c2ed252a0693ebc83424dfb0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 11:24:45.000000000","message":"Patch Set 5: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/vitrage on branch master.  The error was:\n\n  while parsing a block collection\n  in \"openstack/vitrage/.zuul.yaml@master\", line 11, column 9\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/vitrage/.zuul.yaml@master\", line 15, column 13","accounts_in_message":[],"_revision_number":5},{"id":"712cfd1602394517bf049ab98020648206f4e6d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-09 11:30:58.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"88a597683df6805f8f39c33927260b698a42d7a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 12:39:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bca4ba9f83ff400d9bcc57bee84cde51\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6c646444d0b4434fb2cf63e895602444 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fd0c39bb61d4c9396709da4ac7249b6 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4da3e33260b8454f9afe17b37d441c2b : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/feab0c557c3f43febd2a20f4d5ea2131 : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4bd1e75db74a439b988441a22a32c0e8 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81bc19a78b3c4ea882a44ad61e3a16e1 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/794ac2021f994df9a91e1f35b43fa217 : SUCCESS in 3m 27s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b9f34441b1ff4673b2ba873e033658cc : SUCCESS in 50m 48s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/b3226ea0530d4b389827043cc74fb2ff : SUCCESS in 51m 59s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/aed14f507cb14080b4b2855c400dda3c : FAILURE in 42m 18s (non-voting)\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/daf957e862cf4b4e9206e1dc7b6b6c96 : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":6},{"id":"2e1db7b33d49d6d7884e1b1dc62c5eccf7ff8825","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-08-09 12:45:42.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"3cab2e67438eb645fff80291a4898595d5f715ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 12:46:30.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7f224abfc5b1df4e3035bf4ef08ca5b9a884b1c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 14:00:04.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/022d285e916247b18f11e3c7ba0b8027\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a6f28ae1a6424dd485830251b52103aa : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/790279a74f8a428ea427d5b9fa033c03 : SUCCESS in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59b05370b5a44472946c4ccb6673ce5d : SUCCESS in 8m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0eb8a02c002b4f88b36dd90bcb3afbff : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c663ab4585a342a6a417432e54fe8088 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4f7cbe8993243cfab1b6edb519ab5c4 : SUCCESS in 7m 26s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/432e525317dc4394a3b7e98a54abb537 : SUCCESS in 1h 02m 47s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/329cce4884b749938ee58228f2915518 : SUCCESS in 57m 29s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/094a0251b54642f4bb6d6220538e7879 : FAILURE in 30m 37s (non-voting)\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/18bc5eb290d141298b505a3f0b0be19b : SUCCESS in 1h 11m 22s","accounts_in_message":[],"_revision_number":6},{"id":"f2a9d0960a620dea7a590fb69f2a110381b22119","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 14:00:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"052d7fec758f35ee1215a4ed06539b8e40d9f219","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 14:02:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90b1c3b2927d4c818ba98b3c5129f0c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1b0e5e333d94de398b89ab01ab0e3ce : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de71ca5999884658b3e5dd30297fc94a : SUCCESS in 50s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7f2098610a97625eb6a73bb25152168f41026ca5","revisions":{"f365594578683112fa6fcf5ad6725804a1ca58ba":{"kind":"REWORK","_number":1,"created":"2022-06-17 12:50:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/846354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/54/846354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/54/846354/1"}}},"commit":{"parents":[{"commit":"f538c886b39c27c66782f877fd1bebe6a501842d","subject":"move queue to project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/f538c886b39c27c66782f877fd1bebe6a501842d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-17 12:49:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-17 12:50:39.000000000","tz":540},"subject":"Get rid of warning caused by [service_credentials] region_name","message":"Get rid of warning caused by [service_credentials] region_name\n\noslo.config automatically replaces - in option names by _. Because of\nthis behavior now we get the following warning always if\nthe [service_credentials] region_name parameter is set.\n\nDeprecated: Option \"region_name\" from group \"service_credentials\" is\ndeprecated. Use option \"region_name\" from group \"service_credentials\".\n\nChange-Id: I2a8d09aaef4621d3f99d1d7025db9b357c44398e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/f365594578683112fa6fcf5ad6725804a1ca58ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/f365594578683112fa6fcf5ad6725804a1ca58ba"}]},"branch":"refs/heads/master"},"e10e0fbd2b9a4b88392ac90a170b826b43fa4ab2":{"kind":"REWORK","_number":2,"created":"2022-08-08 16:44:24.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/54/846354/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/54/846354/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/54/846354/2"}}},"commit":{"parents":[{"commit":"f538c886b39c27c66782f877fd1bebe6a501842d","subject":"move queue to project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/f538c886b39c27c66782f877fd1bebe6a501842d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-17 12:49:13.000000000","tz":540},"committer":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2022-08-08 16:44:24.000000000","tz":0},"subject":"Get rid of warning caused by [service_credentials] region_name","message":"Get rid of warning caused by [service_credentials] region_name\n\noslo.config automatically replaces - in option names by _. Because of\nthis behavior now we get the following warning always if\nthe [service_credentials] region_name parameter is set.\n\nDeprecated: Option \"region_name\" from group \"service_credentials\" is\ndeprecated. Use option \"region_name\" from group \"service_credentials\".\n\nChange-Id: I2a8d09aaef4621d3f99d1d7025db9b357c44398e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/e10e0fbd2b9a4b88392ac90a170b826b43fa4ab2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/e10e0fbd2b9a4b88392ac90a170b826b43fa4ab2"}]},"branch":"refs/heads/master"},"52d1cadf87fbb3d4ee5ac0b33ebbd9e71f67e0bc":{"kind":"REWORK","_number":3,"created":"2022-08-08 18:49:59.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/54/846354/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/54/846354/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/54/846354/3"}}},"commit":{"parents":[{"commit":"f538c886b39c27c66782f877fd1bebe6a501842d","subject":"move queue to project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/f538c886b39c27c66782f877fd1bebe6a501842d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-17 12:49:13.000000000","tz":540},"committer":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2022-08-08 18:49:55.000000000","tz":0},"subject":"Get rid of warning caused by [service_credentials] region_name","message":"Get rid of warning caused by [service_credentials] region_name\n\noslo.config automatically replaces - in option names by _. Because of\nthis behavior now we get the following warning always if\nthe [service_credentials] region_name parameter is set.\n\nDeprecated: Option \"region_name\" from group \"service_credentials\" is\ndeprecated. Use option \"region_name\" from group \"service_credentials\".\n\nChange-Id: I2a8d09aaef4621d3f99d1d7025db9b357c44398e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/52d1cadf87fbb3d4ee5ac0b33ebbd9e71f67e0bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/52d1cadf87fbb3d4ee5ac0b33ebbd9e71f67e0bc"}]},"branch":"refs/heads/master"},"c78c29e39223533a71c665fa6676e4b28ec914c3":{"kind":"REWORK","_number":4,"created":"2022-08-08 19:52:48.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/54/846354/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/54/846354/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/54/846354/4"}}},"commit":{"parents":[{"commit":"f538c886b39c27c66782f877fd1bebe6a501842d","subject":"move queue to project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/f538c886b39c27c66782f877fd1bebe6a501842d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-17 12:49:13.000000000","tz":540},"committer":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2022-08-08 19:52:47.000000000","tz":0},"subject":"Get rid of warning caused by [service_credentials] region_name","message":"Get rid of warning caused by [service_credentials] region_name\n\noslo.config automatically replaces - in option names by _. Because of\nthis behavior now we get the following warning always if\nthe [service_credentials] region_name parameter is set.\n\nDeprecated: Option \"region_name\" from group \"service_credentials\" is\ndeprecated. Use option \"region_name\" from group \"service_credentials\".\n\nChange-Id: I2a8d09aaef4621d3f99d1d7025db9b357c44398e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/c78c29e39223533a71c665fa6676e4b28ec914c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/c78c29e39223533a71c665fa6676e4b28ec914c3"}]},"branch":"refs/heads/master"},"48adf9eeeb948989facb7ff5e2c2c80f20ac8f29":{"kind":"REWORK","_number":5,"created":"2022-08-09 11:24:16.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/54/846354/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/54/846354/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/54/846354/5"}}},"commit":{"parents":[{"commit":"f538c886b39c27c66782f877fd1bebe6a501842d","subject":"move queue to project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/f538c886b39c27c66782f877fd1bebe6a501842d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-17 12:49:13.000000000","tz":540},"committer":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2022-08-09 11:24:16.000000000","tz":0},"subject":"Get rid of warning caused by [service_credentials] region_name","message":"Get rid of warning caused by [service_credentials] region_name\n\noslo.config automatically replaces - in option names by _. Because of\nthis behavior now we get the following warning always if\nthe [service_credentials] region_name parameter is set.\n\nDeprecated: Option \"region_name\" from group \"service_credentials\" is\ndeprecated. Use option \"region_name\" from group \"service_credentials\".\n\nChange-Id: I2a8d09aaef4621d3f99d1d7025db9b357c44398e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/48adf9eeeb948989facb7ff5e2c2c80f20ac8f29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/48adf9eeeb948989facb7ff5e2c2c80f20ac8f29"}]},"branch":"refs/heads/master"},"7f2098610a97625eb6a73bb25152168f41026ca5":{"kind":"REWORK","_number":6,"created":"2022-08-09 11:30:58.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/54/846354/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/54/846354/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/846354/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/54/846354/6"}}},"commit":{"parents":[{"commit":"f538c886b39c27c66782f877fd1bebe6a501842d","subject":"move queue to project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/f538c886b39c27c66782f877fd1bebe6a501842d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-17 12:49:13.000000000","tz":540},"committer":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2022-08-09 11:30:57.000000000","tz":0},"subject":"Get rid of warning caused by [service_credentials] region_name","message":"Get rid of warning caused by [service_credentials] region_name\n\noslo.config automatically replaces - in option names by _. Because of\nthis behavior now we get the following warning always if\nthe [service_credentials] region_name parameter is set.\n\nDeprecated: Option \"region_name\" from group \"service_credentials\" is\ndeprecated. Use option \"region_name\" from group \"service_credentials\".\n\nChange-Id: I2a8d09aaef4621d3f99d1d7025db9b357c44398e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/7f2098610a97625eb6a73bb25152168f41026ca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/7f2098610a97625eb6a73bb25152168f41026ca5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[]}
