)]}'
{"id":"openstack%2Fvitrage~871177","triplet_id":"openstack%2Fvitrage~master~I8e63eec682c4ff5185588cbd88983b0fcce20fdc","project":"openstack/vitrage","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-01-25 15:46:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8e63eec682c4ff5185588cbd88983b0fcce20fdc","subject":"Use new get_rpc_client API from oslo.messaging","status":"ABANDONED","created":"2023-01-19 20:51:19.000000000","updated":"2023-01-25 15:46:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f0778765b62d9aa52d0799f9e37d9abd599bb821","_number":871177,"virtual_id_number":871177,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-19 22:07:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-19 22:07:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9fec2a794526c9abff30a1bdc31ce267a7ba8865","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-19 20:51:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0f3a07c81402d5651663de780ccc1198163b42d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 22:07:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9ba3d6414524b9a927a291cd2d4ba49\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cca10b042132471cabc24080099ba7e7 : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3d1b32846814fa48045863c9861a21c : FAILURE in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8698d6b005784ae6b6c647479e3d15de : FAILURE in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1bcebe2c2f714035897e3127e56d45db : FAILURE in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67e21927f4644150a651148be4955c92 : FAILURE in 2m 08s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c4756a51b25441299d4652025b51c1ff : SUCCESS in 47m 27s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/9dc5a29074cd423f942f08173c4e1f18 : SUCCESS in 59m 18s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/15e74def542e418bbb56c2d57a1ce1be : SUCCESS in 1h 05m 47s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/b8a4f7aa196844beb6351ba7aa752e8b : SUCCESS in 59m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f0778765b62d9aa52d0799f9e37d9abd599bb821","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-25 15:46:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b00bdd6373d484c5f50d9e97044802f1f4f5ac8","revisions":{"2b00bdd6373d484c5f50d9e97044802f1f4f5ac8":{"kind":"REWORK","_number":1,"created":"2023-01-19 20:51:19.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/77/871177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/77/871177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/77/871177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/77/871177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/77/871177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/77/871177/1"}}},"commit":{"parents":[{"commit":"43e8e5f18a26f6cdb64264ebc34a4e633f9cd213","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/43e8e5f18a26f6cdb64264ebc34a4e633f9cd213"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-19 20:51:11.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-19 20:51:17.000000000","tz":0},"subject":"Use new get_rpc_client API from oslo.messaging","message":"Use new get_rpc_client API from oslo.messaging\n\nUse the new API that is consistent with\nthe existing API instead of instantiating the client\nclass directly.\n\nThis was introduced in release 14.1.0 here [1] and\nadded into oslo.messaging here [2]\n\n[1] https://review.opendev.org/c/openstack/requirements/+/869340\n[2] https://review.opendev.org/c/openstack/oslo.messaging/+/862419\n\nChange-Id: I8e63eec682c4ff5185588cbd88983b0fcce20fdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/2b00bdd6373d484c5f50d9e97044802f1f4f5ac8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/2b00bdd6373d484c5f50d9e97044802f1f4f5ac8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
