)]}'
{"id":"openstack%2Fvitrage~873858","triplet_id":"openstack%2Fvitrage~master~Ia015800c59e46ae1cdae4510b8f51132585f92d0","project":"openstack/vitrage","branch":"master","topic":"fix_tox","attention_set":{},"removed_from_attention_set":{"19134":{"account":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"last_update":"2023-02-17 09:31:33.000000000","reason":"\u003cGERRIT_ACCOUNT_19134\u003e replied on the change","reason_account":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}},"hashtags":[],"change_id":"Ia015800c59e46ae1cdae4510b8f51132585f92d0","subject":"Fix tox 4 changes","status":"MERGED","created":"2023-02-16 20:06:23.000000000","updated":"2023-02-17 11:50:44.000000000","submitted":"2023-02-17 11:48:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873858-fix_tox","meta_rev_id":"f0606e2bf20effd1132b564cd1901d2f23b463aa","_number":873858,"virtual_id_number":873858,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 11:48:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-17 07:57:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-17 07:57:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-16 20:58:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a56551c21b513e0c7f2b2617a7f3b08737f2afee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2023-02-16 20:06:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bace18a006fd8e5dc57403007997af0409650393","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2023-02-16 20:07:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3aa15b1048d4ae3359b510ca5a7dca586aa109f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 20:58:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4757259e61ad4979a2c30389aa236637\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/419ed00fec7c46a09f932a651766fcc3 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19556181865f49758a2dbdb763c51ae4 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18e995c8e1ed49d4b1d5a39202992c72 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26d69a805afd4cbc935067c4d297234e : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8695fe858b7d4118a978a8be3d1e7764 : SUCCESS in 2m 46s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a6f08fb79b99491fb93d8a5a8f2a97b2 : SUCCESS in 43m 12s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/fda1e63df7df4db496e7ffd940cf4a8f : SUCCESS in 43m 25s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/a4c1be36ee124e37876c634eaaea3da5 : FAILURE in 27m 28s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/07626313d5ee4063892cd211af0a055f : SUCCESS in 42m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3442e922f0e7e39787ccf8dfd222907b3e634c93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2023-02-17 07:55:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ad8a493019b737d2eeba7184d2a00aaa619d80db","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2023-02-17 07:57:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"73aa29d7d4eef228f332d41cf4fc8664bc711d5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 09:30:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4094e19cf39c4052983bcac0f5e3eeb6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5596595440da46279efaedca8e0aca0e : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbf8655175354328808e91ffaa05f823 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f549abd70a1a433abe25d84960ab9696 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e00e30ca5e484c3e8713ecdef00beb81 : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69b839fe15fb4093ad9a0eda9fcf2e2d : SUCCESS in 2m 41s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7e63c8289fcf4e7b808621723ee24215 : SUCCESS in 45m 22s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/13780114f08e4e29924a70066ac2d022 : SUCCESS in 42m 35s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/5c6076c1bad7482e9eab146b706a6660 : FAILURE in 35m 49s (non-voting)\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/3f2a544bb75044faac4fb86034a79ae9 : POST_FAILURE in 1h 17m 25s","accounts_in_message":[],"_revision_number":2},{"id":"d6ebd99484c03b3e1f91669dbfd5e9da919ef5d2","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2023-02-17 09:31:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bd7d48a5a86bbdf089d1a41b8640cd4cbd2c28b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:42:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4eb692aa1ce4eb6b0adc9012f0c671f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27ab44498a804bf28743ccaf07836195 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be74a8f9d4544caabaa1657b834533ad : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fda4aba27d094ce2ba8b017638b65581 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed6dfe5fe23c407598283c6ee43f5033 : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1a31397d30745f1bdc67f38a376adf8 : SUCCESS in 4m 13s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/11b06a91cac74422927bdd5773f67c06 : SUCCESS in 44m 31s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/927a91746a4f49ddad588e0532caeacd : SUCCESS in 46m 14s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/17da32f08a3d48b3bb3a588b0901e475 : FAILURE in 38m 33s (non-voting)\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/1edb51b361a84d0c92dfa40a7e18d4a9 : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":2},{"id":"8fd58d0fee2001f37fe056928e8a87869e038341","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:43:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3af77ee533308e234e900bdffa24b223627e115c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 11:48:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b8a003d0857447cbdc73add77413ecc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2834760f04f41b6b769dcd87fe72584 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca834e68c3c14f0d85c4fc01788a3416 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d806224ebe0469e89dab488d1022dd8 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a1788a6386f49298abd30351b9c5e10 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6bb7b652ab848d3a98226e87ef26fbb : SUCCESS in 2m 56s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b5022e142d4249c293cfe560d3e9e998 : SUCCESS in 53m 58s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/1649348e4c7a4e47b81ce01dfa6dcc4f : SUCCESS in 50m 13s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/a890cfd542e54fa190cb391d450491e5 : FAILURE in 36m 23s (non-voting)\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/53d5ba66fc944a45aed608c71e8622ee : SUCCESS in 58m 45s","accounts_in_message":[],"_revision_number":2},{"id":"98e1e6d0cb72ea73a446113d22c63c63b567cf77","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 11:48:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f0606e2bf20effd1132b564cd1901d2f23b463aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 11:50:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f479241ba3824e59ae6d495525c717d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb8e529d4ffa4818b5c03a8fd1f4b8c8 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ccd80dd2403444f8b5f78dce5b88c96 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f3feec42643c21b664233264b213a471e0f795f7","revisions":{"bc0882eee1da6d0d424796bea1c899fb39f802bb":{"kind":"REWORK","_number":1,"created":"2023-02-16 20:06:23.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/58/873858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/58/873858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/873858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/873858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/873858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/58/873858/1"}}},"commit":{"parents":[{"commit":"6450d9cc44c1caef204c3d3393103a356aea3c5f","subject":"Merge \"Fix missing oslo.messaging options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/6450d9cc44c1caef204c3d3393103a356aea3c5f"}]}],"author":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2023-02-16 20:03:49.000000000","tz":120},"committer":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2023-02-16 20:06:21.000000000","tz":120},"subject":"Fix tox 4 changes","message":"Fix tox 4 changes\n\nChange-Id: Ia015800c59e46ae1cdae4510b8f51132585f92d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/bc0882eee1da6d0d424796bea1c899fb39f802bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/bc0882eee1da6d0d424796bea1c899fb39f802bb"}]},"branch":"refs/heads/master"},"f3feec42643c21b664233264b213a471e0f795f7":{"kind":"REWORK","_number":2,"created":"2023-02-17 07:55:50.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/58/873858/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/58/873858/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/873858/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/873858/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/58/873858/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/58/873858/2"}}},"commit":{"parents":[{"commit":"6450d9cc44c1caef204c3d3393103a356aea3c5f","subject":"Merge \"Fix missing oslo.messaging options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/6450d9cc44c1caef204c3d3393103a356aea3c5f"}]}],"author":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2023-02-16 20:03:49.000000000","tz":120},"committer":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2023-02-17 07:55:50.000000000","tz":0},"subject":"Fix tox 4 changes","message":"Fix tox 4 changes\n\nChange-Id: Ia015800c59e46ae1cdae4510b8f51132585f92d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/f3feec42643c21b664233264b213a471e0f795f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/f3feec42643c21b664233264b213a471e0f795f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[]}
