)]}'
{"id":"openstack%2Fvitrage~905578","triplet_id":"openstack%2Fvitrage~master~I78353ccec3074a10e10dad5c6583456016d962a1","project":"openstack/vitrage","branch":"master","hashtags":[],"change_id":"I78353ccec3074a10e10dad5c6583456016d962a1","subject":"Install pymemcache in devstack when needed","status":"MERGED","created":"2024-01-15 11:30:38.000000000","updated":"2024-01-16 09:14:35.000000000","submitted":"2024-01-16 09:09:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905578","meta_rev_id":"ae92fe2112df0cd8f219d73ecf0d67fb21557c01","_number":905578,"virtual_id_number":905578,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-16 09:09:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-01-16 07:54:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2024-01-16 07:54:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-15 12:39:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d89596921dc1202d81df8745da810b7dc57c6f98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-01-15 11:30:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40dab00cec7e8fe057350445a01111601fb41666","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 12:39:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/834808c938234ed5add6ca6b6d2525e3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/581047a93759482eb360335892de5bcf : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a639293de33c4ba4b66f3671cd429e37 : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96325fc74b5a4669987a441dfb024219 : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2650850f87c2430cbf5f1bfe5d682030 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fe1093e42a548e8a24251d616350abf : SUCCESS in 3m 45s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/6216143fea984572906d5221ff7d0b68 : SUCCESS in 51m 22s","accounts_in_message":[],"_revision_number":1},{"id":"148642966ebd9f6b7ec27f082300bf58d66642d5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-01-16 07:54:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8bd79212b9697b258081cad7ea404457ddd51c68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 07:55:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8fbc991fa6284b9b0e00b66d65d139c0825fa06e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 09:09:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/413fe69d834b467386ccc63d060c0e47\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56e1bb906a544e4794e9539458833360 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd4817d3a1b74124b6b8802e2fbcb12f : SUCCESS in 6m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7450c8a273474871a96592b1a060002c : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1366c201961b47f9a131c860ba88d3b6 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/522992d44e78487aad0fdd666bcc4a20 : SUCCESS in 4m 08s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/d2a85f1d8d804336ad08f6da705043c8 : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":1},{"id":"8dc30b61db7a06075e25d100089aad93ee793afc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 09:09:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ae92fe2112df0cd8f219d73ecf0d67fb21557c01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 09:14:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dadd3b5d920e4ab5a7b78d8145d6743e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9700ffa6b15447e69179cf1e5944148a : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d1d6cd548298d84810579250a2f61e36ae24fb5","revisions":{"0d1d6cd548298d84810579250a2f61e36ae24fb5":{"kind":"REWORK","_number":1,"created":"2024-01-15 11:30:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/78/905578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/78/905578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/78/905578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/78/905578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/78/905578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/78/905578/1"}}},"commit":{"parents":[{"commit":"f200b8df2a6b3cd0221a6d1e511484e1eb07443c","subject":"Adopt for networkx 3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/f200b8df2a6b3cd0221a6d1e511484e1eb07443c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-01-15 11:30:21.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-01-15 11:30:21.000000000","tz":60},"subject":"Install pymemcache in devstack when needed","message":"Install pymemcache in devstack when needed\n\nIn case VITRAGE_COORDINATION_URL is set to memcached, we need to\nensure that all dependencies are met inside the venv.\n\nChange-Id: I78353ccec3074a10e10dad5c6583456016d962a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/0d1d6cd548298d84810579250a2f61e36ae24fb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/0d1d6cd548298d84810579250a2f61e36ae24fb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
