)]}'
{"id":"openstack%2Fvitrage~905731","triplet_id":"openstack%2Fvitrage~stable%2F2023.1~I78353ccec3074a10e10dad5c6583456016d962a1","project":"openstack/vitrage","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-11-14 11:15:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I78353ccec3074a10e10dad5c6583456016d962a1","subject":"Install pymemcache in devstack when needed","status":"ABANDONED","created":"2024-01-17 08:26:28.000000000","updated":"2024-11-14 11:15:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":905578,"cherry_pick_of_patch_set":1,"meta_rev_id":"71b9610e96f9117b6b299601c4778be6f9693a33","_number":905731,"virtual_id_number":905731,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-17 09:37:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-17 09:37:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-14 11:15:52.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"5e1533157fffa02d22df1c3fc7172ebfa0f9dc2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-01-17 08:26:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a1207db934879120d14ca3ca788716216e64aa39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 09:37:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b276232cb6de43c5b6ba239c9b123318\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60c7472e7ef645c888d7cd2177d3a6c8 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73129f0c6c52411196032556bb2e19ed : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3423818d3d9145f4aa9285bb96d4a1dd : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f80aba3335a447fa3b8df282415c311 : SUCCESS in 3m 48s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/efc03240cb964935b319fb90aa1dc10d : FAILURE in 50m 15s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/7683371989ab47c4affbcdf02fbf758c : FAILURE in 52m 35s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/f34d504e03bc4f1b95eb305497b018cb : SUCCESS in 1h 04m 46s (non-voting)\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/105dfed4a04f45fcb6aa568b7cfd35ec : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":1},{"id":"71b9610e96f9117b6b299601c4778be6f9693a33","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-11-14 11:15:52.000000000","message":"Abandoned\n\nstable/2023.1 branch of openstack/vitrage is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/2023.1 if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d2781582218453284df75acd7523369aa737de5","revisions":{"0d2781582218453284df75acd7523369aa737de5":{"kind":"REWORK","_number":1,"created":"2024-01-17 08:26:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/31/905731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/31/905731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/31/905731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/31/905731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/31/905731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/31/905731/1"}}},"commit":{"parents":[{"commit":"c0cbd11a8ddd069607ffc87ecda92bd3da403b61","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/c0cbd11a8ddd069607ffc87ecda92bd3da403b61"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-01-15 11:30:21.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-01-17 08:26:28.000000000","tz":0},"subject":"Install pymemcache in devstack when needed","message":"Install pymemcache in devstack when needed\n\nIn case VITRAGE_COORDINATION_URL is set to memcached, we need to\nensure that all dependencies are met inside the venv.\n\nChange-Id: I78353ccec3074a10e10dad5c6583456016d962a1\n(cherry picked from commit 0d1d6cd548298d84810579250a2f61e36ae24fb5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/0d2781582218453284df75acd7523369aa737de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/0d2781582218453284df75acd7523369aa737de5"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
