)]}'
{"id":"openstack%2Fvitrage~905732","triplet_id":"openstack%2Fvitrage~stable%2F2023.2~I78353ccec3074a10e10dad5c6583456016d962a1","project":"openstack/vitrage","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-04-30 14:47:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I78353ccec3074a10e10dad5c6583456016d962a1","subject":"Install pymemcache in devstack when needed","status":"ABANDONED","created":"2024-01-17 08:26:48.000000000","updated":"2025-04-30 14:47:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":905578,"cherry_pick_of_patch_set":1,"meta_rev_id":"9403d1519eba4d6a50bebd6ee04f200febd06b6d","_number":905732,"virtual_id_number":905732,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-17 09:36:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-17 09:36:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-30 14:47:57.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"fc02e8df7f3f5cbf12c3cf386f78b43f489c49e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-01-17 08:26:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1986f3150c217c8db7fe71e5aca770343f7d3b25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 09:36:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2f8f286cd2c4b46ac8bc4e4dacf3328\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60772d48f6c9431abf16a643a34a0a6d : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/217ee7c1e9634c2e8533f5163cbdaff5 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20731e7e17b04170a40a2369df16d797 : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c86e6f7144d40ca9621eea8e1e28222 : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dfb08ea6355d40acb8fd584d809ef4b4 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfe37f5b24404aeab93aef47deab142d : SUCCESS in 3m 32s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/abf8472f3192448eaeb42d7ea9c8936d : FAILURE in 37m 42s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/604eb76f6de34a3b951db2d8ba982ebd : FAILURE in 48m 15s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/5de5642134f648e7a0c292f44e5032bf : SUCCESS in 1h 02m 08s (non-voting)\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/04473a07d3614519af91cf7b6de85cd9 : SUCCESS in 1h 03m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9403d1519eba4d6a50bebd6ee04f200febd06b6d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-04-30 14:47:57.000000000","message":"Abandoned\n\nstable/2023.2 branch of openstack/vitrage transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de04ff45a41a221dc89b776661893e155200a690","revisions":{"de04ff45a41a221dc89b776661893e155200a690":{"kind":"REWORK","_number":1,"created":"2024-01-17 08:26:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/905732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/32/905732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/32/905732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/32/905732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/32/905732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/32/905732/1"}}},"commit":{"parents":[{"commit":"ad67ad39064c2f3243637b38d11d609b83e87cb7","subject":"[coordination] backend_url should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/ad67ad39064c2f3243637b38d11d609b83e87cb7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-01-15 11:30:21.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-01-17 08:26:48.000000000","tz":0},"subject":"Install pymemcache in devstack when needed","message":"Install pymemcache in devstack when needed\n\nIn case VITRAGE_COORDINATION_URL is set to memcached, we need to\nensure that all dependencies are met inside the venv.\n\nChange-Id: I78353ccec3074a10e10dad5c6583456016d962a1\n(cherry picked from commit 0d1d6cd548298d84810579250a2f61e36ae24fb5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/de04ff45a41a221dc89b776661893e155200a690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/de04ff45a41a221dc89b776661893e155200a690"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
