)]}'
{"id":"openstack%2Fvitrage~912519","triplet_id":"openstack%2Fvitrage~master~I32a1417c001cdb6be0719baeb186f79609875f4d","project":"openstack/vitrage","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-03-30 09:15:50.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I32a1417c001cdb6be0719baeb186f79609875f4d","subject":"Replace use of LegacyEngineFacade and adopt to werkzeug\u003e3.0","status":"MERGED","created":"2024-03-12 10:18:24.000000000","updated":"2024-03-30 11:28:22.000000000","submitted":"2024-03-30 11:22:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"912519","meta_rev_id":"03cc1e075afc91ea37fe5a440838c137e3a58eed","_number":912519,"virtual_id_number":912519,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-30 11:22:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-03-30 09:15:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2024-03-30 09:15:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-12 13:10:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b59046c0121ef8fdce1940d5ad4191e3e050866","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-12 10:18:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae057e7c4091e898ce57af0f4b0bdbc1e6b1bf6a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-12 10:20:30.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d82faafd1a45af1671b2b71221cb21cb99aab49","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-12 10:25:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1853a11869525bc29a0d7aae306ce85d5d4c3d03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-12 11:10:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"aac60947b7596acde8b46c8f438603934a7e9d24","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-12 11:13:26.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51464d00d5b2e2e5683a97d4bc63b74db7fb9b6d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-12 11:13:30.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"7ecf68e9a322b05bd976ddfbea3aba89dddc0c17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 13:10:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f944aed76734403cbb2c626865beda69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc0b8ab66c0042de9621d8fc4756ab58 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38daf89aa43b4b8bbb1b8382638f9505 : FAILURE in 9m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4b9b055e5f0497aad4e485a970cb4e1 : FAILURE in 9m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6971a6fadc884a97be2516ccfbddf4e6 : FAILURE in 9m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3332326aafa341bbbca006f6a70b99c9 : SUCCESS in 5m 28s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e6e0b59220d24c05be24242a42421652 : FAILURE in 50m 07s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/6e487c5bba2f4055b673c3a589ca9837 : FAILURE in 42m 33s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/a2ea38d05fe84689a1ef49abcce53c7b : FAILURE in 1h 23m 43s (non-voting)\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/302aa33f82be41c1bc1ca3554d2a555b : SUCCESS in 1h 00m 50s","accounts_in_message":[],"_revision_number":2},{"id":"e3c62e5976c880414afc6f941bd79e6312825d28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-12 13:39:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"18c7b5e6b7ec530998f13ef91141e4ef0f63cec1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 15:14:11.000000000","message":"Patch Set 3: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1186bdf0e12b4f979e16a9249e7b1f93\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10a7ba0aa6f6454b9af64bcb6d8ec368 : FAILURE in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b560f48a95b47529543bd8c45c79f9b : FAILURE in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/326ac911b5d14d8fa215dd27f03386f9 : FAILURE in 8m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46bc9ac7acd54ca1ada6eda20d1ff3de : FAILURE in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b430b761f273424081a16a366dadf609 : SUCCESS in 4m 51s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a84b39706fc64a51a91766d14c97fba7 : FAILURE in 53m 29s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/20f705d053ca47b38e67a847e4197305 : FAILURE in 46m 33s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/4ec47c53e4384bef9804d77cb44732c9 : FAILURE in 1h 06m 13s (non-voting)\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/b95afe6438e44e8eb2042a1fbec2afee : SUCCESS in 1h 32m 36s","accounts_in_message":[],"_revision_number":3},{"id":"45541d9e53a40805b3c535e99a8a2d989186544f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-29 16:04:05.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"46e6064903c74039ffc6cad39ceb2b39d4c96ece","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-29 16:19:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7e47709ee6f772565d175feace8563275876e2fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-29 16:22:40.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e1e8481bbd32fad0bcb9c5535b121175745e8198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 18:40:39.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b039d94815844eea930fd7abf8485af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bd20f924a804521ab4118c5336c029e : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ea3035b9326430b8629a0afa148148a : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a32e6b4ec6a4c8d8cab210a12e73553 : SUCCESS in 6m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f110aafdf45a46359b034c22919e7122 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94a7e972358540aba322d3481bb853f9 : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/92e2758415b74ad3b30bdddb13fc20c0 : FAILURE in 7m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e357758deb94570910f91e11223bc6f : SUCCESS in 4m 17s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6b60f7b81e4f41828dd010d3020e6104 : FAILURE in 38m 08s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/2eedf2346848483292ffb641bf6ee682 : FAILURE in 30m 20s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/8129af956cf84d7ba8391c49e24f6f38 : SUCCESS in 1h 01m 06s (non-voting)\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/efca6917e3b543a487c521ef169d8794 : TIMED_OUT in 2h 12m 27s","accounts_in_message":[],"_revision_number":6},{"id":"4e1e91df2ca2466e04cf3845d99c4385f54c7070","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-29 19:19:45.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"6086b901dd52ddcd722c5ed490403852a042d900","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 21:34:42.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3eea87f88b344b01ae7d66d0438d1711\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdd43ec93ac5445a84b3d52ff6ad6cca : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89f21779b3bf459d9df22dd121d03ef5 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f30af92664c5457895a47df88f027ef9 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1050b8e1d185458993d972376d85f5b4 : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6827188cb7e04520a315b61a58939f2a : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/20346cef14e846f6a3ba6d1cc36facce : FAILURE in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8686592132d945b79be5156891defe6a : SUCCESS in 3m 49s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/88c1a7239a5a4954a5e24a9d11e64526 : SUCCESS in 29m 22s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/a306c3fbee31422290e50d4648052de6 : SUCCESS in 46m 53s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/1afb8ca898ce43e59356a29fd43f1d46 : SUCCESS in 1h 01m 37s (non-voting)\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/f3fdf36837004056b349bed548d7b2d6 : TIMED_OUT in 2h 12m 29s","accounts_in_message":[],"_revision_number":7},{"id":"290a814bc205a2f75fab5c781a402a0377776463","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-30 09:15:36.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"6eb3a7e8a6880d23eb89aaa8bbd6364f74be372c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-30 09:15:50.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"5e839a76f305e933b5763884f11628067f1ae7ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-30 10:27:58.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16fb1307d4614d1db00d4c98a7642fe2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95a823be47534cb393041b23ca6db946 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/727d405ca6974c22a114b98261236274 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11bea1b3ce14492ba132fa2023fa4e1f : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/951bfd8d38184cd2b4036b4bcc23afd2 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69cc9ed922ce43dd92de3def4c7efad0 : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/776d357b55374bfa8e27a6964bff2823 : FAILURE in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a847dc63be46472197735f3eb57723d5 : SUCCESS in 4m 07s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ec35e3e526144ff085ffe0adc08a76de : SUCCESS in 46m 59s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/3167a49ab27b4a7096b9aaad67f69951 : SUCCESS in 41m 37s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/a04ffe0c470348d69c65e30376731d59 : SUCCESS in 1h 01m 43s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1d34acaec02f7efed3939337c9e96be4420bd293","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-30 10:28:38.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"5d18ec691bdfe8aa67f30d7c94135ba299ff460e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-30 11:22:07.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd291f2de2d745aa955b16b871b9777c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a2c254fe3834be2a129e0a945d107f8 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3e0a2b3f9534118a4934fd8128cfc88 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d6ff16056374d6dbd33158938939dec : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0762dd5eac6b465f8238ecaf852aa2a6 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7199744e6c3c448285bed66dd14cb861 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/326b0b4ed0ef40718960bb87fd5359f8 : SUCCESS in 2m 56s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/18a90ce452d142569835f9d44e1557d1 : SUCCESS in 44m 32s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/f0c8c58e02014badafcf36698b63cd3d : SUCCESS in 48m 23s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/873389b747a04f43803300b2abe2d77b : SUCCESS in 43m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5183e55dbf427c412b64a964d7204acc213bf8f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-30 11:22:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"03cc1e075afc91ea37fe5a440838c137e3a58eed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-30 11:28:22.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3120cc0bb674bbda78f92f49fe03516\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf4a0bc9639b4441a441b0fb35d7184e : SUCCESS in 48s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"d6b5247dbb1acedd43b3dc8e7b1cac06519467c4","revisions":{"38a9c4cba56aea008265fd78e57307c40650a585":{"kind":"REWORK","_number":1,"created":"2024-03-12 10:18:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/912519/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/19/912519/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/19/912519/1"}}},"commit":{"parents":[{"commit":"383844437a1c27fe855fee541ae941c2c51d123a","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/383844437a1c27fe855fee541ae941c2c51d123a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-12 10:17:58.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-12 10:17:58.000000000","tz":60},"subject":"db: Replace use of LegacyEngineFacade","message":"db: Replace use of LegacyEngineFacade\n\nWe aren\u0027t passing context objects around, hence this looks more like\nironic\u0027s use of oslo.db than e.g. nova or cinder. Nothing complicated\nthough.\n\nChange-Id: I32a1417c001cdb6be0719baeb186f79609875f4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/38a9c4cba56aea008265fd78e57307c40650a585"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/38a9c4cba56aea008265fd78e57307c40650a585"}]},"branch":"refs/heads/master"},"c425e956dffdb3e3487b1dced155a3a37cdfb901":{"kind":"REWORK","_number":2,"created":"2024-03-12 11:10:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/912519/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/19/912519/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/19/912519/2"}}},"commit":{"parents":[{"commit":"383844437a1c27fe855fee541ae941c2c51d123a","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/383844437a1c27fe855fee541ae941c2c51d123a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-12 10:17:58.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-12 11:10:32.000000000","tz":60},"subject":"db: Replace use of LegacyEngineFacade","message":"db: Replace use of LegacyEngineFacade\n\nWe aren\u0027t passing context objects around, hence this looks more like\nironic\u0027s use of oslo.db than e.g. nova or cinder. Nothing complicated\nthough.\n\nChange-Id: I32a1417c001cdb6be0719baeb186f79609875f4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/c425e956dffdb3e3487b1dced155a3a37cdfb901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/c425e956dffdb3e3487b1dced155a3a37cdfb901"}]},"branch":"refs/heads/master"},"7b71351ceb554399001e2ef53c8df924114454ad":{"kind":"REWORK","_number":3,"created":"2024-03-12 13:39:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/912519/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/19/912519/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/19/912519/3"}}},"commit":{"parents":[{"commit":"383844437a1c27fe855fee541ae941c2c51d123a","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/383844437a1c27fe855fee541ae941c2c51d123a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-12 10:17:58.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-12 13:39:35.000000000","tz":60},"subject":"db: Replace use of LegacyEngineFacade","message":"db: Replace use of LegacyEngineFacade\n\nWe aren\u0027t passing context objects around, hence this looks more like\nironic\u0027s use of oslo.db than e.g. nova or cinder. Nothing complicated\nthough.\n\nChange-Id: I32a1417c001cdb6be0719baeb186f79609875f4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/7b71351ceb554399001e2ef53c8df924114454ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/7b71351ceb554399001e2ef53c8df924114454ad"}]},"branch":"refs/heads/master"},"11a189bdb9d9d7e03a8fb5847c665b4e8d8da880":{"kind":"REWORK","_number":4,"created":"2024-03-29 16:04:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/912519/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/19/912519/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/19/912519/4"}}},"commit":{"parents":[{"commit":"383844437a1c27fe855fee541ae941c2c51d123a","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/383844437a1c27fe855fee541ae941c2c51d123a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-12 10:17:58.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-29 16:03:57.000000000","tz":60},"subject":"db: Replace use of LegacyEngineFacade","message":"db: Replace use of LegacyEngineFacade\n\nWe aren\u0027t passing context objects around, hence this looks more like\nironic\u0027s use of oslo.db than e.g. nova or cinder. Nothing complicated\nthough.\n\nChange-Id: I32a1417c001cdb6be0719baeb186f79609875f4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/11a189bdb9d9d7e03a8fb5847c665b4e8d8da880"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/11a189bdb9d9d7e03a8fb5847c665b4e8d8da880"}]},"branch":"refs/heads/master"},"5c7c86403c9134d872966c4e61853a32bf607ed6":{"kind":"REWORK","_number":5,"created":"2024-03-29 16:19:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/912519/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/19/912519/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/19/912519/5"}}},"commit":{"parents":[{"commit":"383844437a1c27fe855fee541ae941c2c51d123a","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/383844437a1c27fe855fee541ae941c2c51d123a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-12 10:17:58.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-29 16:18:31.000000000","tz":60},"subject":"Replace use of LegacyEngineFacade and adopt to werkzeug\u003e3.0","message":"Replace use of LegacyEngineFacade and adopt to werkzeug\u003e3.0\n\nWe aren\u0027t passing context objects around, hence this looks more like\nironic\u0027s use of oslo.db than e.g. nova or cinder. Nothing complicated\nthough.\n\nThis also replaces deprecated werkzeug.http.parse_authorization_header\nwith Authorization.from_header\n\nChange-Id: I32a1417c001cdb6be0719baeb186f79609875f4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/5c7c86403c9134d872966c4e61853a32bf607ed6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/5c7c86403c9134d872966c4e61853a32bf607ed6"}]},"branch":"refs/heads/master"},"d2ff8ad5d1f76606356157c80d5cc50c40e458c4":{"kind":"REWORK","_number":6,"created":"2024-03-29 16:22:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/912519/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/19/912519/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/19/912519/6"}}},"commit":{"parents":[{"commit":"383844437a1c27fe855fee541ae941c2c51d123a","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/383844437a1c27fe855fee541ae941c2c51d123a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-12 10:17:58.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-29 16:22:33.000000000","tz":60},"subject":"Replace use of LegacyEngineFacade and adopt to werkzeug\u003e3.0","message":"Replace use of LegacyEngineFacade and adopt to werkzeug\u003e3.0\n\nWe aren\u0027t passing context objects around, hence this looks more like\nironic\u0027s use of oslo.db than e.g. nova or cinder. Nothing complicated\nthough.\n\nThis also replaces deprecated werkzeug.http.parse_authorization_header\nwith Authorization.from_header\n\nChange-Id: I32a1417c001cdb6be0719baeb186f79609875f4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/d2ff8ad5d1f76606356157c80d5cc50c40e458c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/d2ff8ad5d1f76606356157c80d5cc50c40e458c4"}]},"branch":"refs/heads/master"},"16444abb3c10eb0c132c6286db46aded25ab67ec":{"kind":"REWORK","_number":7,"created":"2024-03-29 19:19:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/912519/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/19/912519/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/19/912519/7"}}},"commit":{"parents":[{"commit":"383844437a1c27fe855fee541ae941c2c51d123a","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/383844437a1c27fe855fee541ae941c2c51d123a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-12 10:17:58.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-29 19:19:30.000000000","tz":60},"subject":"Replace use of LegacyEngineFacade and adopt to werkzeug\u003e3.0","message":"Replace use of LegacyEngineFacade and adopt to werkzeug\u003e3.0\n\nWe aren\u0027t passing context objects around, hence this looks more like\nironic\u0027s use of oslo.db than e.g. nova or cinder. Nothing complicated\nthough.\n\nThis also replaces deprecated werkzeug.http.parse_authorization_header\nwith Authorization.from_header\n\nChange-Id: I32a1417c001cdb6be0719baeb186f79609875f4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/16444abb3c10eb0c132c6286db46aded25ab67ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/16444abb3c10eb0c132c6286db46aded25ab67ec"}]},"branch":"refs/heads/master"},"d6b5247dbb1acedd43b3dc8e7b1cac06519467c4":{"kind":"REWORK","_number":8,"created":"2024-03-30 09:15:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/912519/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/19/912519/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/19/912519/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/19/912519/8"}}},"commit":{"parents":[{"commit":"383844437a1c27fe855fee541ae941c2c51d123a","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/383844437a1c27fe855fee541ae941c2c51d123a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-12 10:17:58.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-03-30 09:15:36.000000000","tz":0},"subject":"Replace use of LegacyEngineFacade and adopt to werkzeug\u003e3.0","message":"Replace use of LegacyEngineFacade and adopt to werkzeug\u003e3.0\n\nWe aren\u0027t passing context objects around, hence this looks more like\nironic\u0027s use of oslo.db than e.g. nova or cinder. Nothing complicated\nthough.\n\nThis also replaces deprecated werkzeug.http.parse_authorization_header\nwith Authorization.from_header\n\nChange-Id: I32a1417c001cdb6be0719baeb186f79609875f4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/d6b5247dbb1acedd43b3dc8e7b1cac06519467c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/d6b5247dbb1acedd43b3dc8e7b1cac06519467c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
