)]}'
{"id":"openstack%2Fvitrage~912996","triplet_id":"openstack%2Fvitrage~stable%2F2024.1~I9b44ea0d8465692849d310cb61b98b6d92e24a66","project":"openstack/vitrage","branch":"stable/2024.1","topic":"create-2024.1","hashtags":[],"change_id":"I9b44ea0d8465692849d310cb61b98b6d92e24a66","subject":"Update .gitreview for stable/2024.1","status":"ABANDONED","created":"2024-03-14 11:54:15.000000000","updated":"2025-10-31 13:54:19.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e4a1acb078f9fd45d3370d71275089d5e895c887","_number":912996,"virtual_id_number":912996,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-08 16:22:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-14 12:19:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-08 15:34:47.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"7152b04bba67ea032cdefd79e615f86af9a78912","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-03-14 11:54:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83cf6927fe1df96af87b77fc5b1cb30db7dad6b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 12:19:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/97b7edc82a6d40c29b2857c17fac09b2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53d7c7537bae476da26103157735fd39 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42a87087ac5146feb5ab59ba30eba357 : FAILURE in 12m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/311f61dedbab4b9ca93906d451cc48b5 : FAILURE in 9m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/deaa370ca0ef4b72b6543b789e589e9e : FAILURE in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da36c9a37dae4f258eddfb9268a1b7fa : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"952521f98c605c0ef66310badbcb1779d812a6ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-08-28 10:49:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_22816\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"3562efad3015242114e9ce999bb116fc44f27f23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 11:55:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b8f8b9ab1764e9fbcfd815e1311c7e6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aadb73dbf514b25b558165f08cd16d7 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fdb44d361b14d26990f686d2fc42e03 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c0da845ccb649f898d4e5e845a5dd0e : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/596500ccc79547d99aa81e5e1051949d : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a21d9bee25df4444b1c9fb67398673a5 : SUCCESS in 4m 10s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e779ce4362c145aeb52bcf7ff3684e79 : FAILURE in 38m 02s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/667f206110004f99b87ec577a30732f7 : FAILURE in 38m 27s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/8babe72db2fe4863925bb8c86e9c2e93 : SUCCESS in 58m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"75a21418211cfa667d8ebae6ff9d9d32686bf974","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-10-08 15:34:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8171646ee354c96e9c3685de45153b0deecabd46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 16:22:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/296c2111881c43aba2cf7bb4a958c335\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e3428306d5940cb86022398c003f7ae : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63badd2f9c19437699d037df29f558e7 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/628efd7996fb4aa4ba8dd3f1e41fe41b : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f988d43b2ba4ebbbbda18231c6e17c0 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a810187fca641efabaaa3470431f7dc : SUCCESS in 3m 42s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/90e82fd5c2a847f59a569fb82ea99de0 : FAILURE in 40m 29s (non-voting)\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/82cc96514ad74f35bb965289a71a46cc : FAILURE in 30m 29s (non-voting)\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/09d78dec495c4145913a735ae2cc7b96 : FAILURE in 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e4a1acb078f9fd45d3370d71275089d5e895c887","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-10-31 13:54:19.000000000","message":"Abandoned\n\nstable/2024.1 branch of openstack/vitrage is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/2024.1 if you want to further work on this patch.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"52045578353be250773051215477ed3fcd94ff02","revisions":{"03e0de45c58149d7108aa8b51bb01660a624ecd9":{"kind":"REWORK","_number":1,"created":"2024-03-14 11:54:15.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/96/912996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/96/912996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/96/912996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/96/912996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/96/912996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/96/912996/1"}}},"commit":{"parents":[{"commit":"383844437a1c27fe855fee541ae941c2c51d123a","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/383844437a1c27fe855fee541ae941c2c51d123a"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-14 11:54:14.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-14 11:54:14.000000000","tz":0},"subject":"Update .gitreview for stable/2024.1","message":"Update .gitreview for stable/2024.1\n\nChange-Id: I9b44ea0d8465692849d310cb61b98b6d92e24a66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/03e0de45c58149d7108aa8b51bb01660a624ecd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/03e0de45c58149d7108aa8b51bb01660a624ecd9"}]},"branch":"refs/heads/stable/2024.1"},"52045578353be250773051215477ed3fcd94ff02":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-08-28 10:49:16.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/96/912996/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/96/912996/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/96/912996/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/96/912996/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/96/912996/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/96/912996/2"}}},"commit":{"parents":[{"commit":"2e5eb67d508c06c3f217d970cadeff0739f6086c","subject":"Replace use of LegacyEngineFacade and adopt to werkzeug\u003e3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/2e5eb67d508c06c3f217d970cadeff0739f6086c"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-14 11:54:14.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-08-28 10:49:16.000000000","tz":0},"subject":"Update .gitreview for stable/2024.1","message":"Update .gitreview for stable/2024.1\n\nChange-Id: I9b44ea0d8465692849d310cb61b98b6d92e24a66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/52045578353be250773051215477ed3fcd94ff02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/52045578353be250773051215477ed3fcd94ff02"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
