)]}'
{"id":"openstack%2Fvitrage~918415","triplet_id":"openstack%2Fvitrage~master~I04e817f33fa3142bbc588c6f0dc6b968da75da3d","project":"openstack/vitrage","branch":"master","topic":"remove-pysnmp","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-08-27 06:59:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I04e817f33fa3142bbc588c6f0dc6b968da75da3d","subject":"Replace pysnmp","status":"ABANDONED","created":"2024-05-07 07:44:06.000000000","updated":"2024-08-27 06:59:52.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"67161cda208e6f4fcc935d3a92bc5e5db03864fc","_number":918415,"virtual_id_number":918415,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-08-26 15:17:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-08-26 12:48:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2024-08-26 12:48:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-07 09:00:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-26 12:48:37.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"5c096f1558fafa331cae8d3f5946ae0c3dc7060f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-07 07:44:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"460f9c9d29243ef387cca8c37321aba411cb81a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-07 07:45:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ebd6decdf9d69eab70638fc20ccc3b2f41e869d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 09:00:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2691a3b0595c40f0b0ccc0b501f72d21\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/16c05e7b3fc549f7afa80d9ba5214b6c : SUCCESS in 1m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32480bbf76034cb48cb50ebb965dfd1f : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b11b88330fc481880635b44f197aa88 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a9c48cf40244f019ac5ac2b89a37853 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/933e36df4027481b9ef264d8d8d9ac53 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b0f296b1131498a886bd01f908e57f9 : SUCCESS in 5m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f8c7a15f2b743ed872947431146f609 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae05d94dc1194d41adcbe7ba8669fa2e : SUCCESS in 3m 12s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5f7898dc428041b28f860919b387a393 : SUCCESS in 46m 12s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/ba851face9a7439f832063d8f4a6c790 : SUCCESS in 50m 34s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/ae0c63bab13c4c32803fb41b68b4dc4f : SUCCESS in 1h 04m 46s","accounts_in_message":[],"_revision_number":2},{"id":"441ea4c8008a3a57dc3ff3e3162c96fafe0d821b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-09 04:54:09.000000000","message":"Topic set to remove-pysnmp","accounts_in_message":[],"_revision_number":2},{"id":"0c1afabc0c407aa06f663d1cc833de10d66cc8ad","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-08-26 12:48:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"efd038c6a2e3e782b2b880ea255c9444d2a38475","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 12:49:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a3961a586eb9b2015b640416295c74559a199618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 13:37:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/09ac746d817948d6931c6d1a44eb2161\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e86a37dde28f4f3ab6612f501d3f563b : SUCCESS in 1m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a20e57e45af34dea8b6a5c9d96c30766 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61172b13fdcb46cab3a2a6c8e2c61663 : TIMED_OUT in 40m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d84ea4828ee48209458711d66564594 : TIMED_OUT in 40m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddc30a205f0d49a4bdb01190b569aa47 : TIMED_OUT in 41m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/91a78cac337542e7b06372d0f4ca19bf : TIMED_OUT in 40m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee11af703dd84beb9148e6d114d5974d : SUCCESS in 4m 50s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6a0b48b1b4fd45b4b94c452c8599d620 : FAILURE in 27m 30s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/f4a4b19aaedf4371be95295fc9a8c7a1 : FAILURE in 35m 53s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/7712566437064e63a6f278cd4a1c4f01 : FAILURE in 31m 57s","accounts_in_message":[],"_revision_number":2},{"id":"8bc4c3222b5baf07c50ba1f00ab2a44059dadb25","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-08-26 14:29:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eab3f78a0b0bffa09c9bba34f9ac67e887ca2bf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 15:17:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8cb6820aafa4b9fa042e020cef1ff9a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/19bf8d931003410399df89df10f254fa : SUCCESS in 1m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a35a67716e5c48998665cb6403453708 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/288482b4cde54cf299df888f1d874d90 : TIMED_OUT in 41m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bbe1b764e9d4d1d943288e7d7743c75 : TIMED_OUT in 40m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c0560fcaa684cb890fcd594bc57ef7e : TIMED_OUT in 41m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4dc974621ca54e7fbd5f4daca1faa349 : TIMED_OUT in 41m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df9f629de03042d1bacf13f49abd4081 : TIMED_OUT in 40m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bfb16ec24214ab0ba1e54176e0f8422 : SUCCESS in 3m 57s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5b7913e087b14058a6e7980d803550d1 : FAILURE in 31m 29s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/b113646e77bc49fa9d47ca33d5edd26d : FAILURE in 20m 13s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/fbb35af7fc5143dcae78454c0e943492 : FAILURE in 35m 59s","accounts_in_message":[],"_revision_number":2},{"id":"67161cda208e6f4fcc935d3a92bc5e5db03864fc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-08-27 06:59:52.000000000","message":"Abandoned\n\nIn favor of https://review.opendev.org/c/openstack/vitrage/+/927145","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1a9ae6bef21466d11ab4ab69774faf721c015263","revisions":{"7b392740c816c5cfca9cb065ac9f4a629f2bdca7":{"kind":"REWORK","_number":1,"created":"2024-05-07 07:44:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/918415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/15/918415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/15/918415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/15/918415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/15/918415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/15/918415/1"}}},"commit":{"parents":[{"commit":"3887910d6be0babba77b9fada37ca34e77e59e40","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/3887910d6be0babba77b9fada37ca34e77e59e40"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-07 07:40:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-07 07:43:55.000000000","tz":540},"subject":"Replace pysnmp","message":"Replace pysnmp\n\npysnmp[1] is no longer maintained. Replace it by its fork,\npysnmp-lextudio[2].\n\n[1] https://github.com/etingof/pysnmp\n[2] https://github.com/lextudio/pysnmp\n\nChange-Id: I04e817f33fa3142bbc588c6f0dc6b968da75da3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/7b392740c816c5cfca9cb065ac9f4a629f2bdca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/7b392740c816c5cfca9cb065ac9f4a629f2bdca7"}]},"branch":"refs/heads/master"},"1a9ae6bef21466d11ab4ab69774faf721c015263":{"kind":"REWORK","_number":2,"created":"2024-05-07 07:45:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/918415/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/15/918415/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/15/918415/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/15/918415/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/15/918415/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/15/918415/2"}}},"commit":{"parents":[{"commit":"3887910d6be0babba77b9fada37ca34e77e59e40","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/3887910d6be0babba77b9fada37ca34e77e59e40"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-07 07:40:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-07 07:45:17.000000000","tz":540},"subject":"Replace pysnmp","message":"Replace pysnmp\n\npysnmp[1] is no longer maintained. Replace it by its fork,\npysnmp-lextudio[2].\n\n[1] https://github.com/etingof/pysnmp\n[2] https://github.com/lextudio/pysnmp\n\nChange-Id: I04e817f33fa3142bbc588c6f0dc6b968da75da3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/1a9ae6bef21466d11ab4ab69774faf721c015263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/1a9ae6bef21466d11ab4ab69774faf721c015263"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
