)]}'
{"id":"openstack%2Fvitrage~927783","triplet_id":"openstack%2Fvitrage~master~I9005cd62a1ffcf997a88074f20cad137ccd4121c","project":"openstack/vitrage","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-09-03 15:42:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9005cd62a1ffcf997a88074f20cad137ccd4121c","subject":"Remove test helpers for python \u003c 2.7","status":"MERGED","created":"2024-09-03 07:23:19.000000000","updated":"2024-09-03 15:44:23.000000000","submitted":"2024-09-03 15:42:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927783","meta_rev_id":"eee198fe572a5b6b9d2856013a2618ed12978a24","_number":927783,"virtual_id_number":927783,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-03 15:42:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-09-03 14:11:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2024-09-03 14:11:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-03 08:33:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-03 14:11:57.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"1eeae682122661066ed2c5a80b716932a6168675","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-03 07:23:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b949153ab836b7c4130d46a39d4680884d5227a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 08:33:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8207aaf1d0e44b02a8017bf567247703\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6aa77d0c28b492a840706a89732a5a1 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01df15c017f84a6d9530012608226649 : FAILURE in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8edac022cec140efaf4dc81d2c3f8c7b : FAILURE in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/305e16f2659b4a1ead3d5492e2207efc : FAILURE in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44c8fa486b2b4c00a9072ec452dc7ff5 : FAILURE in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d8cb009731c4337b556827d11d8c6a3 : FAILURE in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a83e86fad74847fba0cba93935fe707a : SUCCESS in 4m 34s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5ed7931102e844b59efb69fba51d45de : SUCCESS in 47m 55s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/cd10fd4503eb46c7add720fb78c2b63d : SUCCESS in 48m 32s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/982f99c1423f48edb2ba32c4abfae4bf : SUCCESS in 1h 01m 59s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/99140f4f0bb0400dab22f1713c5be91d : SUCCESS in 1h 04m 57s","accounts_in_message":[],"_revision_number":1},{"id":"91eae862aec8b923005ca60c22e5fd2ffbe2fab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-03 09:02:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cbc41fa0c50729f33ba8aba736cc22a233057ca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 10:24:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/73c06fb98479497894920bc14cd23f79\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edae09d8f36d4213a433b512e0d0aeb5 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6a53528463145db904d6a536ee189b2 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f53fb09f96f41909b8d62708d333396 : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34672e05766d4cba966b79f2bd505702 : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/acaa18d20278426eb01f9c6914f3fbca : SUCCESS in 5m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b9faeb62d3b4bb4997761edb392d5e4 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/275c4083fddf441b9527db9237a024c1 : SUCCESS in 3m 19s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/897c72f960a64be98a66aa4b53d3c819 : SUCCESS in 55m 19s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/6e9aa2980d904265bad0df9fcc4d6d94 : SUCCESS in 52m 31s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/350bf2fce843486a9d7bbc3dbfcf19e0 : FAILURE in 56m 23s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/3119a85995a84f4399f0c5ebec706f86 : SUCCESS in 56m 33s","accounts_in_message":[],"_revision_number":2},{"id":"a5f753ed70d0d168496d98d32e7c352a5b7bebda","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-03 10:26:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"53f68bb0cff051134cc5d6ef98d18f9bd0359f18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 11:31:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c79a856d1c2640a9a917f4cbdf5dcd5e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47f90eff96074c6eb05d2768ca2e2627 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e637c0676936422e93fb6cbc8582569c : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a72461f3fe0c497bb0dea78fe2ff0be4 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04f5c8ccf9e043aa8ee8deb8cf4882a4 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f7b6b6be7ea4f5eb02078651a0fc89f : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f186e778378f499dbebad7ac7d3fba87 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3abedc88e8af4be9a7990d78a61ed465 : SUCCESS in 3m 13s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/602303bebff642bda53c6ab649d858cd : SUCCESS in 52m 01s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/11af7643a08e48fcae1fc32ac3a8a01e : SUCCESS in 50m 04s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/ef7e05cef3824e239a6340feacec7339 : SUCCESS in 59m 12s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/c6dd879bd6f540bea61d54eac8020534 : SUCCESS in 57m 37s","accounts_in_message":[],"_revision_number":2},{"id":"f9ee04f93aab09ec11dde7e22514c76d53327690","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-03 14:11:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5fd766b5e6fc75eb32ef60266d7b864849ce4ec4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 14:12:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"214cc1a9edbc310812f8954bfc9a3847de6dc0b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 15:42:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39115a8d489d4ceda413c0976360083d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/348d5b878a6b41c1ae5ed5d97126a00e : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65580895ffc140fb96aaf5f7b9f277b6 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/061c9fb2e9a9447085a2a0b0f17bf920 : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd24a9ad14314c47b473838bd6e03c8f : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc9955eb964c4e7a9407afa900418de4 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf7f8da18b244214bbc4bd16134a58ec : SUCCESS in 3m 40s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4b081edea4d148b4be269e06391e497a : SUCCESS in 52m 48s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/a525a2154a944f23bdd873383349fb9e : SUCCESS in 49m 54s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/dbd4415263ae4bdab7ff0380e958d2fc : SUCCESS in 1h 14m 05s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/0b91281fa02040f99b8c616400fcef38 : SUCCESS in 58m 45s","accounts_in_message":[],"_revision_number":2},{"id":"1329f055b3da18eb696f2cecefdf07d45648c935","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 15:42:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"eee198fe572a5b6b9d2856013a2618ed12978a24","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 15:44:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d2f0cd8c845455ba85717ce4682cfc9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c43160c64aaa461b863a724f29fdd848 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5702ae0cf8c55981740e590865b429aad151ed38","revisions":{"4ecea81a5649b0e84cf1707c9054807d42c016ec":{"kind":"REWORK","_number":1,"created":"2024-09-03 07:23:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/927783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/83/927783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/83/927783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/83/927783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/83/927783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/83/927783/1"}}},"commit":{"parents":[{"commit":"24dd49640d9e864f53f63f30e0298019ba6d6f35","subject":"Use sqlalchemy text for ensuring session existence","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/24dd49640d9e864f53f63f30e0298019ba6d6f35"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-03 07:22:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-03 07:22:38.000000000","tz":540},"subject":"Remove test helpers for python \u003c 2.7","message":"Remove test helpers for python \u003c 2.7\n\n... because wow the minimum python version supported is 3.8.\n\nChange-Id: I9005cd62a1ffcf997a88074f20cad137ccd4121c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/4ecea81a5649b0e84cf1707c9054807d42c016ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/4ecea81a5649b0e84cf1707c9054807d42c016ec"}]},"branch":"refs/heads/master"},"5702ae0cf8c55981740e590865b429aad151ed38":{"kind":"REWORK","_number":2,"created":"2024-09-03 09:02:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/927783/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/83/927783/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/83/927783/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/83/927783/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/83/927783/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/83/927783/2"}}},"commit":{"parents":[{"commit":"24dd49640d9e864f53f63f30e0298019ba6d6f35","subject":"Use sqlalchemy text for ensuring session existence","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/24dd49640d9e864f53f63f30e0298019ba6d6f35"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-03 07:22:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-03 09:02:47.000000000","tz":540},"subject":"Remove test helpers for python \u003c 2.7","message":"Remove test helpers for python \u003c 2.7\n\n... because wow the minimum python version supported is 3.8.\n\nChange-Id: I9005cd62a1ffcf997a88074f20cad137ccd4121c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/5702ae0cf8c55981740e590865b429aad151ed38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/5702ae0cf8c55981740e590865b429aad151ed38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
