)]}'
{"id":"openstack%2Fvitrage~935442","triplet_id":"openstack%2Fvitrage~master~Ic3871ee82a795aef6a123436ccf40af87a6c7f8a","project":"openstack/vitrage","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-16 22:21:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic3871ee82a795aef6a123436ccf40af87a6c7f8a","subject":"Add note about requirements lower bounds","status":"MERGED","created":"2024-11-16 15:38:00.000000000","updated":"2024-11-16 22:23:26.000000000","submitted":"2024-11-16 22:21:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935442","meta_rev_id":"db6841c4df0d6cab691ff65c6dba0eee97c49abb","_number":935442,"virtual_id_number":935442,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-16 22:21:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-11-16 20:58:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2024-11-16 20:58:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-16 16:43:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-16 20:58:39.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"c6c2de9414f3b7b79a140322f09cc84db90ed0b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-16 15:38:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80f9fc08e8aa7af8eb4fe46adcbee1f95a34cf37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 16:43:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f264f68df104a40bb088a4da45a85a1\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/42a55c19cbd84267bdb33917a77071f9 : SUCCESS in 1m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5a9979a1f13483586c9b884e4464ed7 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2b46dd0542a4797815fb01c8a914a67 : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/23af972734f041dda65de3cf569a53b9 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c773cfd91874f828a258c654990639c : SUCCESS in 4m 20s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/300ada99b97c4abb939c3e7b684ee840 : SUCCESS in 51m 47s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/df5a528bd13e4ce5ae0ebe9453724530 : SUCCESS in 52m 46s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/66db6606d7374c12b056e4dbed63c271 : SUCCESS in 59m 40s","accounts_in_message":[],"_revision_number":1},{"id":"bda1b84f0aac30ab2cf9734feb0faa0231f0f7b1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-16 20:58:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0449be10c7bab44d2fbe5914a368cd7884a68373","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 20:59:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5cab145235d18ed55693b4ee2af1ed6ea8399978","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 22:21:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c5e6586fa374adfb19d74b6b30b200c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/db44df9b93a84949a91d19762d0c5656 : SUCCESS in 1m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af17b9616efb41e8aea0bde470ed23a2 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6cffb0b8aa64fb09e3d7c6f89ee95ab : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d5e296b9b6b84f44b8d4e91d09030b85 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/954c0db7544143339a116b963402cf2b : SUCCESS in 3m 33s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/17f6e427df0946389a6cbfc66d8318bb : SUCCESS in 33m 20s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/ccb2ec81828241fbad808c38b54d4597 : SUCCESS in 52m 22s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/89de0204b8b542ac90663770135da560 : SUCCESS in 44m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4304ef6178c1861be2e8d3d3af98dd61fc6b6630","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 22:21:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"db6841c4df0d6cab691ff65c6dba0eee97c49abb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 22:23:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eea7b5e518f047d094a30dfa9ef5a7c5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2c808bb388a47f4af59567dde4b3124 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"822d421e519b87c20a61e1d028a8a79b19b6721c","revisions":{"822d421e519b87c20a61e1d028a8a79b19b6721c":{"kind":"REWORK","_number":1,"created":"2024-11-16 15:38:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/935442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/42/935442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/42/935442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/42/935442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/42/935442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/42/935442/1"}}},"commit":{"parents":[{"commit":"366540033f27e11e7e2b6e73df0819aba0bbe23d","subject":"Bump hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/366540033f27e11e7e2b6e73df0819aba0bbe23d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-16 15:37:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-16 15:37:36.000000000","tz":540},"subject":"Add note about requirements lower bounds","message":"Add note about requirements lower bounds\n\nSince lower constraints job was removed the bounds are not actually\ntested. Add a note to explain the bounds are maintained on best effort\nbasis.\n\nChange-Id: Ic3871ee82a795aef6a123436ccf40af87a6c7f8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/822d421e519b87c20a61e1d028a8a79b19b6721c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/822d421e519b87c20a61e1d028a8a79b19b6721c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
