)]}'
{"id":"openstack%2Fvitrage~940618","triplet_id":"openstack%2Fvitrage~master~I91a90f78d30082730f7ac1214ab7ece2f51da084","project":"openstack/vitrage","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-21 13:12:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I91a90f78d30082730f7ac1214ab7ece2f51da084","subject":"Replace oslo_utils.encodeutils.exception_to_unicode","status":"MERGED","created":"2025-02-03 12:01:05.000000000","updated":"2025-08-21 13:12:49.000000000","submitted":"2025-08-21 13:12:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940618","meta_rev_id":"9c6e2d78adfc1829a1d6c2f0687fbffa3e71b8f4","_number":940618,"virtual_id_number":940618,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-21 13:11:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-08-21 12:14:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2025-08-21 12:14:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-03 13:05:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-09 14:55:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-08-21 10:37:56.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2025-08-21 12:14:33.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"dd6c3233bc7aaf2520ace8388ada8787ed9693cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-03 12:01:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33699fd5b1a59959fefce3654bc8092069885250","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 13:05:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3dffc76de9214d3a8b098cb84136abba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f63207023d94f88b3afb4baa80d4570 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f835b06d0f57493cb4ffc38e436adf99 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e469c6f8240c4746b18364ddd4825227 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e377c6f126848598d320a3fc448020d : SUCCESS in 2m 36s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9e319ad7eb84448185020438f3c1b934 : FAILURE in 56m 53s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/14cfa753071c4822aa71c6ce3de3e2f9 : FAILURE in 57m 15s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/96fafc9b1e4a4f42bc93c872d5dc4dbf : FAILURE in 42m 51s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/e745257113b642a494abb24de16e6922 : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":1},{"id":"65d3664ae4926e4558f58d7f5499d3d79a11b2b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-09 14:28:34.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_9816\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"_revision_number":2},{"id":"0b26591e03d881dd1b6493ffec8a453ff222b215","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 15:24:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d77c0cdc7db146469f234e36938c7db7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62f6d5452ff64516b8960159aa02d6f8 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/724e8235720e40f0afa49e2338948505 : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/223aff1cfbb542ffa4409718e7985db5 : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98a33342ae674ecfadd1345f0e22b5c9 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df5d7de378af4b15a577a8bb44e8c545 : SUCCESS in 3m 31s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c67e6f99c655496d865f78eda3908b70 : FAILURE in 45m 04s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/63a99b03a9c247e4b5db0267a54ee099 : FAILURE in 48m 51s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/b0e181cc8c384f81a6cfb7b3a69bdf8e : FAILURE in 54m 50s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/f23e1488c80b4f819ba84e155f489d4a : SUCCESS in 50m 10s","accounts_in_message":[],"_revision_number":2},{"id":"50c0800adbc3ba0b41f691dc55109c65dd000772","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-21 10:37:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b274d1f017288457849a3690f8128654274bf3ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 11:40:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a9c80b720074f408fdafd19304130c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e1ef15de81743148b17b905f701ec6b : SUCCESS in 1m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ff812400c594960bc32929e4126ba3d : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/830e950d603d45da8baf3ac6c518a8df : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0763f484b4ed4be9b1b3ef5c990931f5 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36bf207d51474f4e816a4045296719dd : SUCCESS in 3m 43s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b1c5987321bf4490960ab5d450a135e8 : FAILURE in 48m 08s (non-voting)\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/482a02f650714b2286b352eab9025a61 : FAILURE in 47m 07s (non-voting)\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/a866dcac4f124616b3f66824e0fda39a : FAILURE in 57m 09s (non-voting)\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/b4b72ad828b641bd8097f7b03a1ff62b : SUCCESS in 53m 55s","accounts_in_message":[],"_revision_number":2},{"id":"3b207c62e717afcc5cb3637a5d03a6145306f27d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-21 12:14:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cc65392476f372196fe86450a2a4fc9e36341b25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 12:14:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b650d4eb9f43ac6d259d69e2262fcc34ea19985c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 13:11:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c734e3502eb4bad84dee59701da9845\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33a7e5cdadc14483abebd8ebf9dc7d46 : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/777b62aa2ba1487e84a9a875b167875a : SUCCESS in 6m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0cb34c361f084411ae109b431aa61514 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3c78242f4ff484da786cbac7213680c : SUCCESS in 6m 23s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/d12cf948531a49fd9f296542decc4249 : SUCCESS in 55m 00s","accounts_in_message":[],"_revision_number":2},{"id":"2d312373911f1dd26a53fc58f1a15203ec27b7b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 13:12:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9c6e2d78adfc1829a1d6c2f0687fbffa3e71b8f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 13:12:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90a5410fdc434ba99f83cfa2de582083\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aa2c3db6af344a193229bd5df24f424 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ae12210140a6fbbbb48240acc02691bb0006f289","revisions":{"6139f0f800804f3cd3a8f8a5ffb4712e2a8fcea7":{"kind":"REWORK","_number":1,"created":"2025-02-03 12:01:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/940618/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/18/940618/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/18/940618/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/18/940618/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/18/940618/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/18/940618/1"}}},"commit":{"parents":[{"commit":"880acca844c5a70609b95e962c0ec7e5c021c2c0","subject":"Merge \"Add note about requirements lower bounds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/880acca844c5a70609b95e962c0ec7e5c021c2c0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-03 12:00:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-03 12:00:55.000000000","tz":540},"subject":"Replace oslo_utils.encodeutils.exception_to_unicode","message":"Replace oslo_utils.encodeutils.exception_to_unicode\n\nThe function was deprecated because it is equivalent to str(ex) in\nPython 3.\n\nDepends-on: https://review.opendev.org/c/openstack/oslo.utils/+/938929\nChange-Id: I91a90f78d30082730f7ac1214ab7ece2f51da084\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/6139f0f800804f3cd3a8f8a5ffb4712e2a8fcea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/6139f0f800804f3cd3a8f8a5ffb4712e2a8fcea7"}]},"branch":"refs/heads/master"},"ae12210140a6fbbbb48240acc02691bb0006f289":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-09 14:28:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/18/940618/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/18/940618/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/18/940618/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/18/940618/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/18/940618/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/18/940618/2"}}},"commit":{"parents":[{"commit":"aafc997e3feada114109c752a9fd6d945607773b","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/aafc997e3feada114109c752a9fd6d945607773b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-03 12:00:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-09 14:28:34.000000000","tz":0},"subject":"Replace oslo_utils.encodeutils.exception_to_unicode","message":"Replace oslo_utils.encodeutils.exception_to_unicode\n\nThe function was deprecated because it is equivalent to str(ex) in\nPython 3.\n\nDepends-on: https://review.opendev.org/c/openstack/oslo.utils/+/938929\nChange-Id: I91a90f78d30082730f7ac1214ab7ece2f51da084\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/ae12210140a6fbbbb48240acc02691bb0006f289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/ae12210140a6fbbbb48240acc02691bb0006f289"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
