)]}'
{"id":"openstack%2Fvitrage~957407","triplet_id":"openstack%2Fvitrage~master~I16e57ec45696afb597ba4f4f4f85ff10256a09e9","project":"openstack/vitrage","branch":"master","topic":"release-health-check-flamingo","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-08-21 16:26:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I16e57ec45696afb597ba4f4f4f85ff10256a09e9","subject":"DNM: gate health test","status":"ABANDONED","created":"2025-08-14 09:52:59.000000000","updated":"2025-08-21 16:26:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"365923bad96006c8cecac2643047181a176b7294","_number":957407,"virtual_id_number":957407,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-21 14:11:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-14 12:39:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e6c97c0aaefe72dabde3bc3c82562541bd583f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-14 09:52:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ce12eba18bd09a3746f07dd969ae8f0f75b8072","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 12:39:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da25a24d201e4257a3f795e457d2d8ee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbeb1dc533454dd18c8d36ad58516191 : SUCCESS in 1m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b595c7d58039422cadd47456c0e4b2e7 : FAILURE in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e5f84e8a6af3445a87bf87daf04401ab : FAILURE in 2m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac1a2c3f3d6c4e4bafbf2ee3a7733efe : FAILURE in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f4c64384d6d47e6a0e06f2a468dafd4 : SUCCESS in 4m 08s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c5960f63c4c44511870286717df8e018 : FAILURE in 21m 31s\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/7a2f6bc60f0c43018326d66f471ece4b : FAILURE in 22m 48s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/c494dd06d11244b28a0398dc6cb2b1b2 : FAILURE in 14m 28s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/ee3ce801c9e642efae0c3608c315c147 : SUCCESS in 26m 07s","accounts_in_message":[],"_revision_number":1},{"id":"12d28d194856b20dcc631154c727422181caf881","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-08-21 13:15:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_17685\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"_revision_number":2},{"id":"167554cb8b6d99f30640caf00478079f2c0badac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 14:11:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/240bdab765a64e00b58adc61275700d5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1172f2d2f85e4285a45b36ec562392f4 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d00da7ec781d47a6b45138f4fa70848f : SUCCESS in 4m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d29f18a37b6347079ef546252d846cce : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f38f4e8a66d3483a8f3e5a63f76d102a : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b743aca9d724eceba3ef6eeedf0d0c8 : SUCCESS in 4m 52s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/666d44f88c7047c4acdd6f28de01282e : FAILURE in 49m 18s (non-voting)\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/ef18a19c619148749049ccb47fa5b63f : FAILURE in 47m 20s (non-voting)\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/10afe06f429b486eb65e5080d0df5235 : FAILURE in 42m 14s (non-voting)\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/9eeec4c431ed4a8c9cf338c7245ee5c0 : SUCCESS in 50m 23s","accounts_in_message":[],"_revision_number":2},{"id":"365923bad96006c8cecac2643047181a176b7294","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-21 16:26:24.000000000","message":"Abandoned\n\nthanks Dmitriy for fixing the unit test jobs! i see that tempest based jobs are set now to non-voting, but at least the gate is functional. abandoning this patch now.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fe18279a363fb2004ce984523520eb06e1bb75a7","revisions":{"dec800e591a029a38ba75957029c93de13dda5a8":{"kind":"REWORK","_number":1,"created":"2025-08-14 09:52:59.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/07/957407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/07/957407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/07/957407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/07/957407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/07/957407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/07/957407/1"}}},"commit":{"parents":[{"commit":"aafc997e3feada114109c752a9fd6d945607773b","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/aafc997e3feada114109c752a9fd6d945607773b"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-08-14 09:52:49.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-08-14 09:52:58.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: I16e57ec45696afb597ba4f4f4f85ff10256a09e9\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/dec800e591a029a38ba75957029c93de13dda5a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/dec800e591a029a38ba75957029c93de13dda5a8"}]},"branch":"refs/heads/master"},"fe18279a363fb2004ce984523520eb06e1bb75a7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-08-21 13:15:11.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/07/957407/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/07/957407/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/07/957407/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/07/957407/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/07/957407/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/07/957407/2"}}},"commit":{"parents":[{"commit":"28c018eb6393836690e8f8a554a87c173a62ae3d","subject":"Merge \"Update license defenition in favor of SPDX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/28c018eb6393836690e8f8a554a87c173a62ae3d"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-08-14 09:52:49.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-08-21 13:15:11.000000000","tz":0},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: I16e57ec45696afb597ba4f4f4f85ff10256a09e9\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/fe18279a363fb2004ce984523520eb06e1bb75a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/fe18279a363fb2004ce984523520eb06e1bb75a7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
