)]}'
{"id":"openstack%2Fvitrage~959294","triplet_id":"openstack%2Fvitrage~master~Ia8c2ed6836a6ecaa1d947382d3ad912f2fd4d17f","project":"openstack/vitrage","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-03 12:49:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia8c2ed6836a6ecaa1d947382d3ad912f2fd4d17f","subject":"sqlalchemy: Use built-in declarative","status":"MERGED","created":"2025-09-03 09:51:39.000000000","updated":"2025-09-03 12:51:06.000000000","submitted":"2025-09-03 12:49:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959294","meta_rev_id":"b0bbf42a07472e1bd4c6a72e99981a825ac87a70","_number":959294,"virtual_id_number":959294,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-03 12:49:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-09-03 11:52:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2025-09-03 11:52:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-03 10:53:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-03 11:52:29.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"c7dd64923685b402e469e360598d4ae9ca12a83d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-03 09:51:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b2dd7530a3cd6bbb6338032f219e89897b60f8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 10:53:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04bbee2591cd4e09b0bbb4618014f3ce\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d610352f56fd4fd08cf289c1da56dfe6 : SUCCESS in 1m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14e1c61180564fcf91186cdec8fb1e32 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf2e623822a54aa6add0d52b679bd239 : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fdddf399d841418694449cba83739dbe : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/08b64dad33bb45759c8b10fe16587931 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f905a7c79afd496c823dba67646bd575 : SUCCESS in 3m 42s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/df6bbffa1c784e0db5b3c4a84b139a81 : FAILURE in 47m 09s (non-voting)\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/90052f66f1dd4729a2b829cbb381ddae : FAILURE in 40m 37s (non-voting)\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/2a5083e4a0314d2eb42bdffa6a658d9e : FAILURE in 57m 09s (non-voting)\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/411acf46ac4946b381efdc11656babfe : SUCCESS in 46m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ec2752e55042d2668a1c11d73cacc7f76b785b39","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-03 11:52:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"26962a8b010e52ed8948fb3f7a9573ad58afad6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 11:53:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69554f41a5bd2aa8522f66cf570921c10d7e791b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 12:49:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccc5de7c82bc475fbb73d0f99c369c13\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/23634fc0cdb546819ef33484beefd2aa : SUCCESS in 1m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31793e954b6d497a96977d38d167d872 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f347ca6d50954d288f4a12584b56d754 : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e36120f0e424621a6a47c3e4557d8c9 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45b21fd1e84749a58a8428043e57effd : SUCCESS in 3m 43s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/27eb85f6732c42e9ac8c2e8fd95faabd : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":1},{"id":"852ce48b8dfcb17c647cce478ba32ec71622c26f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 12:49:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b0bbf42a07472e1bd4c6a72e99981a825ac87a70","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 12:51:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/037092af833e4362bb4c7d9f6cf33cf0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b43e557666e42fdbcfb2441af25b8fa : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae9e248dda8c065ab7d770c64661b318f1a7a09e","revisions":{"ae9e248dda8c065ab7d770c64661b318f1a7a09e":{"kind":"REWORK","_number":1,"created":"2025-09-03 09:51:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/959294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/94/959294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/94/959294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/94/959294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/94/959294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/94/959294/1"}}},"commit":{"parents":[{"commit":"4a668e342a2f782538c234c84b3902d60456a48c","subject":"Replace pysnmp-lextudio with pysnmp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/4a668e342a2f782538c234c84b3902d60456a48c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-03 09:51:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-03 09:51:30.000000000","tz":540},"subject":"sqlalchemy: Use built-in declarative","message":"sqlalchemy: Use built-in declarative\n\nsqlalchemy.ext.declarative was deprecated in sqlalchemy 1.4.0, due to\nthe built-in implementations[1].\n\n[1] https://github.com/sqlalchemy/sqlalchemy/commit/450f5c0d6519a439f40\n\nChange-Id: Ia8c2ed6836a6ecaa1d947382d3ad912f2fd4d17f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/ae9e248dda8c065ab7d770c64661b318f1a7a09e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/ae9e248dda8c065ab7d770c64661b318f1a7a09e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
