)]}'
{"id":"openstack%2Fvitrage~962660","triplet_id":"openstack%2Fvitrage~stable%2F2024.2~I2df24cfe627453fd70c91dfc366176b3f5444694","project":"openstack/vitrage","branch":"stable/2024.2","hashtags":[],"change_id":"I2df24cfe627453fd70c91dfc366176b3f5444694","subject":"Set tempest jobs to NV","status":"MERGED","created":"2025-10-01 11:22:31.000000000","updated":"2025-10-01 13:08:07.000000000","submitted":"2025-10-01 13:06:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962660","cherry_pick_of_change":962654,"cherry_pick_of_patch_set":1,"meta_rev_id":"164ce46d70d58b0e332a077c6a8a3431c68cf6af","_number":962660,"virtual_id_number":962660,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-01 13:06:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-10-01 12:35:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2025-10-01 12:35:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-01 12:19:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"feaa9a41d740af49e1abc8af24e502f9d1618088","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-10-01 11:22:31.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.","accounts_in_message":[],"_revision_number":1},{"id":"53771182a4ec4d066d9a045c232f0d6a9800c24d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 12:19:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79a8ed9bd670413f903f748e308f7682\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0d42f7298ce4bc493f95cdc0e32c31c : SUCCESS in 1m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/741ed9f6b4a04a2eae41e4dcd2755a09 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5370ecf58354f3082760a25c7f6203b : SUCCESS in 7m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13d74ae6e01648e0a6f6906b29569935 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/21c0aab8ed9d4995b47db639fc6e9926 : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a46fed05e55244278f6189bc807b595d : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ef1378c07ea44b5bbe673ef670e3ce4 : SUCCESS in 6m 54s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6ad6574f54944f5392bfce45df9fc471 : FAILURE in 44m 09s (non-voting)\n- vitrage-tempest-plugin-api-ipv6-only https://zuul.opendev.org/t/openstack/build/208bbae3ec954823b1a9ef0a67ee887c : FAILURE in 43m 42s (non-voting)\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/a85dd0eac3ce49b2b5aa8b37a049f66d : FAILURE in 51m 01s (non-voting)\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/8f3ab2785b3344aebcb5882169f474f1 : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":1},{"id":"79b64e667e1238d88dfc17985b39f32475376c95","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-10-01 12:35:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"89867f89223bdd62f98fed2eb7dcf9f2fd6bf30c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 12:36:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4dd0e549871a3c7f9ecd4b0e695b13392b004f2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 13:06:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d98992bdd8b04a929ea9868e64e165ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5927998630424a28a50ec43fa5b50c2d : SUCCESS in 2m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a039eed882f3448aa622f811ba6a3da4 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab76e82c39484f4f8d613fc165b5f7d9 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/479cffd52ea34452963440818d1b5788 : SUCCESS in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/027d7688adb14a1c83458d5e85466c6b : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5e2833fb9a44a8fac192d7738e70834 : SUCCESS in 2m 19s\n- vitrage-grenade https://zuul.opendev.org/t/openstack/build/fba1458481ad4f78ad7219547b9d41f6 : SUCCESS in 28m 45s","accounts_in_message":[],"_revision_number":1},{"id":"df044fd3a9331c7514c9521cb9ad3aaf59dece85","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 13:06:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"164ce46d70d58b0e332a077c6a8a3431c68cf6af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 13:08:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0f7140c90684f61894cbaae7c932237\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c5a5c7197d4d368e3ae550c870937d : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f29c3dffefbef38d285de109c554c387ff2ae18a","revisions":{"f29c3dffefbef38d285de109c554c387ff2ae18a":{"kind":"REWORK","_number":1,"created":"2025-10-01 11:22:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/60/962660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/60/962660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/60/962660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/60/962660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/60/962660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/60/962660/1"}}},"commit":{"parents":[{"commit":"5abde2e3c4ab5a285ea9d3a13c70c66319da9e05","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/5abde2e3c4ab5a285ea9d3a13c70c66319da9e05"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy@adria-cloud.com","date":"2025-10-01 09:37:17.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy@adria-cloud.com","date":"2025-10-01 11:22:31.000000000","tz":0},"subject":"Set tempest jobs to NV","message":"Set tempest jobs to NV\n\nTempest started failing for all branches at some point.\nAs of now failure reason is unknown, though it is important to\nunblock\nCI to land patches and figure out failure reasons later.\n\nChange-Id: I2df24cfe627453fd70c91dfc366176b3f5444694\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy@adria-cloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/f29c3dffefbef38d285de109c554c387ff2ae18a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/f29c3dffefbef38d285de109c554c387ff2ae18a"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
