)]}'
{"id":"openstack%2Fwatcher-dashboard~448571","triplet_id":"openstack%2Fwatcher-dashboard~master~I3753ea7d596086f3b0f6ec6c7fd8aaf9d34658f7","project":"openstack/watcher-dashboard","branch":"master","topic":"bp/workload-fingerprinting-fingerprint-visualization","attention_set":{},"removed_from_attention_set":{"24321":{"account":{"_account_id":24321,"name":"praveenraj","email":"praveenrajr9@gmail.com","username":"praveenrajr9"},"last_update":"2025-07-21 18:44:47.000000000","reason":"Change was abandoned"},"3":{"account":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"last_update":"2025-07-21 18:44:47.000000000","reason":"Change was abandoned"},"22714":{"account":{"_account_id":22714,"name":"Ananth Narayan S","email":"ananth.narayan@gmail.com","username":"ananth"},"last_update":"2025-07-21 18:44:47.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3753ea7d596086f3b0f6ec6c7fd8aaf9d34658f7","subject":"[WIP] Support visualization of workload fingerprint","status":"ABANDONED","created":"2017-03-22 13:28:19.000000000","updated":"2025-07-21 18:44:47.000000000","total_comment_count":7,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"0c9eec2a5e4d5e0f6550b7d6fdd9a5a26a7d1d97","_number":448571,"virtual_id_number":448571,"owner":{"_account_id":22714,"name":"Ananth Narayan S","email":"ananth.narayan@gmail.com","username":"ananth"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24321,"name":"praveenraj","email":"praveenrajr9@gmail.com","username":"praveenrajr9"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":22714,"name":"Ananth Narayan S","email":"ananth.narayan@gmail.com","username":"ananth"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-01-09 23:33:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19055,"name":"Alexander Chadin","email":"joker946@gmail.com","username":"joker946"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24321,"name":"praveenraj","email":"praveenrajr9@gmail.com","username":"praveenrajr9"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22714,"name":"Ananth Narayan S","email":"ananth.narayan@gmail.com","username":"ananth"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19055,"name":"Alexander Chadin","email":"joker946@gmail.com","username":"joker946"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24321,"name":"praveenraj","email":"praveenrajr9@gmail.com","username":"praveenrajr9"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":22714,"name":"Ananth Narayan S","email":"ananth.narayan@gmail.com","username":"ananth"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19055,"name":"Alexander Chadin","email":"joker946@gmail.com","username":"joker946"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":24321,"name":"praveenraj","email":"praveenrajr9@gmail.com","username":"praveenrajr9"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22714,"name":"Ananth Narayan S","email":"ananth.narayan@gmail.com","username":"ananth"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":19055,"name":"Alexander Chadin","email":"joker946@gmail.com","username":"joker946"}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}],"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":19055,"name":"Alexander Chadin","email":"joker946@gmail.com","username":"joker946"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22714,"name":"Ananth Narayan S","email":"ananth.narayan@gmail.com","username":"ananth"},{"_account_id":24321,"name":"praveenraj","email":"praveenrajr9@gmail.com","username":"praveenrajr9"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-03-23 04:47:45.000000000","updated_by":{"_account_id":24321,"name":"praveenraj","email":"praveenrajr9@gmail.com","username":"praveenrajr9"},"reviewer":{"_account_id":24321,"name":"praveenraj","email":"praveenrajr9@gmail.com","username":"praveenrajr9"},"state":"REVIEWER"},{"updated":"2017-03-23 04:52:38.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-04-12 12:19:22.000000000","updated_by":{"_account_id":19055,"name":"Alexander Chadin","email":"joker946@gmail.com","username":"joker946"},"reviewer":{"_account_id":19055,"name":"Alexander Chadin","email":"joker946@gmail.com","username":"joker946"},"state":"REVIEWER"},{"updated":"2025-01-09 23:33:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-10 16:05:15.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"CC"}],"messages":[{"id":"ec7e6892f029ea9c0143cf12a7aaf7158a0f2430","author":{"_account_id":22714,"name":"Ananth Narayan S","email":"ananth.narayan@gmail.com","username":"ananth"},"date":"2017-03-22 13:28:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad9434eb7d4e4b5637de9e4d25ed407e6b886f4e","author":{"_account_id":22714,"name":"Ananth Narayan S","email":"ananth.narayan@gmail.com","username":"ananth"},"date":"2017-03-22 13:29:19.000000000","message":"Patch Set 1: Code-Review-1\n\nThis patch is WIP.","accounts_in_message":[],"_revision_number":1},{"id":"27b09f43ec47f61ec0ab28ad289c97e62b2323b2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-22 13:35:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-watcher-dashboard-docs-ubuntu-xenial http://docs-draft.openstack.org/71/448571/1/check/gate-watcher-dashboard-docs-ubuntu-xenial/158d109//doc/build/html/ : SUCCESS in 4m 33s\n- gate-watcher-dashboard-pep8-ubuntu-xenial http://logs.openstack.org/71/448571/1/check/gate-watcher-dashboard-pep8-ubuntu-xenial/dc60902/ : FAILURE in 4m 48s\n- gate-watcher-dashboard-python27-ubuntu-xenial http://logs.openstack.org/71/448571/1/check/gate-watcher-dashboard-python27-ubuntu-xenial/3a957c6/ : FAILURE in 2m 39s\n- gate-watcher-dashboard-python35 http://logs.openstack.org/71/448571/1/check/gate-watcher-dashboard-python35/502da59/ : FAILURE in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d1ee590ad79593c43d46044161b84694d035e05d","author":{"_account_id":24321,"name":"praveenraj","email":"praveenrajr9@gmail.com","username":"praveenrajr9"},"date":"2017-03-23 04:47:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"29d1ab74d364b398641de6d49bb8d63b568927a0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-23 04:52:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-watcher-dashboard-docs-ubuntu-xenial http://docs-draft.openstack.org/71/448571/2/check/gate-watcher-dashboard-docs-ubuntu-xenial/8576e7b//doc/build/html/ : SUCCESS in 4m 10s\n- gate-watcher-dashboard-pep8-ubuntu-xenial http://logs.openstack.org/71/448571/2/check/gate-watcher-dashboard-pep8-ubuntu-xenial/0770b55/ : SUCCESS in 4m 41s\n- gate-watcher-dashboard-python27-ubuntu-xenial http://logs.openstack.org/71/448571/2/check/gate-watcher-dashboard-python27-ubuntu-xenial/f7e6c3a/ : SUCCESS in 3m 46s\n- gate-watcher-dashboard-python35 http://logs.openstack.org/71/448571/2/check/gate-watcher-dashboard-python35/2e6bbdf/ : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"b40dccd6180e4324d91b70877b70349ab16c88bd","author":{"_account_id":19055,"name":"Alexander Chadin","email":"joker946@gmail.com","username":"joker946"},"date":"2017-04-12 12:19:22.000000000","message":"Patch Set 2:\n\nAnanth,\n\nIs it still WiP?","accounts_in_message":[],"_revision_number":2},{"id":"bee4961bb75c5c854866d8436702ad60a86df56e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24321,"name":"praveenraj","email":"praveenrajr9@gmail.com","username":"praveenrajr9"},"real_author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-09 23:20:06.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_24321\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":24321,"name":"praveenraj","email":"praveenrajr9@gmail.com","username":"praveenrajr9"}],"_revision_number":3},{"id":"d419497af6d652d84d8767f36ab30267efecb59f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 23:33:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bd71f4ea39c4bebb98546c6f3de597f\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2721fc6481f64bc4a870467da63aef6c : FAILURE in 5m 16s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/0b51c86076c44a97bd4f6ac0ba13dadf : FAILURE in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bf9b7d3cbd946bba7e5e2b12c959382 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/977b0b300834470489586d2bcdd7abc9 : FAILURE in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/850b7d32f9174ea1bf1ff4f08fcb9eb0 : FAILURE in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aee5095315b42de92a81f0bf681566e : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":3},{"id":"e4ae41d8533a20b9a0b70eae2601b167cb7b8226","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-02-10 16:05:15.000000000","message":"Patch Set 3:\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0c9eec2a5e4d5e0f6550b7d6fdd9a5a26a7d1d97","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-21 18:44:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6bcff15232a150913a7ae5d5941d3cd87b94c7f8","revisions":{"b60c8c5796fa2bbec49ae2e3c32bd917671dcc0f":{"kind":"REWORK","_number":1,"created":"2017-03-22 13:28:19.000000000","uploader":{"_account_id":22714,"name":"Ananth Narayan S","email":"ananth.narayan@gmail.com","username":"ananth"},"ref":"refs/changes/71/448571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/71/448571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/448571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/448571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/448571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/448571/1"}}},"commit":{"parents":[{"commit":"0a2e5973dee8edd945a4631c6a48ca10398e9942","subject":"Merge \"Remove the extra required\u003dTrue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/0a2e5973dee8edd945a4631c6a48ca10398e9942"}]}],"author":{"name":"Ananth Narayan S","email":"ananth.narayan@gmail.com","date":"2017-03-22 13:05:34.000000000","tz":330},"committer":{"name":"Ananth Narayan S","email":"ananth.narayan@gmail.com","date":"2017-03-22 13:08:50.000000000","tz":330},"subject":"[WIP] Support visualization of workload fingerprint","message":"[WIP] Support visualization of workload fingerprint\n\nThis patch adds support for visualizing workload fingerprints which are\ngenerated by the workload fingerprint generation strategy.\n\nChange-Id: I3753ea7d596086f3b0f6ec6c7fd8aaf9d34658f7\nImplements:bp workload-fingerprinting-fingerprint-visualization\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/b60c8c5796fa2bbec49ae2e3c32bd917671dcc0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/b60c8c5796fa2bbec49ae2e3c32bd917671dcc0f"}]},"branch":"refs/heads/master"},"480ab83261ad6710c9b18cac965163a490cee0a3":{"kind":"REWORK","_number":2,"created":"2017-03-23 04:47:45.000000000","uploader":{"_account_id":24321,"name":"praveenraj","email":"praveenrajr9@gmail.com","username":"praveenrajr9"},"ref":"refs/changes/71/448571/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/71/448571/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/448571/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/448571/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/448571/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/448571/2"}}},"commit":{"parents":[{"commit":"0a2e5973dee8edd945a4631c6a48ca10398e9942","subject":"Merge \"Remove the extra required\u003dTrue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/0a2e5973dee8edd945a4631c6a48ca10398e9942"}]}],"author":{"name":"Ananth Narayan S","email":"ananth.narayan@gmail.com","date":"2017-03-22 13:05:34.000000000","tz":330},"committer":{"name":"praveenraj","email":"praveenrajr9@gmail.com","date":"2017-03-23 04:56:24.000000000","tz":330},"subject":"[WIP] Support visualization of workload fingerprint","message":"[WIP] Support visualization of workload fingerprint\n\nThis patch adds support for visualizing workload fingerprints which are\ngenerated by the workload fingerprint generation strategy.\n\nChange-Id: I3753ea7d596086f3b0f6ec6c7fd8aaf9d34658f7\nImplements:bp workload-fingerprinting-fingerprint-visualization\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/480ab83261ad6710c9b18cac965163a490cee0a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/480ab83261ad6710c9b18cac965163a490cee0a3"}]},"branch":"refs/heads/master"},"6bcff15232a150913a7ae5d5941d3cd87b94c7f8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-01-09 23:20:06.000000000","uploader":{"_account_id":24321,"name":"praveenraj","email":"praveenrajr9@gmail.com","username":"praveenrajr9"},"real_uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/71/448571/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/71/448571/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/448571/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/448571/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/448571/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/448571/3"}}},"commit":{"parents":[{"commit":"793bf9f51b7c0605916e64e37d4a7d8de7326f78","subject":"Update python versions, drop py3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/793bf9f51b7c0605916e64e37d4a7d8de7326f78"}]}],"author":{"name":"Ananth Narayan S","email":"ananth.narayan@gmail.com","date":"2017-03-22 13:05:34.000000000","tz":330},"committer":{"name":"praveenraj","email":"praveenrajr9@gmail.com","date":"2025-01-09 23:20:06.000000000","tz":0},"subject":"[WIP] Support visualization of workload fingerprint","message":"[WIP] Support visualization of workload fingerprint\n\nThis patch adds support for visualizing workload fingerprints which are\ngenerated by the workload fingerprint generation strategy.\n\nChange-Id: I3753ea7d596086f3b0f6ec6c7fd8aaf9d34658f7\nImplements:bp workload-fingerprinting-fingerprint-visualization\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/6bcff15232a150913a7ae5d5941d3cd87b94c7f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/6bcff15232a150913a7ae5d5941d3cd87b94c7f8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
