)]}'
{"id":"openstack%2Fwatcher-dashboard~682578","triplet_id":"openstack%2Fwatcher-dashboard~master~I9691132f542242716992d2f842f1261b7cd30389","project":"openstack/watcher-dashboard","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"I9691132f542242716992d2f842f1261b7cd30389","subject":"Build pdf docs","status":"MERGED","created":"2019-09-17 08:26:40.000000000","updated":"2019-09-18 01:59:51.000000000","submitted":"2019-09-18 01:57:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682578-1568771833498-795c03e7","meta_rev_id":"4cb7a652780f8610220dc78bd8802684b5c2fb76","_number":682578,"virtual_id_number":682578,"owner":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"value":2,"date":"2019-09-18 01:57:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"all":[{"value":2,"date":"2019-09-18 01:17:15.000000000","_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"all":[{"value":1,"date":"2019-09-18 01:17:15.000000000","_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 01:57:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5a8b8e2d2d95df5c31736bac80d31ee86009a460","author":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"date":"2019-09-17 08:26:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85772e1006367eae67bc408759c199f570ea9a6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 09:14:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/71391ca29e7c4119b846216754b12587 : SUCCESS in 2m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de4d6e3c216549eab37e52c36aa31fc5 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a231806a4d740d7a27968e04920a222 : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a98efc5694446618d4309b56affbc74 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c605c1bcdaf1477db28686409a865c4c : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/551cb14d00824374b4997ae6fe8a9070 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a351fdc8d79244388b6b9950638adbe1 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7c58199191a4d5b9620095df3a886ec : SUCCESS in 2m 55s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/48b2417a8325415995f3c427c74d102a : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"5361b1a4b0829a8f7b52dbc9a2e34b9f8b83d3b4","author":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"date":"2019-09-18 01:17:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80e79dc9f38d18ca58965598eb549dcbc5863caf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 01:17:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3c24a94cc0ed0cb360b285f17a69a67d9d39d33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 01:57:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/850ce2ce2c7647e3a2d836218a8029bd : SUCCESS in 2m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3697c55702e4d02b78f7aacde84b7c7 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e705c12a72d24d3f9854acdfe00c1d82 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2f03134ddda84ad8ae35af6d52b610af : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0349dfb163c487098df2b5853f755d6 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a669ec4563474654b1456c049a1890ca : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6f2ca1d9e004255828dc7545bc6def7 : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a8252da6f734ec3be2cbac1145052b7 : SUCCESS in 3m 17s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/debf86a6cfb74d4db0786a73cb78c267 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"77aba3569bd2484aa6fd3c234b49995e08df7690","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 01:57:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5df9242fd3d07c70db67aec1ff687c60b28dc41c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 01:59:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f8b24bb7a2e46b08d8a1d75d9ad301f : SUCCESS in 2m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1925bbb9613f4301b5e5d7723ed0ff1a : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0405ceff6d3499e8f3786df9d678ffaab6ad95f","revisions":{"e0405ceff6d3499e8f3786df9d678ffaab6ad95f":{"kind":"REWORK","_number":1,"created":"2019-09-17 08:26:40.000000000","uploader":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"ref":"refs/changes/78/682578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/78/682578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/78/682578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/78/682578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/78/682578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/78/682578/1"}}},"commit":{"parents":[{"commit":"2e3c8223187532f0840c8acd9c9e9c09ff3e90db","subject":"Merge \"Add blueprints link for README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/2e3c8223187532f0840c8acd9c9e9c09ff3e90db"}]}],"author":{"name":"licanwei","email":"li.canwei2@zte.com.cn","date":"2019-09-17 08:26:17.000000000","tz":-420},"committer":{"name":"licanwei","email":"li.canwei2@zte.com.cn","date":"2019-09-17 08:26:37.000000000","tz":-420},"subject":"Build pdf docs","message":"Build pdf docs\n\n Add a new pdf-docs environment to enable PDF build.\n\nChange-Id: I9691132f542242716992d2f842f1261b7cd30389\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/e0405ceff6d3499e8f3786df9d678ffaab6ad95f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/e0405ceff6d3499e8f3786df9d678ffaab6ad95f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
