)]}'
{"id":"openstack%2Fwatcher-dashboard~747272","triplet_id":"openstack%2Fwatcher-dashboard~master~I33469e566e745f7a0394ea12a515dac56e27a5f6","project":"openstack/watcher-dashboard","branch":"master","hashtags":[],"change_id":"I33469e566e745f7a0394ea12a515dac56e27a5f6","subject":"[DNM] test watcher-dashboard with horizon jobs on focal","status":"ABANDONED","created":"2020-08-20 18:37:57.000000000","updated":"2020-08-21 11:24:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"196da15f4d4224f636958f782a2e0f17b7fe8d38","_number":747272,"virtual_id_number":747272,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-20 19:01:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 19:01:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"464869eb8932ead817bba9308369bad206d65472","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-20 18:37:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39e4ba3309a9664c17470f09c2d744dd794cf3b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 19:01:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72af42957c1943e3adb5db200e0b0333 : SUCCESS in 5m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d705c07ce094329b709e947b8b92c1f : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8644f7be61fe4d9999aeb9d54c630868 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d792d3bb9ff433694dacf7fa4035d8f : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0c73e86adcb4b06a7bf069313e0c703 : SUCCESS in 14m 39s","accounts_in_message":[],"_revision_number":1},{"id":"196da15f4d4224f636958f782a2e0f17b7fe8d38","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-21 11:24:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83c0482590d6b9d3778da58db2764c2828dfefa9","revisions":{"83c0482590d6b9d3778da58db2764c2828dfefa9":{"kind":"REWORK","_number":1,"created":"2020-08-20 18:37:57.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/72/747272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/72/747272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/72/747272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/72/747272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/72/747272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/72/747272/1"}}},"commit":{"parents":[{"commit":"a72fd198c9f6bfe614ab2d1aae6ffc70857fcda8","subject":"[goal] Migrate testing to ubuntu focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/a72fd198c9f6bfe614ab2d1aae6ffc70857fcda8"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-08-20 18:37:19.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-08-20 18:37:19.000000000","tz":0},"subject":"[DNM] test watcher-dashboard with horizon jobs on focal","message":"[DNM] test watcher-dashboard with horizon jobs on focal\n\nDepends-On: https://review.opendev.org/#/c/744647/\nChange-Id: I33469e566e745f7a0394ea12a515dac56e27a5f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/83c0482590d6b9d3778da58db2764c2828dfefa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/83c0482590d6b9d3778da58db2764c2828dfefa9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
