)]}'
{"id":"openstack%2Fwatcher-dashboard~800081","triplet_id":"openstack%2Fwatcher-dashboard~master~I79cd70085e9e423fa3a7835e4a7fbd2b0cbf4922","project":"openstack/watcher-dashboard","branch":"master","hashtags":[],"change_id":"I79cd70085e9e423fa3a7835e4a7fbd2b0cbf4922","subject":"Fix broken devstack install","status":"MERGED","created":"2021-07-08 16:24:42.000000000","updated":"2021-07-09 05:58:22.000000000","submitted":"2021-07-09 05:57:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800081","meta_rev_id":"5ced373218540d25fcfd74b9967ee830fea115ae","_number":800081,"virtual_id_number":800081,"owner":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-09 05:57:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31455,"name":"Min Li","email":"limin.lc@inspur.com","username":"limin0801"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"all":[{"value":2,"date":"2021-07-09 05:48:40.000000000","_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-09 00:51:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31455,"name":"Min Li","email":"limin.lc@inspur.com","username":"limin0801"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"all":[{"value":1,"date":"2021-07-09 05:48:40.000000000","_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31455,"name":"Min Li","email":"limin.lc@inspur.com","username":"limin0801"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31455,"name":"Min Li","email":"limin.lc@inspur.com","username":"limin0801"}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"reviewers":{"REVIEWER":[{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},{"_account_id":31455,"name":"Min Li","email":"limin.lc@inspur.com","username":"limin0801"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-08 17:10:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-08 19:12:55.000000000","updated_by":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-07-08 19:12:55.000000000","updated_by":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"reviewer":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"state":"REVIEWER"},{"updated":"2021-07-08 19:12:55.000000000","updated_by":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"reviewer":{"_account_id":31455,"name":"Min Li","email":"limin.lc@inspur.com","username":"limin0801"},"state":"REVIEWER"},{"updated":"2021-07-08 19:12:55.000000000","updated_by":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"reviewer":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"state":"REVIEWER"},{"updated":"2021-07-08 19:12:55.000000000","updated_by":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"}],"messages":[{"id":"8eaa47932c588c92cd6c37e4c1136da65ab1d7fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"date":"2021-07-08 16:24:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"883e5b4734b78e4ab3a1ddb80da73eaba731cfec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 17:10:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9db664da83a047dbb1390e908c7c7cee : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2283383b9c4f46a899b81b7b84e0a0c1 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be58b2c06f4f49feba828d9b6d91e6c1 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/628f1f1ef4a74b35950a09fb7a25cdc1 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8796139a7fbc404e86caa53019191c41 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e02fda44aef04a7e94d2b12c8fb5070d : SUCCESS in 6m 56s\n- watcher-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/3fcdc45fdf2b4ac4a5563b40818e5703 : FAILURE in 38m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e603c1ff9bab2cf2c3f2ddfc130de89bcb3dbc94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"date":"2021-07-08 18:43:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f801733c70bd46f3335595454c25ee4b9535dd07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 19:27:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e8cca969a6c46bebea1bd42debe6cb3 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f5ca255b7764f33968b13720599fe68 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e84acff9aff44423905d833e4e4e1546 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/129dd8f4f43e48949520c4280a19dd5c : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c9a4b9c1ab548d1ab43ccc3bc434fd9 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d9b46e76b6e4cf48667396208e2b161 : SUCCESS in 5m 55s\n- watcher-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/cc1df9b5f83e4e8ba85de3b97914a486 : FAILURE in 38m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eba39bbe8052787c4c7b841531682a29c0fbd2eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"date":"2021-07-08 19:34:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0f507f67eecf4fdc71cdad2b2f1157fd0da32897","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 20:06:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88b372b9048448a8aec83b911e945388 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aabcbce5076b43a4a80496ebf3c5f8fb : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f1425425e5841b7ad5da42e21116f39 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03e7e93504c84ff78c9e2ab81d56f5f7 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2d257f0fc9a4c99b5ebbbd52204e704 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5da26b6d44d04f09a63b9af78d90be2d : SUCCESS in 7m 35s\n- watcher-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/161bd700808a46edba882fb31d1cde8e : FAILURE in 30m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a49166dbb83eae5a268166e103020dcb385e78f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"date":"2021-07-08 20:25:48.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f6c07302545d41d0289493aa7b063caf10d0964e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"date":"2021-07-08 20:26:18.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"05c7d57ba52e0227beab336865a66a1046350500","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 20:36:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5de9edd3b48b4639b65295abd65d1e83 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf3ce61edbbf45638fe6d84bb9cbce40 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ee3de2119114ed0a6410920c0193c90 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1586d574ab924f02af9ceef373b11ed3 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b739c759f7aa4d2ba582ef6eb9f1bd5f : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/560e27741a9b4944ae0286be26c5daae : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":5},{"id":"b53a2f1a1efb30e6f52c283ac92aa79f71017e7f","author":{"_account_id":31455,"name":"Min Li","email":"limin.lc@inspur.com","username":"limin0801"},"date":"2021-07-09 00:51:13.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"f790c5c115d0f30b68a22605c6b81ee38a308dfa","author":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"date":"2021-07-09 05:48:40.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"67cb1d5e8d40a7b783d14c09d69160ee21d75181","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 05:48:55.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4574e34b2aca0daec0e1912a368653c019f0d98f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 05:57:11.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d9bf75faa0e4779b4349a01be41fdac : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3e1bae933514de39b6e5701cfc535d9 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd3a66d17f8b450fb968b25561234fe6 : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e39594efb1a24217a07bd4d14d41a982 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":5},{"id":"a9d2ca8b586802eec0b1c7782fc11fe6cb365f47","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 05:57:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"5ced373218540d25fcfd74b9967ee830fea115ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 05:58:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41c0d96ac8c7402e83013e06f36c9797 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4192182302a9f81b2cac5aff484161afc776c750","revisions":{"fd2b72e56e07d85519aaa3128cc472bee323eb79":{"kind":"REWORK","_number":1,"created":"2021-07-08 16:24:42.000000000","uploader":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"ref":"refs/changes/81/800081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/81/800081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/1"}}},"commit":{"parents":[{"commit":"fd466a7e184bbd4f210a7a3aacf0680838e56fb1","subject":"DNM: Test watcher-dashboard integration test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/fd466a7e184bbd4f210a7a3aacf0680838e56fb1"}]}],"author":{"name":"Dantali0n","email":"info@dantalion.nl","date":"2021-07-08 16:22:57.000000000","tz":120},"committer":{"name":"Dantali0n","email":"info@dantalion.nl","date":"2021-07-08 16:22:57.000000000","tz":120},"subject":"Fix broken devstack install","message":"Fix broken devstack install\n\nChange-Id: I79cd70085e9e423fa3a7835e4a7fbd2b0cbf4922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/fd2b72e56e07d85519aaa3128cc472bee323eb79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/fd2b72e56e07d85519aaa3128cc472bee323eb79"}]},"branch":"refs/heads/master"},"029343f9b06b46278fa65b94c2c87a7a7a488e0c":{"kind":"REWORK","_number":2,"created":"2021-07-08 18:43:06.000000000","uploader":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"ref":"refs/changes/81/800081/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/81/800081/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/2"}}},"commit":{"parents":[{"commit":"fd466a7e184bbd4f210a7a3aacf0680838e56fb1","subject":"DNM: Test watcher-dashboard integration test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/fd466a7e184bbd4f210a7a3aacf0680838e56fb1"}]}],"author":{"name":"Dantali0n","email":"info@dantalion.nl","date":"2021-07-08 16:22:57.000000000","tz":120},"committer":{"name":"Dantali0n","email":"info@dantalion.nl","date":"2021-07-08 18:41:37.000000000","tz":120},"subject":"Fix broken devstack install","message":"Fix broken devstack install\n\nChange-Id: I79cd70085e9e423fa3a7835e4a7fbd2b0cbf4922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/029343f9b06b46278fa65b94c2c87a7a7a488e0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/029343f9b06b46278fa65b94c2c87a7a7a488e0c"}]},"branch":"refs/heads/master"},"362fa4d6b39941074edf58961ef7e93bdcb6efbf":{"kind":"REWORK","_number":3,"created":"2021-07-08 19:34:52.000000000","uploader":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"ref":"refs/changes/81/800081/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/81/800081/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/3"}}},"commit":{"parents":[{"commit":"fd466a7e184bbd4f210a7a3aacf0680838e56fb1","subject":"DNM: Test watcher-dashboard integration test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/fd466a7e184bbd4f210a7a3aacf0680838e56fb1"}]}],"author":{"name":"Dantali0n","email":"info@dantalion.nl","date":"2021-07-08 16:22:57.000000000","tz":120},"committer":{"name":"Dantali0n","email":"info@dantalion.nl","date":"2021-07-08 19:33:41.000000000","tz":120},"subject":"Fix broken devstack install","message":"Fix broken devstack install\n\nChange-Id: I79cd70085e9e423fa3a7835e4a7fbd2b0cbf4922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/362fa4d6b39941074edf58961ef7e93bdcb6efbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/362fa4d6b39941074edf58961ef7e93bdcb6efbf"}]},"branch":"refs/heads/master"},"3b53f0e3127545cc46c382fd98702da50ed8be69":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-08 20:25:48.000000000","uploader":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"ref":"refs/changes/81/800081/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/81/800081/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/4"}}},"commit":{"parents":[{"commit":"a9fc6a8ab3b22ccca20e467e3486610a15c68176","subject":"Enable tox cover jobs on zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/a9fc6a8ab3b22ccca20e467e3486610a15c68176"}]}],"author":{"name":"Dantali0n","email":"info@dantalion.nl","date":"2021-07-08 16:22:57.000000000","tz":120},"committer":{"name":"Dantali0n","email":"info@dantalion.nl","date":"2021-07-08 20:25:48.000000000","tz":0},"subject":"Fix broken devstack install","message":"Fix broken devstack install\n\nChange-Id: I79cd70085e9e423fa3a7835e4a7fbd2b0cbf4922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/3b53f0e3127545cc46c382fd98702da50ed8be69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/3b53f0e3127545cc46c382fd98702da50ed8be69"}]},"branch":"refs/heads/master","description":"Rebase"},"4192182302a9f81b2cac5aff484161afc776c750":{"kind":"REWORK","_number":5,"created":"2021-07-08 20:26:18.000000000","uploader":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"ref":"refs/changes/81/800081/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/81/800081/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/81/800081/5"}}},"commit":{"parents":[{"commit":"a9fc6a8ab3b22ccca20e467e3486610a15c68176","subject":"Enable tox cover jobs on zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/a9fc6a8ab3b22ccca20e467e3486610a15c68176"}]}],"author":{"name":"Dantali0n","email":"info@dantalion.nl","date":"2021-07-08 16:22:57.000000000","tz":120},"committer":{"name":"Dantali0n","email":"info@dantalion.nl","date":"2021-07-08 20:26:07.000000000","tz":0},"subject":"Fix broken devstack install","message":"Fix broken devstack install\n\nChange-Id: I79cd70085e9e423fa3a7835e4a7fbd2b0cbf4922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/4192182302a9f81b2cac5aff484161afc776c750"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/4192182302a9f81b2cac5aff484161afc776c750"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
