)]}'
{"id":"openstack%2Fwatcher-dashboard~866308","triplet_id":"openstack%2Fwatcher-dashboard~master~Ie9cdf3e87a06a0af8cf36c8bbdef0f14336e8f70","project":"openstack/watcher-dashboard","branch":"master","topic":"866308","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-01-17 13:00:12.000000000","reason":"\u003cGERRIT_ACCOUNT_29313\u003e replied on the change","reason_account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2023-01-17 13:20:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie9cdf3e87a06a0af8cf36c8bbdef0f14336e8f70","subject":"Replace deprecated inspect.getargspec","status":"MERGED","created":"2022-12-01 09:24:59.000000000","updated":"2023-01-17 13:22:03.000000000","submitted":"2023-01-17 13:20:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866308-866308","meta_rev_id":"a7789966ac7560628599a1b834250a7f801a2ef4","_number":866308,"virtual_id_number":866308,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-17 13:20:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2023-01-04 11:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2023-01-16 07:02:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-01 09:33:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-04 11:20:32.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"2d3fa56e13392f3a1d253642f17710e0a08d1a9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-12-01 09:24:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3db8c35321a057085101c2c0cd5a7018f7bb1a0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-12-01 09:25:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1e1173cfc3d658d1f09356ca48a6a648266b0b8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 09:33:58.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eaf1c795b0d846f8b1ee5914df983a3e\n\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34225791a324432184987f5d0285ef98 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c34c40fa42794bc495aee76a1e5aad9e : FAILURE in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/728ba5e50eb9475a8a2f1ddd672b87a3 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57a66457da2d4fe781fb65ddbdd5bef8 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1cec38459b94352b017c4b01b54677c : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b7ee2121d8f4d35b6aa9d6f0c2c9e27 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":2},{"id":"08089143f9ef638bf0d6462a08a72e45ae898701","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-01-04 10:23:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fe44173481c514ed7a5bb75538f75eb8864820be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 10:33:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/804d0f6ffda540a392bbd2ba185b74e6\n\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a34d6b6be6404220b947bbca60e030d2 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39687036173f4c7ebc263bbeef7a29ec : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50ecbfc1b68f47cbaa34dcf68bd1f8e9 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da157dba2bc34740831333d2b6716a99 : SUCCESS in 3m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d30c66f363cc4091be779b5b8eb5951a : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/216812c7283e4cc2a7a5e9ce523f5702 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":3},{"id":"f015ea2efc332b2665f05a7ce66c1459a49cfdb9","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-01-04 11:20:32.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0ae2ff0bc5d2af4b476b96e10a7912adf7894252","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-01-16 07:02:40.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a2ac5ec0b5eb7696003ca4026a5fcf066060967f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 07:03:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6edf10d0392ae0dc5fd14b0b10e3b69b76b29c39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 07:22:55.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa5080fab6424e9396bd81ed8db84372\n\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cddb562222244240a3e91b6b6fd3f7c6 : FAILURE in 2m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/698a49247d9f4c5aa30db04494d42e70 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddc17e71c11d46baa21909af4a138851 : FAILURE in 2m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4cefe1d0dc124d8f8cb037b7a4612527 : FAILURE in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35f4d329842843738afb0dbccd26e1c6 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":3},{"id":"2717e50867ca505e89ed3e684c0c5aaee3798035","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-01-17 13:00:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1f6d139b823257bc5d5194082d7b0181659059b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 13:10:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a86c71e53ff40c68ae2b4ed00d79244\n\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d07d401d7c840c3aa2f111090c2dbe3 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8355024949b04525ae65d670b1bb11b3 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b508bada451542839860c1a37ce7cdab : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5cd6072c2e04452a3f26d2d6b7b883b : SUCCESS in 5m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b357a8b54534674b6985e0bd8c504c4 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b38f345a56ef47bd917eca0140c8b4d2 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":3},{"id":"038c52abe2c252d8f70ef0a9d597801d1ee73cf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 13:11:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f4317740beb1ba53c3143c08feaa0d6d6320e47f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 13:20:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/503aa94a930b4a358b31809fe6b0bd10\n\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5079a95c4a194853bc0942eb90be3513 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac49cb81df0840bbaa23f180f1b8c4b5 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01d3403362ad4d58bcc8290979f5d0ac : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8935ce5e4215444d951b1618715784ab : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f05ba5e905f8451cb35f53e269828729 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":3},{"id":"33cbdb20773c921457be6721c3962273814eab9c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 13:20:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a7789966ac7560628599a1b834250a7f801a2ef4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 13:22:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86469016f0d548d3898599b6d521dcbf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/512d45d56a4c44e0b7c18a75914267eb : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2f5345a6d8c49fdb473dbd60cd075aee2b08f4bf","revisions":{"ee3328d3e0db53a8d94cf762a24054585f4f8ee1":{"kind":"REWORK","_number":1,"created":"2022-12-01 09:24:59.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/08/866308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/08/866308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/08/866308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/08/866308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/08/866308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/08/866308/1"}}},"commit":{"parents":[{"commit":"b526383ebae37f3bdb6ef4be9245684967c46d65","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/b526383ebae37f3bdb6ef4be9245684967c46d65"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-12-01 09:23:19.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-12-01 09:24:57.000000000","tz":60},"subject":"Python 3.11: use getfullargspec, not getargspec","message":"Python 3.11: use getfullargspec, not getargspec\n\nIn Python 3.11, inspect.getargspec. We shall use getfullargspec from now\non instead. This patch was added to Debian. See Debian bug:\n\nhttps: //bugs.debian.org/1025195\nChange-Id: Ie9cdf3e87a06a0af8cf36c8bbdef0f14336e8f70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/ee3328d3e0db53a8d94cf762a24054585f4f8ee1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/ee3328d3e0db53a8d94cf762a24054585f4f8ee1"}]},"branch":"refs/heads/master"},"e5331420c3742d35621e650e481fd74d0f04bb02":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-12-01 09:25:22.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/08/866308/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/08/866308/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/08/866308/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/08/866308/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/08/866308/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/08/866308/2"}}},"commit":{"parents":[{"commit":"b526383ebae37f3bdb6ef4be9245684967c46d65","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/b526383ebae37f3bdb6ef4be9245684967c46d65"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-12-01 09:23:19.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-12-01 09:25:07.000000000","tz":60},"subject":"Python 3.11: use getfullargspec, not getargspec","message":"Python 3.11: use getfullargspec, not getargspec\n\nIn Python 3.11, inspect.getargspec. We shall use getfullargspec from now\non instead. This patch was added to Debian. See Debian bug:\n\nhttps://bugs.debian.org/1025195\n\nChange-Id: Ie9cdf3e87a06a0af8cf36c8bbdef0f14336e8f70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/e5331420c3742d35621e650e481fd74d0f04bb02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/e5331420c3742d35621e650e481fd74d0f04bb02"}]},"branch":"refs/heads/master"},"2f5345a6d8c49fdb473dbd60cd075aee2b08f4bf":{"kind":"REWORK","_number":3,"created":"2023-01-04 10:23:26.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/08/866308/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/08/866308/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/08/866308/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/08/866308/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/08/866308/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/08/866308/3"}}},"commit":{"parents":[{"commit":"b526383ebae37f3bdb6ef4be9245684967c46d65","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/b526383ebae37f3bdb6ef4be9245684967c46d65"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-12-01 09:23:19.000000000","tz":60},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-01-04 10:23:23.000000000","tz":330},"subject":"Replace deprecated inspect.getargspec","message":"Replace deprecated inspect.getargspec\n\ninspect.getargspec was deprecated since Python 3.0 and\ninspect.getfullargspec is its replacement with correct handling of\nfunction annotations and keyword-only parameters[1].\n\n[1] https://docs.python.org/3/library/inspect.html#inspect.getargspec\n\nCo-Authored-By: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\n\nChange-Id: Ie9cdf3e87a06a0af8cf36c8bbdef0f14336e8f70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/2f5345a6d8c49fdb473dbd60cd075aee2b08f4bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/2f5345a6d8c49fdb473dbd60cd075aee2b08f4bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}}]}],"submit_requirements":[]}
