)]}'
{"id":"openstack%2Fwatcher-dashboard~905170","triplet_id":"openstack%2Fwatcher-dashboard~master~I84c8571e017cbb39b364b9f94bbb35747cf75e8a","project":"openstack/watcher-dashboard","branch":"master","topic":"py3.11","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-02-13 08:00:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I84c8571e017cbb39b364b9f94bbb35747cf75e8a","subject":"Update python classifier in setup.cfg","status":"MERGED","created":"2024-01-10 03:21:59.000000000","updated":"2024-02-13 08:01:05.000000000","submitted":"2024-02-13 08:00:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905170-py3.11","meta_rev_id":"7ecf63b448770981d81481712298451a5fb9dea3","_number":905170,"virtual_id_number":905170,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-13 08:00:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2024-01-11 16:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2024-02-13 04:57:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-10 03:39:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-11 16:13:20.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"296a66d3efbe029ec607fe326a607981730ac3d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-01-10 03:21:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"186eedbcd79ffbefa12f86c5be7fabe81c8338db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 03:39:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a24539d4334b4a7d8313c5c6207f0a38\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/82a4416db4f341639f1834b1f53f70d9 : SUCCESS in 2m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b789841980814f1f8593a515103c9774 : SUCCESS in 4m 50s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/2a690d633b4a4b93be6b405a0c563ee8 : FAILURE in 5m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9d78924b901438db955e808f5a7663e : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21c4a67bb1ff44dd859fdc3d99a232a7 : SUCCESS in 4m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6808bc500d0540d59a750791edaa2f1b : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3daa99554e2f4150903a30d1014ad852 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":1},{"id":"49ff47227dc4565aa34343d64d62f2d3b3bd4c7d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2024-01-11 16:13:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cf0f2f1407b9c2f8ec620c5f5bc682381493f29a","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2024-02-13 04:57:21.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b45454ad94a07713d3a28a2dcf608a17e091ac7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 04:57:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db4de4fd9ca939bf5ee64748987c77edf95634ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 05:09:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee034e89d3c94bcfa3dfef6b5837a2d6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6d337261cf9a4b7087b8802213a7bab5 : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de351119def14ff7aad7d08a397005d5 : FAILURE in 3m 42s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/84ce3f52f73143629af94ad2455c6265 : FAILURE in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ce6af119f9040299d5c63d185c7acd2 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb4408a496e14be7895fe8af833d8026 : FAILURE in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7b76c477496467d81fc02e2446c3532 : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":1},{"id":"6d215aa385a3c5719e085bcbeca0ec2aef03a805","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2024-02-13 07:36:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f96b092445d718dcaff617da36d4f1490f89ac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 07:48:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac0ba8d8290c4715bc658472cceaf5a7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5e592163554241619367a8f353e4d9e0 : SUCCESS in 1m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd02b016c9b54bc386aef1c6f53055ec : SUCCESS in 4m 17s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/0a3297ccc136480eb54fc240dc343a2d : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/489d3fbbbd8840de9d177440362ca64a : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dabe4f7e9ba4ed3be90f33370ca61b2 : SUCCESS in 4m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/306dbf558b62470ba6d1d9717afda859 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d79a2f1ab587480096069bed97d83275 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"28aed16989a73f45473b94f34508895a348f3c69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 07:48:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"900b5c8f6f5025f4ff5eb9bb9e3f5ad63fee30fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 08:00:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/235ed603eeda45fb9e52cc17e7f211e8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2bedaeed833445e4b9677f6b365b26b9 : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84f763457ecd4a14a5fb93038148f390 : SUCCESS in 4m 18s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/04120755bdff470eb7cea4c7f5761e45 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e74e1322bc74e22a5d036c4b1e13c40 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/944dbebabd834a01a931266f7fe55a8d : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e53026d1f44443c9eec82df6936b85c : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":1},{"id":"df25d8943792fa426520c80e02d9e8be37cf3d9d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 08:00:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7ecf63b448770981d81481712298451a5fb9dea3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 08:01:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d8e41ebbbf244e0b8d16a90ca2e9ab8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d64464dc186a495987d756f1e96b027d : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b1db180990e325d4009f3f9ed8f651937761ae90","revisions":{"b1db180990e325d4009f3f9ed8f651937761ae90":{"kind":"REWORK","_number":1,"created":"2024-01-10 03:21:59.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/70/905170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/70/905170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/70/905170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/70/905170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/70/905170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/70/905170/1"}}},"commit":{"parents":[{"commit":"eaf8db4c84d6c95dcd1957186cbdd8596255f3e9","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/eaf8db4c84d6c95dcd1957186cbdd8596255f3e9"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-10 03:14:55.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-10 03:21:57.000000000","tz":-480},"subject":"Update python classifier in setup.cfg","message":"Update python classifier in setup.cfg\n\nAs per the current release tested runtime, we test\npython version from 3.8 to 3.11 so updating the\nsame in python classifier in setup.cfg\n\nChange-Id: I84c8571e017cbb39b364b9f94bbb35747cf75e8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/b1db180990e325d4009f3f9ed8f651937761ae90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/b1db180990e325d4009f3f9ed8f651937761ae90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
