)]}'
{"id":"openstack%2Fwatcher-dashboard~932693","triplet_id":"openstack%2Fwatcher-dashboard~master~I5a45357ccd69da26443818ae165ba08963004358","project":"openstack/watcher-dashboard","branch":"master","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2024-11-18 07:01:34.000000000","reason":"\u003cGERRIT_ACCOUNT_8449\u003e replied on the change","reason_account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-11-04 15:59:02.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2025-01-09 23:35:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5a45357ccd69da26443818ae165ba08963004358","subject":"Fix efficacy indicators in action plans","status":"MERGED","created":"2024-10-18 11:09:31.000000000","updated":"2025-01-09 23:36:13.000000000","submitted":"2025-01-09 23:35:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"932693","meta_rev_id":"934e188882f16c92b3601dd304592fae14b87e45","_number":932693,"virtual_id_number":932693,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-09 23:35:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"recommended":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2024-11-05 07:42:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2024-12-11 11:28:31.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2024-12-03 01:09:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2024-12-12 11:41:24.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2025-01-09 23:05:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-18 11:20:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-04 15:10:03.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-11-05 07:42:09.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"CC"},{"updated":"2024-11-05 07:42:27.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2024-12-11 11:28:31.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2024-12-12 11:41:24.000000000","updated_by":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"reviewer":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"state":"REVIEWER"}],"messages":[{"id":"94cf403ed339ffbe497481e0324d50379972b3ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-10-18 11:09:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"773006cf13ec9a5841480bddcda81abe63b1f641","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-10-18 11:12:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c4db52453b97bd4b057219cf730ec3ba771c1574","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 11:20:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab27ae853b6a4d1e9dcf45439fd78d8f\n\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/75b8657147e14079a9ea1e46f73a2eb0 : SUCCESS in 4m 10s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/6b965fabe30e41728f6739e684fa0442 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/766e41a409294bdbb25bd06a55b85f60 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb5cf04b72574ce198ddaa3ccc4b0e1d : SUCCESS in 5m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c3ecd03b1734bda8fbcee1ea71f9ee4 : SUCCESS in 6m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac5d7bd65fa9462895cfc4b0b3e878fb : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/530f74e602574f3fb14b101ba4db1b6c : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":2},{"id":"55f502faf2d94091226d223edcbbeeb8bde464dd","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-04 15:10:03.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0c65365c55b1184ac29cea51b5ae11d47b64fb9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-11-04 15:34:48.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4706aa2613a43194f840cd365d6dbd9462c1c3a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 15:45:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca7ba14173bd483590ea47291a83ed45\n\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85e390e2687f476fa4b9154d48c0e1cc : SUCCESS in 3m 34s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/08f8f8cd8c6444afbaa27dfae0734e1b : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a3711a26db845568531e6bbaf6feb73 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14f74aee0a0c4bcc9c600c0e820e5faa : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e415ecec534467e8d8b3c0f891c5859 : SUCCESS in 4m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9a6a81f70954ab18d7aaac0ff1109d9 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7dd489168a346fd8b312ffd63e93564 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":3},{"id":"a265c1ca9fd8bf976d69e893c9a8565bf11e7818","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-04 15:59:02.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2b297ab0977fe58e1d7ee2de3f34ce8f370d9a4f","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2024-11-05 07:42:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b952814d0aec7c4c2f39354239826f0a936a9ae1","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2024-11-05 07:42:27.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f9083fc4df0b1815edcec5dbfee14d502a27461f","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-11-05 09:29:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"80d3c3a4d42430fcd536a0c8ff981f7fb8999a27","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2024-11-06 08:03:19.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e46ae2485fbd04ca2f0665a990dac38953a90301","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-11-13 17:53:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d83f1f462a7bf27f81d79ae3682f7ca9a724bd02","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-11-14 08:31:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"18db475d6091c45c8bbd40cc1929ad61f745b4c3","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2024-11-18 07:01:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5904e408588cf9115510fa2d43850ad7a67dc344","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-12-03 01:09:48.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3aec63a9847fc0ed4e19639c7c6796c2fd93597d","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2024-12-11 11:28:31.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b3a6f463ed3343c30566d95ec26541ab5c514656","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2024-12-12 11:41:24.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fda1541ccfc7c592e119aed6396656d3adbc5376","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-09 23:05:17.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8e5a7629356ff142a5c622879aae49e64342cb0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 23:05:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e4168a18fc7f6f994094885533df7ed316319567","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 23:35:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ebf21df2667490c9a532e710c3d7420\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f2dfcf8bee7b4691893c0576fa6f9910 : SUCCESS in 4m 55s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/912bea2854b04d77bb4966243d8b1582 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1376d0a0e2c4a15a05a6c63cd922d1c : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43d3b25be7ac436a86406683125fbd1f : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ac6c331f6d34f2289bb6c7f9d191125 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":3},{"id":"e8056ea85a2926a235e15c28624514c847d4b6e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 23:35:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"934e188882f16c92b3601dd304592fae14b87e45","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 23:36:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a73223c938d948c895f6e1e8914eb3c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a481a43ea454e63b5fc4f7bf1fa1b14 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"42d8914dcfc404f76e8e6e67dfb918b845ab9316","revisions":{"f46101de39b786d8b491187dff3b9d4a872f251a":{"kind":"REWORK","_number":1,"created":"2024-10-18 11:09:31.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/93/932693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/93/932693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/93/932693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/93/932693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/93/932693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/93/932693/1"}}},"commit":{"parents":[{"commit":"0451ec9c5fbaf1a4a362df91088d3d937061b3b3","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/0451ec9c5fbaf1a4a362df91088d3d937061b3b3"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-10-18 10:56:12.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-10-18 11:09:29.000000000","tz":120},"subject":"Fix efficacy indicators in action plans","message":"Fix efficacy indicators in action plans\n\nCurrently, Efficacy in action plans list and efficacy indicators in\naction plan details are empty.\n\nThere are two different problems:\n\n- While EfficacyIndicator expects a dict, global_efficacy is a list\n  with one dict inside.\n- EfficacyIndicator is using getattr function to initialize the\n  attributes while the correct way for dicts is get, which also provides\n  None when the attribute does not exist.\n\nThis is effectively reverting [1].\n\n[1] https://review.opendev.org/c/openstack/watcher-dashboard/+/587032\n\nChange-Id: I5a45357ccd69da26443818ae165ba08963004358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/f46101de39b786d8b491187dff3b9d4a872f251a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/f46101de39b786d8b491187dff3b9d4a872f251a"}]},"branch":"refs/heads/master"},"ae7746aae283a33157005c7df4ca124860b2a7d9":{"kind":"REWORK","_number":2,"created":"2024-10-18 11:12:23.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/93/932693/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/93/932693/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/93/932693/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/93/932693/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/93/932693/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/93/932693/2"}}},"commit":{"parents":[{"commit":"0451ec9c5fbaf1a4a362df91088d3d937061b3b3","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/0451ec9c5fbaf1a4a362df91088d3d937061b3b3"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-10-18 10:56:12.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-10-18 11:12:02.000000000","tz":120},"subject":"Fix efficacy indicators in action plans","message":"Fix efficacy indicators in action plans\n\nCurrently, Efficacy in action plans list and efficacy indicators in\naction plan details are empty.\n\nThere are two different problems:\n\n- While EfficacyIndicator expects a dict, global_efficacy is a list\n  with one dict inside.\n- EfficacyIndicator is using getattr function to initialize the\n  attributes while the correct way for dicts is get, which also provides\n  None when the attribute does not exist.\n\nThis is effectively reverting [1].\n\n[1] https://review.opendev.org/c/openstack/watcher-dashboard/+/587032\n\nChange-Id: I5a45357ccd69da26443818ae165ba08963004358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/ae7746aae283a33157005c7df4ca124860b2a7d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/ae7746aae283a33157005c7df4ca124860b2a7d9"}]},"branch":"refs/heads/master"},"42d8914dcfc404f76e8e6e67dfb918b845ab9316":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-11-04 15:34:48.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/93/932693/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/93/932693/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/93/932693/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/93/932693/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/93/932693/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/93/932693/3"}}},"commit":{"parents":[{"commit":"0451ec9c5fbaf1a4a362df91088d3d937061b3b3","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/0451ec9c5fbaf1a4a362df91088d3d937061b3b3"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-10-18 10:56:12.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-11-04 15:34:46.000000000","tz":60},"subject":"Fix efficacy indicators in action plans","message":"Fix efficacy indicators in action plans\n\nCurrently, Efficacy in action plans list and efficacy indicators in\naction plan details are empty.\n\nThere are two different problems:\n\n- While EfficacyIndicator expects a dict, global_efficacy is a list\n  with one dict inside.\n- EfficacyIndicator is using getattr function to initialize the\n  attributes while the correct way for dicts is get, which also provides\n  None when the attribute does not exist.\n\nThis is effectively reverting [1].\n\n[1] https://review.opendev.org/c/openstack/watcher-dashboard/+/587032\n\nCloses-Bug: #2086627\n\nChange-Id: I5a45357ccd69da26443818ae165ba08963004358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/42d8914dcfc404f76e8e6e67dfb918b845ab9316"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/42d8914dcfc404f76e8e6e67dfb918b845ab9316"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
