)]}'
{"id":"openstack%2Fwatcher-dashboard~949737","triplet_id":"openstack%2Fwatcher-dashboard~master~I28715183ae9c1f9b75d6d68d6a8a0d554bffd9ce","project":"openstack/watcher-dashboard","branch":"master","topic":"pip-23.1-support","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-05-20 00:05:08.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"hashtags":[],"change_id":"I28715183ae9c1f9b75d6d68d6a8a0d554bffd9ce","subject":"add pyproject.toml to support pip 23.1","status":"MERGED","created":"2025-05-14 12:48:50.000000000","updated":"2025-05-20 00:30:06.000000000","submitted":"2025-05-20 00:29:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949737-pip-23.1-support","meta_rev_id":"c999ad87145ce5664f8a21b29498320ea86978ca","_number":949737,"virtual_id_number":949737,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-20 00:29:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2025-05-20 00:05:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2025-05-14 15:37:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2025-05-20 00:05:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-14 13:01:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-14 15:37:54.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"b5872d1e8c8e6f8ccb53ad843f6de4025928ad19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-14 12:48:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"934aeb3c96ffe6ac9bc34ba622324df46d660fd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 13:01:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bee58ab42b7b4bc0b8ec795ba26b3321\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c2b9b79b5ed41e587e1e44ea7c4c174 : SUCCESS in 4m 07s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/5ad5e30156c349d498175d1ba1517aca : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a6d563d804e454ca20248ecbb4eff93 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abde4f94f930473db85d82860a59ccd9 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a478b252c1a49bbb4bd49155e1c572c : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3da8620ec1e248e4af95c365bebdee41 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c229642e3ac4ec0884bf5a8a99b88c6 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"84572e9355b6a4d1396a42c37505744743f26e41","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-05-14 15:37:54.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/3c1048104b444cdabd45053264c9e838\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/37fac7cb70ab46bba58d461807043aa6 : SUCCESS in 2h 47m 44s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/8844787da5d24edabd181b3a051258e7 : SUCCESS in 1h 53m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ba8e0001b6cc3b5428f47efa6e78c878e4b6aec3","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-20 00:05:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"04c3b7df3900ad4921a2d4d779d6952ddff19486","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 00:05:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c81d08e72f91f04cc0716cbfce7fb03454d15a84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 00:29:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/482aa0c5d63d4aa2822e47def079ae54\n\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/56ebd8165aec4f7aa36fffffb050074f : SUCCESS in 3m 54s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/e698b4a8ab6747508bde2e89c1078438 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6aaf151c70d458184b213a7804070d5 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4f30b9accb3494fabad8a1be32491bf : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf0bb8cc35594fb4a1dc55d87012637b : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":1},{"id":"cdd66e78054eba1d90084d575828587403b06047","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 00:29:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c999ad87145ce5664f8a21b29498320ea86978ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 00:30:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e21a15cae084601857e76becba27c50\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f872db5fd10243cd8ecc208a43ee6a11 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20a3637c6de90178f0401296eadb7c1919e846a8","revisions":{"20a3637c6de90178f0401296eadb7c1919e846a8":{"kind":"REWORK","_number":1,"created":"2025-05-14 12:48:50.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/37/949737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/37/949737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/37/949737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/37/949737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/37/949737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/37/949737/1"}}},"commit":{"parents":[{"commit":"0451ec9c5fbaf1a4a362df91088d3d937061b3b3","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/0451ec9c5fbaf1a4a362df91088d3d937061b3b3"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2023-11-02 11:16:49.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-05-14 12:48:47.000000000","tz":60},"subject":"add pyproject.toml to support pip 23.1","message":"add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects\nthat do not have pyproject.toml and now uses a pyproject.toml\nwhich is vendored in pip.\nTo address that, this change adds the minimal pyproject.toml\nto enable pbr to be properly used to build editable wheels.\n\nThis is required to support installing devstack on\ncentos stream 9 and related distros with GLOBAL_VENV\u003dTrue\nWithout this change the wsgi scripts are not generated in\neditable mode. i.e. pip install -e /opt/stack/keystone\n\nSee https://pip.pypa.io/en/stable/news/#v23-1\nand https://github.com/pypa/pip/issues/8368 for more\ndetails on the removal of the fallback support.\n\nsetuptools v64.0.0 is used to support editable installs\nvia its PEP-660 implmentation\nhttps://github.com/pypa/setuptools/pull/3488\n\nChange-Id: I28715183ae9c1f9b75d6d68d6a8a0d554bffd9ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/20a3637c6de90178f0401296eadb7c1919e846a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/20a3637c6de90178f0401296eadb7c1919e846a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
