)]}'
{"id":"openstack%2Fwatcher-dashboard~958209","triplet_id":"openstack%2Fwatcher-dashboard~master~Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273","project":"openstack/watcher-dashboard","branch":"master","topic":"blueprint-add-skip-actions","attention_set":{},"removed_from_attention_set":{"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2026-03-03 20:36:24.000000000","reason":"\u003cGERRIT_ACCOUNT_30002\u003e replied on the change","reason_account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},"34452":{"account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"last_update":"2026-03-03 20:50:07.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-03-03 20:10:08.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"28006":{"account":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"last_update":"2026-03-03 20:50:07.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-03-03 13:44:45.000000000","reason":"removed on reply"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2026-03-03 20:50:07.000000000","reason":"Change was submitted"},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2026-03-03 20:50:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273","subject":"Add skip action UI controls","status":"MERGED","created":"2025-08-21 14:51:30.000000000","updated":"2026-03-03 20:51:54.000000000","submitted":"2026-03-03 20:50:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":194,"unresolved_comment_count":11,"has_review_started":true,"submission_id":"958209-blueprint-add-skip-actions","meta_rev_id":"868ddc89751514493cf08eaca9fe32b76d7b97de","_number":958209,"virtual_id_number":958209,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-03 20:50:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2026-03-03 20:36:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2026-03-03 20:10:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":1,"date":"2026-03-03 13:44:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2026-03-03 20:36:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}],"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-21 15:04:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-21 17:58:21.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-08-28 12:33:11.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"CC"},{"updated":"2025-08-28 19:03:10.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2025-09-02 10:44:18.000000000","updated_by":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"reviewer":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"state":"CC"},{"updated":"2025-09-02 13:59:12.000000000","updated_by":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"reviewer":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"state":"REVIEWER"},{"updated":"2025-09-03 16:28:11.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-09-08 09:39:44.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2025-12-01 00:41:57.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"}],"messages":[{"id":"33a715906cec01922c984cc22ea39bed3063a585","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2025-08-21 14:51:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f42bba9c5747cad652e19e4aed8c1cc8408c78b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 15:04:22.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/60ba00646848443ead52ea5b9a8b5b98\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d4c7e6309fad4dd7accababad330873f : SUCCESS in 1m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d177fa7c681c444f9046f4428c9eede5 : SUCCESS in 6m 04s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/84e4727daa9948afa9bfb6e3f1caeb24 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a6e141c87ef4d72bbda44d1bf39d098 : FAILURE in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb5ae5b9dc8d44aeb530ed90601d9f92 : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/928c1bc1031440f2b859406b5d176969 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6c1759cfddc42428ce81c0aed6ddc51 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da70ea57a3414fd2a6ad63d5b897a361 : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":1},{"id":"670686a6bb2b3aeaba5562c2e5150ebbe918a545","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-08-21 17:58:21.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/cf5d3134525c4885ac956a23351f2bc7\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/e6f1f777a758491a9364b3ddbbeacdc4 : SUCCESS in 3h 05m 34s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5c23da440fc14ebe98368bf6bda808af : SUCCESS in 1h 55m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea29f5505fcc933475305e89fc1ae6eaf7de1a45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-08-28 12:19:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"74415af2453e3909716acc70e11d570a37da563b","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-08-28 12:33:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2a4f80c16cde548558e6bf1b2a4f144ec752a018","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 12:35:44.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d4f0819debf4819af67495d47d7edc1\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/54a66a9c28524a12a089c94fa8e7ab0d : SUCCESS in 1m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/93ea1189a732467c8851e2156e92fad9 : SUCCESS in 4m 57s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/2d79b2d4d7394e1e87fabd154da0b0b3 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/651dace9a8da49fdae899d591200d015 : FAILURE in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e62b3ff920604ce392e0278fff43dc7f : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c7a26fea78654a9bb649109d5268695b : SUCCESS in 10m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b06ca54796244b2fbe5951f00cad40c3 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/116b536ecc2a43e7b0cdf157a1a94569 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":2},{"id":"d66eda3ab040540f18e8a14abdddc33375e2870c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-08-28 12:40:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"14eea9e202b25a7725bd3c92e22c127625aa3d50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 12:51:30.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/956911 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"ff03c4a745c3bf8bbeedce4f10a7bd69c06904ff","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-08-28 12:51:37.000000000","message":"Patch Set 3: Code-Review-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/python-watcherclient/+/956911 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"6515c4595d4bdab6215aabba50d4362a682eba0a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-08-28 12:52:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"250a8f9552684d4af5835ef860df9fea25ecad65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 13:03:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f429686aebba4574a8b08417e6c900e0\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ba6de9382d3c4c6da687e5740aa20bb4 : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4186e2662e754b6a8642070c23004eb7 : SUCCESS in 5m 57s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/4b3c7eb83a264e99a336e39216063e41 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39818c4889f142e3909fb40d954aec07 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2499bf70e2784910ae8236dde5868056 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c52545176a484abfbda4416ee386ad05 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0a4219c0f0044db83be9a93d4707800 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f88e29eb85e4ad48a19ebada27094b1 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":3},{"id":"760bc4c32aa5d85226d034f8a309cfa06ec94754","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-08-28 13:19:52.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/python-watcherclient/+/956911 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"9c5ff80cddbefc56227dc0ae261152458617a18b","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-08-28 19:03:10.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"dccdd3fc60e616f06405c6778791022a28664613","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-08-29 04:08:49.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"27b6b9bb4b79160d83977dd8beea4cd39bfd1591","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-08-29 04:10:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cde7a02b69e8ec4c0c08b8303a87f50d314c512a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-08-29 04:17:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ed410f766e79f0fcacdf10dd169f32b44ab47e5c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-08-29 04:20:06.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e9f1f23c02f76b6d77fc36d241322430ee00b3da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-08-29 04:28:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ed827ab469252581fc0c97e055f49273d58d6b1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 04:41:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09df9327f8b4439d86670b652137a8b5\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1721b2bc6f11402bb9b5b02226885bbf : SUCCESS in 1m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b7d4c0f327a54959baa1b8c3e30af0e7 : SUCCESS in 4m 23s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/990aa19640c742e5855c462107c2061f : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/815dce3f804a40218b4bbd4d6d226aea : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c11242de6a94aa8af00518ad07308e8 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb8875d7000c491a9349deccb8ffe786 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/716c5a9a5c334df4a87b3df031143afc : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27c6884c5f4a4185bf6afb82b61f9b53 : SUCCESS in 9m 32s","accounts_in_message":[],"_revision_number":7},{"id":"599840e3cc10bf6618d7e8b46474336056090c28","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-08-29 07:06:10.000000000","message":"Patch Set 7: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/bacd2c8babb94229a6f52da0b1d97dbe\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5443d5837871464abf61d68198de5fb5 : SUCCESS in 2h 36m 09s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/7091207bc8bd4a1eb010dc947b06a613 : SUCCESS in 1h 53m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"27e5d316317bbd76824204b9a6976c41f286b843","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-08-29 11:53:02.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"cf521818cc76d167f283e3c23d8ad05ed494d80d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 12:11:01.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01341c9e6b034c9f917cd969de40f0af\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/46718c3f3bb14a80bc9d6d84d7dc5b66 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/456fe1627e264de98a9a5d118b5dc759 : SUCCESS in 4m 01s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/799b7052901f4fcca124e77af60423e5 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/372cd159e0594d1faf61613ae247fe8b : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7397477644b84431a0677877bee740ef : SUCCESS in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/babb26059bcd4e9181b8513de73abc74 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e4552c78d6149f994046cf5e030937b : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb6f8465a766400285c41252bf658386 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc94c92d70a243a2a16b7cb4fe0f1778 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":8},{"id":"672f380379c3047eb18c296ca16a6b2f00389b1a","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-08-29 12:30:04.000000000","message":"Patch Set 8: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"d4425dfb9e712f31becbb4235258631cadd535f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-08-29 14:07:21.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"3e771548626025cfcb0585cac4e0fa35c41d3134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 14:19:22.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a8c87736fd44c338e73d3542f815b74\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/483c003d1cd54a8a8f0997724d7524b3 : SUCCESS in 2m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5145dd049acb4b3796aecf69ee5d5171 : SUCCESS in 6m 36s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/c6efbe0d1ebf449ebd518a7195e11eb7 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8f52db2e77944fb80f9a4327cedf291 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12025aab6a8148d69234cca236d6bc37 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21e2f89ede294bf19cc801bd145aa950 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/289cf04c6ff24d9992a9b7803e5c2e0a : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ac14b2ebf5b45979e7fbfb44dbceb79 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd18559e93fe4590877e76d34a60e51c : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":9},{"id":"9f873ba3da44729e499b50f105d36d2369d8dbf7","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-08-29 16:54:11.000000000","message":"Patch Set 9: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/e637cc895ebf4f329dcc3625ebd7a32f\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/ddc75d9919944bde86de851202d39ea9 : SUCCESS in 2h 45m 22s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/4e9a83a821e741b1abd25bbd38f5498e : SUCCESS in 1h 58m 18s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0cae1e1aeadf5a4556bf2e10c656d59753a5358f","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-08-29 18:30:37.000000000","message":"Patch Set 9: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"ef42b74a4b01d2d072679ab6b3063493724d961a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-01 04:46:55.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"8e9c83075051b2328089605087391c3dd3d1fe2a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-01 04:47:11.000000000","message":"Patch Set 10: Code-Review-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/watcher-dashboard for 957232,8","accounts_in_message":[],"_revision_number":10},{"id":"8add59b40e4023b17ed488412ec991f6dc52572a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-01 04:48:15.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":11},{"id":"9111e136da066ba21eb2dc3b15af6b9778abbd57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 04:59:07.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25e03bb6143f4014a065a66a2113152c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0378d258bf38455db7d6f2a95edbb8c5 : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4157064bdbde48fa9da96c3a94b530e1 : SUCCESS in 5m 54s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/bbafe6c85c4a4eeab2739bfadd830ba0 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0d3973d225440859839726d7d99360a : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88895c0b42774ddaafff7d9ce5c9ad59 : SUCCESS in 2m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee006a6e8e20459ab1f3b3d63fd38fdd : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50420a06e9b447499f14cd3d0b5f376d : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c430872d4984c8b8f91351d6a6ecbff : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3650db30d1f24bc7a4e628cae43e1cbf : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":11},{"id":"d1c00181e923c8e6fd15ff19564d03818b693606","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-01 05:01:03.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"cb0fa95b495196cf210972e7704bd4abe1aa39d5","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-01 05:19:00.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f67bbd187ae30023d7c106b3c3caee6778f1941c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-01 07:28:31.000000000","message":"Patch Set 11: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/1a9f9ce18f304ea99104e7e872487190\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/fe153811fbd14dd882ca94c1186ae229 : SUCCESS in 2h 39m 06s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/85c3492814774055aad668a670b4fa16 : SUCCESS in 1h 54m 39s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ba7467bde3921ed1df0b14630a4a5e6a9b8db38c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-01 14:43:26.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"19af374eb351d91d6bc0add88028904fb2e11dda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-01 14:43:47.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"c5b5ad99c0b1270809249ae2fd3cee91f88ae88c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-01 14:54:06.000000000","message":"Patch Set 13: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/656228f7ea2e46c6a0820b363322c8ea\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/728733752c2b465981e08953e294a38f : FAILURE in 9m 09s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/a884d0785315492aa2177e3416c98514 : SKIPPED Skipped due to failed job openstack-meta-content-provider-master (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"d26c60d90aff76cac5f11f6416d84859b89aaa4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 14:54:59.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f37aaf04e7904727b14b3dfbb08eafab\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/237baf26936b421d80573d473b882529 : SUCCESS in 2m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/acf6d8c593624e5a9a0709ee9cc9887d : SUCCESS in 4m 20s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/4d7786eff8a1446e9875dbb53ef1eb2e : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afaba04c13974b88aaff65192679545a : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/972f9b64c5884d5eba306a40a2877611 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f9b44a21e68497abadbcee4df225f54 : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d22bab7764d4c6ba054b03530d95bde : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e57f1b2f0e46ad9f64d6f5c235c35f : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e73a26bef37409997d68b8f5b494a0b : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":13},{"id":"6cce8acceb9642efa1f29ab9e251768140dd52c8","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2025-09-02 10:44:18.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"494c90ec6d0a500ca8a2fbfb828f41abda39428f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-02 10:54:11.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"4d4395aa0995134755cbb74f0e64a54669e3358f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-02 12:57:16.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"aab06dba6f8f4a123dbe2bd31784548ba339bc4f","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2025-09-02 12:59:25.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"0885488e3d9c5ce7a588e8801c699773a6a50b4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-02 13:12:50.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"d617856184843211a22e53f9c8ea7ad338f9042c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-02 13:13:00.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"0a4c5987ad3035a8690c6bf1e0972e20e358dd13","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-02 13:13:06.000000000","message":"Patch Set 14: Code-Review-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/watcher-dashboard for 957232,8","accounts_in_message":[],"_revision_number":14},{"id":"70ff182ef4bbd2f7deff8756db3ea2d6f8bcab07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-02 13:14:21.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":15},{"id":"92d6e0de0621e473167053529e2abddbe1be5113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 13:28:25.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f9dc083dd0e4e6d89efe882ab452967\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/86a418c23ea34ad8b7349c04cfc02b0b : SUCCESS in 2m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6df3f1daad8e480f82375f649e867bd2 : SUCCESS in 4m 13s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/8ec26f15f60845b9a77420dc78de69a4 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7516f17d135f4cb383ffade838d29ff5 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e0e0a3f47164e708cb7369f5d208026 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20dd29c744ba465f99c601fc6ad8d8ee : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08dbb15e076d4fee847631d41f017249 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d12eee48e134a97b8022613cc41c812 : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e5cca825a944b278237e11310aa54c1 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":15},{"id":"918bbc6815a702d5a0e897f3e28c1d685ce7586c","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-09-02 13:38:24.000000000","message":"Patch Set 15: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"994e5c396aa7e3240fc3aa43506c8b0aeff7ce4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-02 13:41:14.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"3061c39dde99b267555459edc3db39f88aded01a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-02 13:41:29.000000000","message":"Patch Set 16: Code-Review-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/watcher-dashboard for 957232,8","accounts_in_message":[],"_revision_number":16},{"id":"ee9c08edee6842b70396340a85312c580f1461f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-02 13:41:36.000000000","message":"Patch Set 17: Patch Set 16 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":17},{"id":"76feef4681ace347404cd21ce982d89597883b56","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-02 13:41:47.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"ecdd5656856fcffd1e512b2108e100169fde2239","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-02 13:51:44.000000000","message":"Patch Set 17: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/14e82ded64494f758bd7f5bee6548609\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/82a5bb2997d6417dbccb875af9a83420 : FAILURE in 8m 56s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/4e5f5658df284cea9a1384dde8bffebd : SKIPPED Skipped due to failed job openstack-meta-content-provider-master (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"d21af1321e70f0554f36e9cb39dd991a809d86c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 13:57:21.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed9fc40934b3435ab8f87e7fd004e3d8\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/889e94e57bc643eba18df20c605f6f56 : SUCCESS in 1m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ebb288734cbb4fe2b40a56d8d65acbf6 : SUCCESS in 3m 40s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/ad52b1a2ba54432b8c462f3d25f9b04e : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc11c605f9a74451a83a388cf180037a : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b2a60be7da64d13be2c3b16742c2d20 : SUCCESS in 3m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81187c8418dc430e9f0c7f60c912c01d : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01c88b8838414423a8d8ba37d8512c78 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/062b8831284142c7a9cb74220f83822e : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e849c1ffa4944ea59a0296e2aaf9b6d0 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":17},{"id":"5ec8e9c12a00b4a9fc988202d882e2fd92a6ca49","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2025-09-02 13:59:12.000000000","message":"Patch Set 17: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"dc6f9c768689ba7192237068bfe7628345e2fd2e","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-09-02 14:09:40.000000000","message":"Patch Set 17: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"f32aee100c890dc0fef8216cf731df7a586e63c1","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-02 14:29:23.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"72762335346460df1c418685701469bef59a4759","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-09-03 16:28:11.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"8d93d742d0d2a6b80ff72a3dfae2294808a4162b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-09-03 16:33:57.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"55fe77838f91cb57be680e542fbd86e71610949c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-04 07:24:31.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"7afce0951cda84e58f4632237d55a2a0d0052def","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-04 07:28:02.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"2a6f1835d958cf1a6e844367bd311f8d0e7f05c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-04 07:28:39.000000000","message":"Uploaded patch set 19: Commit message was updated.","accounts_in_message":[],"_revision_number":19},{"id":"f91f366d2d93a7cea522436c2d2f935ef358747e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 07:37:34.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b9254361a824f0b92093b6578ea10a6\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/965a8485333c495794a7c99f4bbb3860 : SUCCESS in 1m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa275e38eb374c35bcb58bda4d46a2c7 : SUCCESS in 4m 47s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/de4e9e78860a42b1b18db77b283ebefa : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ebc6d7c287f4a6189b870a5c1f03865 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb1c7e6e26e74b1cb679dde034361e4a : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/085123ec8fa144148376e684b3997269 : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c090dcaf32045e49e753c249155f53c : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f21f43795df4bbab0a6e06111154009 : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bb0e8365fa74c5384b91abe84cf9935 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":19},{"id":"2376c75a23cf437f0b9a2328cc52f2714f44a657","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-04 07:38:18.000000000","message":"Patch Set 19: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/c0bbbf42adb54b06b4795fc4426159cc\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/e779ed8a30ed4692a249f13541f17037 : FAILURE in 8m 32s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/3fca04218d074a32bee36d5184078162 : SKIPPED Skipped due to failed job openstack-meta-content-provider-master (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"9ee3cf77e02c156a15bb5d8d602ef098173f5c34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-04 07:51:55.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"0e85d5dc9e10c1670c1286a6487eff403338b660","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-04 08:02:51.000000000","message":"Patch Set 20: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/6e0653a6132d4f71a8a4024c8c02727b\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/606ba7259d40496cb3b68a8c2ce45867 : FAILURE in 9m 31s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/c9aa24cd36be40a0af7597dfc8009c83 : SKIPPED Skipped due to failed job openstack-meta-content-provider-master (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"e2f13dcaf4c04e9dbcadf879ba1c4d07b7ffda02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 08:05:10.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f09dc1b1d50a40a5a8a076f750973dc7\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d9a51b8f83774ad3b4c147b84a111bdf : SUCCESS in 1m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0db04807eda742a0af0c728a4fe87cc3 : SUCCESS in 4m 15s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/4e319ee5ee6642689a02e19830a9f3cf : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb8fd869a904ed6934b3aebc60db30a : SUCCESS in 2m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29155cd7b00149aaa48754189243ee6d : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8272e275f78e493eb7fd94176a42dfe2 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44294536c3dc4ccdb3d69c1d1ba7af8d : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63d26fc2274e4ad193194934b95e7784 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ed01979d8e5491bbea55fed363fddd5 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":20},{"id":"4d3249004d76227a2d13dc11087c1857bb4bd00b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-09-04 10:31:31.000000000","message":"Patch Set 20: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"e645c9f574d51ae8602b168470164373c3fb1645","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2025-09-04 10:40:20.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"981778584747474b33cbe0ff3c4cfbc9ea68b8bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-08 08:05:18.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"5703a345f8f23ae77c379adbafe72898bf421381","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-08 08:05:33.000000000","message":"Patch Set 21: Code-Review-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/watcher-dashboard for 957232,10","accounts_in_message":[],"_revision_number":21},{"id":"033015f90dac88b14e726fb8705b5438bf72d91f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-08 09:39:44.000000000","message":"Patch Set 21: Code-Review-1","accounts_in_message":[],"_revision_number":21},{"id":"3543232cfceba2116c72438e552304aed0d8bb2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-08 15:22:00.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":22},{"id":"5d49fb2273b30d0215b5c0bc1d8258f9963bc4d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 15:48:21.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfed93ae82ff4e128846c3c9222081d2\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/641f080c5f6c4ade8991bfd2dc084e7c : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5767f67b36846c0ac12e01321baf006 : SUCCESS in 4m 21s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/4501a50eae434ba19a97e936944f5063 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/183b2d91cefc4de48c96e90368c300c5 : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1831c6991ca040539759f4c278e9fac1 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/814ec3804109408f80c1673be4c25cf8 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eccad9c22db84b12b531846a817c3262 : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a80c83802a364eb4b59e99b06b426635 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c493a8cb0ba54a02a1cb52f3e249bc58 : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":22},{"id":"799e996b603bd39d04cecb2102b0f3fc8a7fc3af","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-08 18:16:55.000000000","message":"Patch Set 22: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/2c87d01a951348f795e186e18dbacaf2\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/9157c7c261b14d0a94bcec537020451a : SUCCESS in 2h 53m 29s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/f4f1c6aec3604c799b1a7183a3b65eae : SUCCESS in 2h 08m 53s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"2f6908884cb30a5bfe075dbe1cf3263e9c6d6ac5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-09 05:47:47.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"e082fc6bb0491daef7c9fcdafb800167c33a913a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 05:58:42.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4717a60d267446fbb17787444f8746e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/85269b99e0d74458a60c6b403ee75513 : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d3d2d00ed934a2ca46070f284f8c349 : SUCCESS in 4m 09s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/357f6d2110c64a72b675a089d533ff14 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1c9f95e18df4d469a622210a1e683a3 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d214007d3f9496dbe42f2db702c12a2 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44f43694a3bf467b86e97d8c60d3fcdd : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae2205fa663a4726b26dbbbd4fe4d5ee : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ccc9de190844447afa9cef43fc01f25 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d9fde93d8314a72a0184141177399be : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":23},{"id":"de0dddb76fc168b73e428fd2c3589d6f0f464b16","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-09 08:39:34.000000000","message":"Patch Set 23: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/2b255cece9e2438ea75d3c9002af13c8\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/1b32555dd558473db3fd48f166bea1fe : SUCCESS in 2h 50m 32s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/14b7507114934c1ab9229830ee761402 : SUCCESS in 2h 06m 22s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"90edb189d881773270175181de001ad9057ba630","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-09 09:01:41.000000000","message":"Patch Set 23:\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"5a6564f45012354c94972ac1361c92ac1c0ad859","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-16 04:28:11.000000000","message":"Patch Set 24: Patch Set 23 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"697dd028a15fa58ef9f5531d1bad9b210b2ca296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 04:40:13.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd20b1da56504cb2ab1053c590fad128\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f0a46364393241e187363abce4d4c30e : SUCCESS in 2m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c32747eaa9354f1ba5b0dd03c01589af : SUCCESS in 4m 34s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/ce8edcbda8e547e79b8f7014d00809e2 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e6f297a0355498db264def13088dfea : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4efb824aa1794dd1881b06f8881a72bf : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d3b5dfa473349d58d3de359e4d81c8b : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3b7d24be3974a92ab9f7ee6b988662d : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4a3b4b9001e453db738cb2816e51050 : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04589302d1e8438fa31975c2f7537af5 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":24},{"id":"1843243ca621351f91d866ba660c09b557b50c9a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-16 09:05:32.000000000","message":"Patch Set 24: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/505dd79800e24b779929fa2ad5e5c134\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/c97c394d9f0948b09702bee76e2d9764 : SUCCESS in 4h 36m 08s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/e83d5616aefa496ebdc79649d52d52e3 : TIMED_OUT in 3h 40m 12s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"395be0d44ee5c38957a8733ffd44f0711ced500f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-17 12:01:51.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"78c31e3be3af90e09c5db6e44253318635a7cb9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 12:22:48.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cad572b3b4db4138926bae76bf84212a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0feedb71296941c996ffab16d51455cd : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a10b4c8b6e841d59c8d57420b34471a : SUCCESS in 5m 38s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/36241c8185144a66958f046efad08d3d : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d0c64c54efa476b94f228631a10726f : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4cfe9282cf184af5b64af4fe1cf82592 : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/41536686350e44e181e85bc6f0a732e7 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/628a556e51a448cc800da2f29ab65041 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91ee8c5d044149359a315bcc427d1edc : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4dc021a1f924a278bd5f7d61c95431b : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":25},{"id":"63f6f5610d58b6adcf489046d6574fa9c659d517","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-17 14:46:32.000000000","message":"Patch Set 25: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/2413b693f2de4534aa463e238dc17e7c\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/c6c357603f934e2d9d159c74bfd967fb : SUCCESS in 2h 43m 22s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5113033d2b204efcb2ab9cd1cb48d50b : SUCCESS in 1h 55m 49s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"1b5529c04686f9af568508cf272830eb4c99bb82","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2025-09-19 09:20:15.000000000","message":"Patch Set 25:\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"755bc5e695b7525b9f98fa10a88edb03f49b299f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-19 09:36:13.000000000","message":"Patch Set 26: Patch Set 25 was rebased\n\nThe following files contain Git conflicts:\n* watcher_dashboard/common/microversion.py\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"ff9d7c271065655ed3be701727a4affc2b9ab85c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-19 09:47:10.000000000","message":"Patch Set 26: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/f23c4e26624e400a8f26b83251639d63\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/f9432f193c7342439a9388881d08d5df : FAILURE in 9m 52s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/ce1cba522df94beaa187e14b45a8dac2 : SKIPPED Skipped due to failed job openstack-meta-content-provider-master (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"6d5c0b04a21b82819c76f132625c66b1040a3c01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 10:00:22.000000000","message":"Patch Set 26: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84e3832390cd4bc39c41e8f88e0c2cba\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1ad131a644e44a038fd163004c3b6343 : SUCCESS in 1m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e0dd16ba1454ace98b29ded4ef7f212 : FAILURE in 5m 38s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/fb5405f46eec4db8a9f58d8de4418b4d : FAILURE in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c7143274ce5411e8d908b0992049e22 : FAILURE in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc46a385ea84439e8bb0aa94a6789a27 : FAILURE in 4m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7db73bd0e9aa4cea8e6515cb29dd32e0 : FAILURE in 6m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a953b309e58b4b898f20dcc1e79b01ac : FAILURE in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f69f84978ed4bb1b94684ea785ea6f0 : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44a3afa3467f4d72a1d8fc3d1bd5e50b : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":26},{"id":"56abd7f3d46555a71f6aa0b30408dcf0623836ba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-19 14:41:04.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"faeb34af3f13e316fe2b13305b2687572fbce82a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-19 14:49:47.000000000","message":"Patch Set 27: Code-Review-1","accounts_in_message":[],"_revision_number":27},{"id":"c6eefe56ed44d9eb179a89f7f6af5968c073ba74","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-19 15:15:33.000000000","message":"Patch Set 27: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/8ba9d69e5ce34c9093fc14fce9e079bc\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/4ba0413b4c9b4c768ab65e02bda577b0 : FAILURE in 33m 20s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/bf5c96ff73cd493bb018b1f3f090af1c : SKIPPED Skipped due to failed job openstack-meta-content-provider-master (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"d95c4af7808ec9cfec5d0b0abda302ddb4b14d27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 15:17:51.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b9ba47b8f414b928643ee02cc7ad5d7\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/78966a42adfb440189ba2ed68c1330b5 : SUCCESS in 2m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1a729b2c4ca746f7981c7fb6c4f55e52 : SUCCESS in 4m 29s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/1e7deffc970d48049f2f14beaca25115 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee6afc817e414523a5bf26f3d0bb125c : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17ea47bd12154b13b230753e535ff764 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48fc9df9392d439ea867a17aeeed3b5e : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb61152cde3544e283dd63529596a4d6 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/098b0a844bec48568779a3e1b17f06fc : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/327efd790daf49c58d64d8ff6e72c375 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":27},{"id":"05eee18a28931a1ae3e4ec7f3d4f2da897d4201a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-25 05:23:39.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"370d3bdc7ffed93f5d1c8aec6d57473013ab7e58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 05:37:31.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a434ee273a824315b5946f19dbf38c30\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/147f1a4b98e94ff8a50edcffff2bf222 : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b35aa359a36640c9b0d40d284c32e002 : SUCCESS in 5m 09s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/ce0d617ba25742aab7d2f16264dc78a0 : SUCCESS in 5m 00s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/629154d7fb9e495c8e2cf7dd455961ab : FAILURE in 4m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/618fefb5fbc646d3934d569ca0f7a21d : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acf0aa621d5645ec8e5f0b9e923305b3 : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/742805e1c5d8421385f67c2811b0d0d1 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7eb7a2bef96a4c0cba9b5dfe9964f720 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0221c811e2684f21ab459ac952f72406 : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43efc1629cab41f7894db93e9640b263 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":28},{"id":"55fd3d305608441e2cad207b373786d2fbaf3457","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-25 08:11:25.000000000","message":"Patch Set 28: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/721669282da54bcf9bb9292a808ec9b6\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/305733a1be2442a38816cf4e7d6bacdf : SUCCESS in 2h 46m 20s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/2b522872587b44e19efc965cac007e6e : SUCCESS in 1h 56m 05s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"de0b2d6c898e58ae62f6fbed45c614961dcc6536","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-09-29 04:59:16.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"7f39b7fe46f200c0f0d2d12e94397aa83940be45","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2025-09-30 09:10:40.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":28},{"id":"68863a3836d63e956dba368842361530e8d43489","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2025-09-30 10:03:07.000000000","message":"Patch Set 28:\n\n(6 comments)","accounts_in_message":[],"_revision_number":28},{"id":"31410901559bfc20dad77564010b16b43b4e8f31","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-10-03 03:52:57.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"141dddd403df9427865c1b69db732da3a02264fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-10-03 03:57:29.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"2f54c5f4e57268fe0e879860761bb89e356702d4","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-10-03 04:00:29.000000000","message":"Patch Set 29:\n\n(2 comments)","accounts_in_message":[],"_revision_number":29},{"id":"8bfaa92658acc814bf5828a8f9a72624460d2bb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 04:10:09.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8fc7a6d516343f18c38168e8fa0ce31\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/16b07ed0f2994b5cbcc20b57a92b84c8 : SUCCESS in 2m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f0a6f671dd1f4d02be0467e10e06969f : SUCCESS in 2m 14s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/4e03c352bc6446adb0decf74d47d5e6d : SUCCESS in 4m 34s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/8430cfc483444ac0910d8a78a2001263 : FAILURE in 2m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9de9baf37cfb45ae9653f0bc466f412f : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3b7835a2ad843a386374720c25ef9e6 : SUCCESS in 2m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10b8a70f27814c5db7fd0558300d0188 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b647aaa91c8c4240a2cb2b89d5f5c61d : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31b76aa50724613bf5c096f0422c291 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32c202e396414782951c8bd2a162d5a0 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":29},{"id":"3223545839db614fed6a255782ea27d6a49dc432","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-10-03 06:43:56.000000000","message":"Patch Set 29: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/719fc315d6f54069aaa1f14195bc4855\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/676d3a12474b41d79885185fe2bce4e4 : SUCCESS in 2h 44m 46s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/77a3de433e404372a5b9b097087f487e : SUCCESS in 1h 59m 41s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"747d896fa56d5a3e222e4992fcc43042981b1dd2","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2025-10-03 08:14:40.000000000","message":"Patch Set 29: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":29},{"id":"d4a68b090c0713d590184f7980634719be79e9e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-10-06 07:03:50.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":30},{"id":"8187d3b182dd4b6555558df211c6a799572f622f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-10-06 07:05:41.000000000","message":"Patch Set 30:\n\n(3 comments)","accounts_in_message":[],"_revision_number":30},{"id":"2b626c92c66fc0dec37c75da9ca36379151c4cb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 07:16:06.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4d64d7a5fc342cc94025a05f5f3a1ca\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/785d475701324e0b9ef941f2e8a75c99 : SUCCESS in 1m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44a1066f408f4f56ab88e7e5d65210eb : SUCCESS in 2m 35s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/6a4f2df09dea4924b43827dcf048e3d8 : SUCCESS in 2m 17s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/2b27f0f431d1422da7f8abb740bd5eb9 : FAILURE in 4m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d5a82901a3643d3aa8833c59c304f5a : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51297c88a0dd4a46a36de85284d2525b : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f856ec3fff84de7a208a0cd2bda7bbf : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a2ca8f36ebe49149ff5e3118db69fbf : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2029c678f62e433f9b9c29ab84984305 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd8d3a424dae4b4aa51364508a1ca75c : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":30},{"id":"f10ed2b9567098c09d51fdaf94778399665a49d2","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-10-06 10:04:47.000000000","message":"Patch Set 30: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/33ab46dd2fe04263a75be76c75acd757\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/90cb6218e95d4a95937acd17be68549a : SUCCESS in 2h 53m 29s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/6c34186990384710ab11d09297479343 : SUCCESS in 1h 54m 45s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"fc888a62e65ade12ac21f27d87d09403c42fb274","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2025-10-06 11:25:51.000000000","message":"Patch Set 30: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"b6488e35e0e73ffb0f1cd36e5e705843412b0b5d","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2025-10-07 08:06:21.000000000","message":"Patch Set 30:\n\n(4 comments)","accounts_in_message":[],"_revision_number":30},{"id":"4c4d560680313ed43ee02ee34cc364feb2261a90","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-10-08 15:08:16.000000000","message":"Patch Set 30: Code-Review-1\n\n(23 comments)","accounts_in_message":[],"_revision_number":30},{"id":"e17de8b7d51d2db482f67bb6ea37f9dd0d1ef7c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-10-10 15:03:33.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"7809b5389bf66349108172fd95231e4f14d3bf0a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-10-10 15:10:41.000000000","message":"Patch Set 31:\n\n(22 comments)","accounts_in_message":[],"_revision_number":31},{"id":"911ea878c5cfc4cca87d4b0066ab279457d525fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 15:20:31.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/960ec4f8365d42c099f1cca23b5196d7\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c91de1cbaecf4cb3839475357f81480a : SUCCESS in 2m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64618c4110384f11bc11a8b30f1188c3 : SUCCESS in 2m 25s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/9929ac4ba7c9419a9d3c7ecbf66e401b : SUCCESS in 4m 34s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/25130121d7894238acc758f06fa8fcc3 : FAILURE in 4m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75e6fbe6493041059351d282f39f3f1d : SUCCESS in 7m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8fe5c33bb8724c3e9ca04c02955d37ee : SUCCESS in 2m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8c17bad5e3342eabba569a82116a21c : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a28496115b84895bb1a01b0bd2ae27d : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbe751023e6a4d829c31c69fe11c7bdf : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78e8dcdbe3c241d2972d30fdc7fad0b4 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":31},{"id":"f25606d2f57408934874e6d2ed1ad83bde76dba0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-10-10 17:58:39.000000000","message":"Patch Set 31: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/9be49ba8d5fe4895b7eeed92f4d447e5\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/170ee1ecd8d140239d61182e9090a7cf : SUCCESS in 2h 49m 00s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/d04300d47959408cac64163927747cb2 : SUCCESS in 1h 56m 52s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"847fb78c530878a30f135922c8e76b844307b531","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-10-24 12:06:17.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"ba4c7c18b72e558142f6e75ad1229549ee7fb77a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 12:18:54.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c553ecb7db3d49e9bb49a6db77358cca\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/93c4116d864d4334a77240fa633d98cd : SUCCESS in 2m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2bfa0dad492c41fd9fc81704d30838dd : SUCCESS in 2m 37s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/2c6cff82e539412d96d82734288f7296 : FAILURE in 4m 27s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/644144fed85a45a59d8d3e320a7f5305 : FAILURE in 2m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77b714ac1bef48269295590597d8ab4c : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/162faf6a8a374ef48ded0a83e43c4187 : FAILURE in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b88da2220d9401f94766b62a02ead31 : FAILURE in 6m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8dd4f16b308c423d810b0466ecf45688 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a87dfd22219b48bda21f2987d10eee6f : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c830ef8165364a6ab150fc046667194b : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":32},{"id":"5779a5ff9ea81d852aa9c9f8f1c9871763c4688c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-10-24 13:15:27.000000000","message":"Patch Set 33: Patch Set 32 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":33},{"id":"2efef96b602bec297c2d202557f89644d8b7f369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 13:30:04.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c971877608454c8784fce3cafded710e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/dfd9ae725e844eb38e850b65bfc98d6a : SUCCESS in 2m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5bafe581f414e68a10f17105a61be7e : SUCCESS in 3m 49s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/75d6587d9ec14bf3a45055b808d4ba88 : SUCCESS in 5m 37s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/2f990b86ec45400db281899a252bc91e : FAILURE in 4m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f642fb02276f4ed6a55fda29e9daed08 : SUCCESS in 2m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17a23953e77b475a911b0e0b01bcffab : SUCCESS in 5m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/959db2fc812e46a78cbd1729785f7df9 : SUCCESS in 8m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b9afbee81854cacad8ee3c822ba32ee : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abf3ea8558f64c9c89333315458729f6 : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8e477dfc8ae4fd2b2611f2974d78716 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":33},{"id":"9e6020f882086fd4c28cbb0d630f78bcf9eeee90","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-10-24 16:05:04.000000000","message":"Patch Set 33: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/d7a633a9bdd84a94b0d9bba667bbfd3f\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/35a62154994143eb93539cd214a01cd6 : SUCCESS in 2h 48m 24s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/4d1dce633f1a421aae72cc986074c0eb : SUCCESS in 1h 54m 15s","accounts_in_message":[],"_revision_number":33},{"id":"201ac643c997c625f13729d92a0023bfaf69a7d2","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2025-11-04 10:09:35.000000000","message":"Patch Set 33: Code-Review+1","accounts_in_message":[],"_revision_number":33},{"id":"b054c398e3a7dbacebdd8468a4ff1a8f33f5414e","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-11-04 12:08:38.000000000","message":"Patch Set 33: Code-Review+1\n\n(8 comments)","accounts_in_message":[],"_revision_number":33},{"id":"94ea97344e76d0c6005ca9fe2fa81bb0c8b31d94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-11-06 05:15:19.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"7e8de1f232fef1b8641dee1a78bc3f273a5658db","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-11-06 05:15:53.000000000","message":"Patch Set 33:\n\n(2 comments)","accounts_in_message":[],"_revision_number":33},{"id":"5b700faebfaaf5785ba9344d17158bd5b83dabf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-11-06 05:16:03.000000000","message":"Patch Set 35: Patch Set 34 was rebased","accounts_in_message":[],"_revision_number":35},{"id":"2f33c9fcd6e936b92f472aeaeb4c491987bbb865","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-11-06 05:30:19.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"c857c3972793cdee9fbd310945667f44fd59e56b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 05:57:27.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc46b4e7f73c4253aede9d2b1d1febcc\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f2a42c3a036440c68bd9348f793439b0 : SUCCESS in 1m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1873033f4ef44e57aaf62f0603eabcdf : SUCCESS in 5m 11s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/db4436240c58498cb613bf4b3cc699b4 : SUCCESS in 6m 24s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/2f302f5d13144ba491b5f42e502b6917 : FAILURE in 4m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7055101a3db447eaa00b7c2fa64a535 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7420f04dea3420da28ee2b0c8b4346a : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a1933f519aa14f1a89ed504fc73517cf : SUCCESS in 8m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27642454d38642268a63db56420d21c0 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d789d3a618744a828275cc220721c8aa : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9418abe952a4aa6978fb44292baf719 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":36},{"id":"173fbb93fa0139461e1330b748c7f365f72b66bd","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-11-06 08:13:09.000000000","message":"Patch Set 36: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/fe1cfb1c876742c6963da538aec275c5\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/a44d23c2ca644e4a8753a01a3609077a : SUCCESS in 2h 41m 44s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/fa3e012e9a7547c5b8259562c1da6f12 : SUCCESS in 1h 53m 51s","accounts_in_message":[],"_revision_number":36},{"id":"83c15644d59d7b5edd392d771d4dda461aa9a91b","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2025-11-19 10:33:51.000000000","message":"Patch Set 36: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"fbec116b80147af4d6c9ff4452c5acae8baaa4e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-11-20 15:41:34.000000000","message":"Uploaded patch set 37.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":37},{"id":"948aaa397e849b2a0e2c3e8f43e81ad64d772e56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 15:52:30.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/027ae415e5b84a899b859dbdd3b41bbd\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a0a271d8e3f3470b9bff257299b4ac2c : SUCCESS in 1m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a486cea6ca5b4742a12c6c2ed71f2edd : SUCCESS in 2m 37s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/b3168e33e9214dc0892c0dc339f226d5 : SUCCESS in 2m 35s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/06083a2a129f471c9fd3d1cbe07340b6 : SUCCESS in 2m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eaa4fb75ea14e89987cb2909176d859 : SUCCESS in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c76514176f245749446469b55be7e2c : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a38e13a9a3484a5eb5f2094e0f74a5b5 : SUCCESS in 3m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61e73648234349f281b84f9b442e1a52 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d06911179b0456f917365b8c11bd271 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7497c60a28f4b1697c8107904890e0a : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":37},{"id":"e579bcd2575d3e0a68fcb17dea3fb09df95062ca","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-11-20 16:09:30.000000000","message":"Patch Set 37: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/adce2f3ceabe479ab0c495c660b5c078\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/8b44c6bdf19442e3bd4f927671524bea : FAILURE in 26m 25s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/0768446307b54abcbe5991fdd607bb2a : SKIPPED Skipped due to failed job openstack-meta-content-provider-master","accounts_in_message":[],"_revision_number":37},{"id":"ca56d701388ca74db085f2efbcb838c9b0e0b780","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-11-26 05:46:11.000000000","message":"Patch Set 37:\n\n(6 comments)","accounts_in_message":[],"_revision_number":37},{"id":"db7eb94b633970eb65afc68f7d00925c19633b7d","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2025-12-01 00:41:57.000000000","message":"Patch Set 37:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":37},{"id":"e0ebac1f2cb7ac5ceea685aa9169632489cc06b6","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2025-12-01 00:53:37.000000000","message":"Patch Set 37:\n\n(9 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/83c8ce7971a84e548c28fdd39e487e86\n\n- teim-code-review https://zuul.teim.app/t/main/build/0b6236607d6b4b1381eacce245409056 : SUCCESS in 11m 27s","accounts_in_message":[],"_revision_number":37},{"id":"2b558ed8a757592936c253bc077bece7529ddc71","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-02-18 08:38:29.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"ff3e1932fb664ac55704c0274b88791fa87b8f63","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2026-02-18 14:06:01.000000000","message":"Patch Set 37:\n\n(2 comments)","accounts_in_message":[],"_revision_number":37},{"id":"642af6aff6d2c89de4b47bb149fe704d3e290e94","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2026-02-18 14:06:40.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"69e75a28edf020dea6cc1d980f1e51eeaa9df3e2","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2026-02-18 14:19:30.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"8c0b59ce984f67cc8bef42dbe56db400c97dc61b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-02-20 14:49:23.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":38},{"id":"7133c869b044719f3f426cee1564c108f0021324","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-20 14:49:34.000000000","message":"Patch Set 38:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":38},{"id":"ecc138a644a0263487bce735bb31c00008d7df3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 15:02:14.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ecb8b8bfc614b7faf0da1ec05e1cb99\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ec64323dbf5c49dcbd08e8703e14eb0b : SUCCESS in 2m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a62074690cac45a1a315c48daeccf893 : FAILURE in 7m 29s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/05fc1a0f4cd043acb8c49e9212b3eb43 : FAILURE in 4m 06s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/a5fcbe8fe59b4f108e022c7520f1c3a5 : FAILURE in 3m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd282b225e734ff8b7d73d4836eb074a : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49d30125e52e4035851c697891cde8b9 : FAILURE in 5m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/724306dfa666422792a4058f1c317449 : FAILURE in 6m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/968a9561edaa4c8581cad531b54f2499 : FAILURE in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c05e5e306db545a2af9f922b58a12241 : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f71cff672cc4a7ba70078251b03963c : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":38},{"id":"5154d7490a2b597bfe7d23b7c81b97467b5fd399","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-20 15:14:58.000000000","message":"Patch Set 38:\n\n(8 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/beb529bb394448fda852f3d8ef2d3d15\n\n- teim-code-review https://zuul.teim.app/t/main/build/6f3e41c8ad894ce4875bfb20571c50f6 : SUCCESS in 25m 15s","accounts_in_message":[],"_revision_number":38},{"id":"0f8903d9861f66c3e5f5f1f51fa257df026dbe8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-02-20 15:39:05.000000000","message":"Uploaded patch set 39.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":39},{"id":"c407b2e018b10a42f895ac930c53c7ab2120f833","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-20 15:39:11.000000000","message":"Patch Set 39:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":39},{"id":"d67836d8d3285e5f6ee2e56a90632a98f1c9f209","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-02-20 15:43:48.000000000","message":"Uploaded patch set 40: Commit message was updated.","accounts_in_message":[],"_revision_number":40},{"id":"c80906dfad27cc7279d2adb9b2531efbbf624299","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-20 15:43:56.000000000","message":"Patch Set 40:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":40},{"id":"ccc306592bd7273e8d5a471850b85a44e582d2de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 15:55:50.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51266c360380417cacdcca640426e99b\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/91091e3a48fe458fa11cb3bb18d5a2f4 : SUCCESS in 2m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ffa37f86acf445d97c6b5aa70278763 : SUCCESS in 5m 47s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/ef6ccbb6b8c04d15841ceab3a7996ab7 : SUCCESS in 4m 24s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/3fad547727c74bd9be3e9e9ff84cb564 : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/548de7f359ca448a86f7b445e1b6084b : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/831e560cfe974d029ba8fccfe0e6d5fa : SUCCESS in 2m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0cbac1b8b4b840c8ad0611c962c3757b : FAILURE in 8m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3c6786985ed4b7f9de22fe13838adca : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/059511787c384e37a82e41238a9faf2a : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33c5e3b8d0c74eb98395d1500f47919f : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":40},{"id":"abeb21a54acd9ec32e209f86890edf4438d84fd8","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-20 15:56:58.000000000","message":"Patch Set 40:\n\n(9 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/4918f7e19ab6417babc470ce2bf69140\n\n- teim-code-review https://zuul.teim.app/t/main/build/de202cc7b57e4527bbd59a2460e33498 : SUCCESS in 12m 53s","accounts_in_message":[],"_revision_number":40},{"id":"a27dbf9f8f4cfe5149ff6f2755710f31a4353f7c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-02-20 18:26:52.000000000","message":"Patch Set 40: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/2564ebe41f1049e49e5de017f41a0f68\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/99e0b24c89734e3ba59913277ac4d5fa : SUCCESS in 2h 41m 51s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/a4b9c5ae9b0941f1a80f04e27d2c4f24 : SUCCESS in 1h 59m 13s","accounts_in_message":[],"_revision_number":40},{"id":"8a12529de0fc36b4ef1bab742369e4f0d8cf9c15","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-02-22 07:42:53.000000000","message":"Patch Set 40:\n\n(2 comments)","accounts_in_message":[],"_revision_number":40},{"id":"932cfbc603a1e7f0dd5bd5b6a60974c7d9629692","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2026-02-24 16:54:27.000000000","message":"Patch Set 40: Code-Review+1","accounts_in_message":[],"_revision_number":40},{"id":"cf5770ef24939fbdc459a03a9b1bc6d75d2c0ac8","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-25 18:30:48.000000000","message":"Patch Set 40: Code-Review-1\n\n(15 comments)","accounts_in_message":[],"_revision_number":40},{"id":"cb85b8994073c52a5d99aae84c144a69325e1f26","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-25 18:52:41.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"bbca1ab015dc7c0e6406f9342274aab9782e7e54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-02-26 13:53:52.000000000","message":"Uploaded patch set 41.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":41},{"id":"a62adb01c0e6d0f1cc101ea2bff8c91d33bab39f","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 13:54:18.000000000","message":"Patch Set 41:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":41},{"id":"95ad43b894ddf01d938e9c5e708e3d757ec4a26c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-02-26 13:54:56.000000000","message":"Patch Set 41: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"ef6ec44091d0cff5b35a3528475a10aa7aced3db","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 14:04:14.000000000","message":"Patch Set 41:\n\n(8 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/c262f78c91214a8284187e7b5b77394e\n\n- teim-code-review https://zuul.teim.app/t/main/build/3e47b1dbbc5c4cadbea1c89074cb82a7 : SUCCESS in 9m 47s","accounts_in_message":[],"_revision_number":41},{"id":"34679ffd0d6f188d0b177c7069f3d778fc52bfc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 14:07:17.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bffd1a0fda5842dea326fcb75463e3ec\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/45433e251e8f4f79b0f97c18f29e0617 : SUCCESS in 1m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/698bd7280ee144c4bfbac315ff201b36 : SUCCESS in 2m 46s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/3b18463593fd4560923a1f372ed8ce2c : SUCCESS in 4m 28s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/1555f139743648b3bc865c1cce920803 : SUCCESS in 2m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9910b890ee3443da6d08b0488b40295 : SUCCESS in 2m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de80eeedfbc14f498be913298bcf4b9f : SUCCESS in 2m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/30f2f19d6d7a4e2d914565e318c11ac0 : FAILURE in 3m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29af86413c584c919c9bb5b2bb2cb4a3 : SUCCESS in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e685ff163a1045ab984ca81fe7a651fe : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac38dbc837b84a75972c2e306920137b : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":41},{"id":"b2bd30c359365dae54c9a54e1b0946fe1f07b70d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-02-26 15:03:37.000000000","message":"Uploaded patch set 42.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":42},{"id":"f7a84ac72729e254711b6a71a4eeb8a412a4a5a1","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 15:03:43.000000000","message":"Patch Set 42:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":42},{"id":"ce6cfa0c5222ca4db3a0e187412cf94094c505fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-02-26 15:05:04.000000000","message":"Uploaded patch set 43: Commit message was updated.","accounts_in_message":[],"_revision_number":43},{"id":"613c52acb3e139bb795542ba4a03b1d751791f01","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 15:05:10.000000000","message":"Patch Set 43:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":43},{"id":"7f56c772b4a79ffd5a07914d9587488585113fe4","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 15:16:42.000000000","message":"Patch Set 43:\n\n(12 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/c627237a696a459fbecc6070368c52f3\n\n- teim-code-review https://zuul.teim.app/t/main/build/e3aa69981a204b2ca4ddb2103eab5655 : SUCCESS in 11m 23s","accounts_in_message":[],"_revision_number":43},{"id":"00de9102b592f80d4eb07e17dc114a3ae5baf141","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 15:17:23.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c3739c341cf49ec82b190670a08ef0f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3fbf432fb7874d70ad5c1a195c122d22 : SUCCESS in 2m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/39dc4e7f0bc74da49728cbd650df3a4e : SUCCESS in 4m 41s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/1ff5a08362484ea79a2d9185cdd24f9d : SUCCESS in 3m 37s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/1e2428daf7a148baba530d68898445bb : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a5c29306d164e86956b8d4cb55f123a : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14c3c2f628954c59bd63dfa3e5ddbb34 : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3094f38d4b1f481889f8321071d3b0a8 : FAILURE in 6m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1be31ab5f29740f88f8cdaca09c86055 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b2dac114a5f4cfd9c5f2ab20738691c : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2205b016e40748668b762e4d7215f861 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":43},{"id":"ffff5a8378c3a631db8e5ff1502fcb009d4330d5","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-02-26 15:18:26.000000000","message":"Patch Set 43:\n\n(11 comments)","accounts_in_message":[],"_revision_number":43},{"id":"4c7ee07edad1fac8000686f71f370e6848388a32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-02-26 15:18:43.000000000","message":"Uploaded patch set 44.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":44},{"id":"5cad0a4bb8e492217f02b1019568c3ce78fda86e","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 15:18:48.000000000","message":"Patch Set 44:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":44},{"id":"815d4eceabe995e091173355e3f9bf6918aa677e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-02-26 15:22:45.000000000","message":"Patch Set 44:\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"b031446a0d09740f7aa93d35f39f75858acd12f4","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 15:29:18.000000000","message":"Patch Set 44:\n\n(10 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/de68e193d8d0442c98fea3a63e7f41be\n\n- teim-code-review https://zuul.teim.app/t/main/build/22fc973b27c5449795d8ae319758f348 : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":44},{"id":"a6aac28a70e97eb3394a5f0b1dfd022e550ffbd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 15:33:51.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/497f01e56a7147dea19ae7ea5b92bf23\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6243c9629fa242e294d6f3119ab83e83 : SUCCESS in 1m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e514c39e1f1c4fd1bb0dd793ccc17188 : SUCCESS in 3m 46s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/d6c766eb0a164979b8b6fc0096039894 : SUCCESS in 5m 54s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/68ac1706defb4fb8a2aaf355dcc52efd : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f7ac7d8b4264047b9c9022602295b25 : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/045d9a6f4b9743eb8a4435f01923d387 : SUCCESS in 4m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6e0140f2e4dc466b8a297b6d5976d21d : FAILURE in 8m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f665e6ff24ee4e0daa3c61286169ce7c : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22951916be1c4171baa29b97ff5f095f : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fa372dc1f7b40208c157bae8433a461 : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":44},{"id":"77d7a5581009bdb6ebc376f2ec98c41b04bcada0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-02-26 18:23:05.000000000","message":"Patch Set 44: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/9250bcd27a9441b8bbc9ef2bfed812c1\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/90149d4c984940f4b27cba0fa3325005 : SUCCESS in 3h 03m 11s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/1233f70724dd4ec38eeecfea9c7fdb24 : SUCCESS in 2h 04m 39s","accounts_in_message":[],"_revision_number":44},{"id":"72a2089ade5b16ecb858dfe4a819221f7257eeac","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-26 20:02:33.000000000","message":"Patch Set 44: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":44},{"id":"7d3d0986fb1dc653e515a4351eb7d7b40b816512","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-02-27 10:59:20.000000000","message":"Patch Set 44:\n\n(2 comments)","accounts_in_message":[],"_revision_number":44},{"id":"90f5b36b047005e07fe03ddac25057c9ed37e04e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-03-03 07:33:38.000000000","message":"Uploaded patch set 45.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":45},{"id":"2ecb303b11619232d12233915c8a880d8630da13","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-03-03 07:33:51.000000000","message":"Patch Set 45:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":45},{"id":"71f3046d41eea82d04eeee796017e9477ea9bf21","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-03-03 07:34:31.000000000","message":"Patch Set 45:\n\n(1 comment)","accounts_in_message":[],"_revision_number":45},{"id":"a34cbf1cd07b2a57b758783f1eba4945f418c09e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-03-03 07:36:49.000000000","message":"Patch Set 45:\n\n(6 comments)","accounts_in_message":[],"_revision_number":45},{"id":"499e3b7d6942ab1a0e5e1917235c4a76914ca200","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 07:50:37.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7184035a69f145a5bf7514a048c22538\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f05a39105714a7cb356fb5d28340130 : SUCCESS in 3m 49s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/06e2155582cd43de9d5782fa2169ebb3 : SUCCESS in 5m 36s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/c5ba90aca79f492ba899dfa6bec0dfb9 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19d6c7e7443a4370b2038efcb1837385 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28879a175bbb42e9819be8e0738c2759 : SUCCESS in 7m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/29da3f4cfeb04a5f9efc53b99726dd17 : FAILURE in 5m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65b6e56d7e6744f4a63dc05b67b6ac44 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1c868624dd42d38955bff142dc8555 : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2161ed7f4a584a8bbcaacc4cdac7bb23 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":45},{"id":"17651796a5f42d3d00e4b3ffa84ae6a7325f98a6","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-03-03 07:56:39.000000000","message":"Patch Set 45:\n\n(9 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/7dd199ee9d8e46f2b99d11bd8be007e8\n\n- teim-code-review https://zuul.teim.app/t/main/build/306155e958494d34b5edebe8c439c1c0 : SUCCESS in 22m 40s","accounts_in_message":[],"_revision_number":45},{"id":"d1c643f1e14b68675e801398cf2c30067183044f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-03-03 08:37:38.000000000","message":"Patch Set 45: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":45},{"id":"0ac15ee553a9620a3ff9dd61d29517882282bfa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-03-03 09:58:10.000000000","message":"Uploaded patch set 46.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":46},{"id":"e58172c0e18443bda2f9efb367bda9c4b1a057dc","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-03-03 09:58:24.000000000","message":"Patch Set 46:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":46},{"id":"948517ddc4512267e989cfc42c0ef95723e41f9c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-03-03 10:04:43.000000000","message":"Patch Set 46:\n\n(1 comment)","accounts_in_message":[],"_revision_number":46},{"id":"0f7dd92b5fb85225154e1a71e8f82b0b4b2fc918","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 10:10:04.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bd63fc101a445a394aac378f18ef503\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2517d634ed74e059448097b963ffbdd : SUCCESS in 5m 44s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/28d47d160934423f80a3eb07486ce490 : SUCCESS in 2m 23s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/73c38f1239974a5fb3e82ba9282faf82 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6025c42615747b19b6edb4412bd590d : SUCCESS in 2m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8fcadf5fb91a4f33bbecd7e98ec039f6 : SUCCESS in 3m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/de599a3d1984425e97a94614c15fcce5 : FAILURE in 3m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10463030e1ca45d4ae6fa7c267be4ff6 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05a9649704d94a8e923ed9ea8ef0eee2 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d88e500b3ccd4e209ee15e420fcd773e : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":46},{"id":"67b914539f9deaeaaa423390da1c3dfdadd115a4","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-03-03 10:23:03.000000000","message":"Patch Set 46:\n\n(7 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/b9b3b3eb32254ee6be3dd126428b8837\n\n- teim-code-review https://zuul.teim.app/t/main/build/fd0e68917f934f8f9b6df57e534f924d : SUCCESS in 24m 28s\n\nWarning:\n  Job teim-code-review: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path watcher_dashboard/release_notes/notes/skip-actions-option-a1b2c3d4e5f6.yaml in the commit\u0027\n  Job teim-code-review: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path watcher_dashboard/content/actions/templates/infra_optim/actions/_skip.html in the commit\u0027\n  Comments left for invalid file watcher_dashboard/content/actions/templates/infra_optim/actions/_skip.html\n  Comments left for invalid file watcher_dashboard/release_notes/notes/skip-actions-option-a1b2c3d4e5f6.yaml","accounts_in_message":[],"_revision_number":46},{"id":"248382c83814da79c073893e762f3db6da645fd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-03-03 11:04:55.000000000","message":"Patch Set 47: Patch Set 46 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":47},{"id":"3e5d05920490da10dc94f134c0ef9d2219721547","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-03-03 11:05:04.000000000","message":"Patch Set 47:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":47},{"id":"c582b375c7c0e6af4985521da5a14564bde3c165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 11:15:48.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baef30f6071c4d5d8f8bfe03bb850838\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0dd5512788e946019c0d403dd27f3e2d : SUCCESS in 5m 27s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/ef0cb2ef908149cab345efc429c94c58 : SUCCESS in 5m 34s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/e7517df569db4b73bf96c9d555e07387 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87c85bd8418f48c4afb4df9bf09615d5 : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/639191b99eb942e98b8175ef153ed9e2 : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2f5e9000370e459787ddf374f16ab56c : FAILURE in 7m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1281b2f642cc48d7b909b5468b82c98e : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/513ffbcab27847028cdb914e6788344e : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc3953dcf96d4433b3903874dea7bb53 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":47},{"id":"3eb04a06ff008cfaae62dfe04f25d7b115f86768","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-03-03 11:36:16.000000000","message":"Patch Set 47:\n\nBuild failed (automatic-ci pipeline). To rerun just this ci comment\n\"teim-ci: auto\". To rerun other ci jobs comment \"recheck\".\nNote this ci does run on \"recheck\".\n\nhttps://zuul.teim.app/t/main/buildset/e56e112b57af4add978c89ee8e5d20dd\n\n- teim-code-review https://zuul.teim.app/t/main/build/9f33d695969b4ad2bc7435b4e34fe69d : TIMED_OUT in 30m 50s","accounts_in_message":[],"_revision_number":47},{"id":"a7e9af0284c552ab954b7e6e29d32359d01111bd","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-03-03 13:44:45.000000000","message":"Patch Set 47: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/01c0ecaf7ace4d9587186ffc6194b7ac\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/a8bc1be752a346c79125991c600f7c73 : SUCCESS in 2h 38m 33s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/fc9bbf66282544c8886d9160e73c5655 : SUCCESS in 2h 00m 26s","accounts_in_message":[],"_revision_number":47},{"id":"adabc2734fd6d4c6d6f4d13bba4777f35b0a4e21","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-03-03 20:10:08.000000000","message":"Patch Set 47: Code-Review+2\n\n(10 comments)","accounts_in_message":[],"_revision_number":47},{"id":"29e65cdcf2cdf61a23dc814228f0bfdfe2fcddc1","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2026-03-03 20:36:24.000000000","message":"Patch Set 47: Code-Review+2 Workflow+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":47},{"id":"2199cf8690a59bc8a78f76e91ee317b1aadd3818","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 20:36:36.000000000","message":"Patch Set 47: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":47},{"id":"9165e3a1eda43f4f6c7e3c6d489be2a134a65f15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 20:50:06.000000000","message":"Patch Set 47: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb794279b6fe409a96d651b149ea088f\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df755abab3d045b49ee3e46b689bdf05 : SUCCESS in 5m 17s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/c73fbb55abe04cddb356ea2b6683b570 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ca24935e40348d19006fa42c01ffcbb : SUCCESS in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3604f7a7f9024df0b8e28827a01ff572 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95f6ebb36de04d3883ea8fa1173c08fa : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99f987a6c5b648509d02122e4aa5f8b9 : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":47},{"id":"b0bd4a94682fe45abdbb222781768238d97b75b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 20:50:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":47},{"id":"868ddc89751514493cf08eaca9fe32b76d7b97de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 20:51:54.000000000","message":"Patch Set 47:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52570d8e4cf342e384457fec90f17d63\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aacb283e8aea45fb9e240c04cd59305b : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/436509f0ef8c4044920aa012e6a11a27 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":47}],"current_revision_number":47,"current_revision":"3a884013ecbca509d2379b090f2a16bb6206d2d5","revisions":{"7b6fa36b073714cc3c044acecbf928543008a99c":{"kind":"REWORK","_number":1,"created":"2025-08-21 14:51:30.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/09/958209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/1"}}},"commit":{"parents":[{"commit":"4a0f64b5cbd48af7c44788e0f86e313a6e7865a5","subject":"finalize python 3.9 support removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/4a0f64b5cbd48af7c44788e0f86e313a6e7865a5"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:51:19.000000000","tz":120},"subject":"[WIP] Add option to SKIP Actions","message":"[WIP] Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/7b6fa36b073714cc3c044acecbf928543008a99c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/7b6fa36b073714cc3c044acecbf928543008a99c"}]},"branch":"refs/heads/master"},"9a2b32b732466a8360f382eb33cef534f2f87595":{"kind":"REWORK","_number":2,"created":"2025-08-28 12:19:23.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/2"}}},"commit":{"parents":[{"commit":"4a0f64b5cbd48af7c44788e0f86e313a6e7865a5","subject":"finalize python 3.9 support removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/4a0f64b5cbd48af7c44788e0f86e313a6e7865a5"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-08-28 12:18:06.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/9a2b32b732466a8360f382eb33cef534f2f87595"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/9a2b32b732466a8360f382eb33cef534f2f87595"}]},"branch":"refs/heads/master"},"df7a850bc63e0ee72ecf389491d7f310f0ac2b06":{"kind":"REWORK","_number":3,"created":"2025-08-28 12:40:41.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/3"}}},"commit":{"parents":[{"commit":"4a0f64b5cbd48af7c44788e0f86e313a6e7865a5","subject":"finalize python 3.9 support removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/4a0f64b5cbd48af7c44788e0f86e313a6e7865a5"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-08-28 12:40:24.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/df7a850bc63e0ee72ecf389491d7f310f0ac2b06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/df7a850bc63e0ee72ecf389491d7f310f0ac2b06"}]},"branch":"refs/heads/master"},"d07bc4f293dcaf80005ffa82985f65cc64864f0a":{"kind":"REWORK","_number":4,"created":"2025-08-29 04:08:49.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/4"}}},"commit":{"parents":[{"commit":"4a0f64b5cbd48af7c44788e0f86e313a6e7865a5","subject":"finalize python 3.9 support removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/4a0f64b5cbd48af7c44788e0f86e313a6e7865a5"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-08-29 04:08:40.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/d07bc4f293dcaf80005ffa82985f65cc64864f0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/d07bc4f293dcaf80005ffa82985f65cc64864f0a"}]},"branch":"refs/heads/master"},"159d3fc93c8e80b68178b769c4f67b341fbdc6f4":{"kind":"REWORK","_number":5,"created":"2025-08-29 04:10:49.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/5"}}},"commit":{"parents":[{"commit":"4a0f64b5cbd48af7c44788e0f86e313a6e7865a5","subject":"finalize python 3.9 support removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/4a0f64b5cbd48af7c44788e0f86e313a6e7865a5"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-08-29 04:10:38.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/159d3fc93c8e80b68178b769c4f67b341fbdc6f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/159d3fc93c8e80b68178b769c4f67b341fbdc6f4"}]},"branch":"refs/heads/master"},"9d98d5f00088fffd136ec736801a7c758e7c5dbf":{"kind":"REWORK","_number":6,"created":"2025-08-29 04:17:35.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/6"}}},"commit":{"parents":[{"commit":"4a0f64b5cbd48af7c44788e0f86e313a6e7865a5","subject":"finalize python 3.9 support removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/4a0f64b5cbd48af7c44788e0f86e313a6e7865a5"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-08-29 04:17:28.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/9d98d5f00088fffd136ec736801a7c758e7c5dbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/9d98d5f00088fffd136ec736801a7c758e7c5dbf"}]},"branch":"refs/heads/master"},"9d394b1e1cda85f3f29a8f8e742411d3a75e1f9f":{"kind":"REWORK","_number":7,"created":"2025-08-29 04:28:50.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/7"}}},"commit":{"parents":[{"commit":"4a0f64b5cbd48af7c44788e0f86e313a6e7865a5","subject":"finalize python 3.9 support removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/4a0f64b5cbd48af7c44788e0f86e313a6e7865a5"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-08-29 04:28:43.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/9d394b1e1cda85f3f29a8f8e742411d3a75e1f9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/9d394b1e1cda85f3f29a8f8e742411d3a75e1f9f"}]},"branch":"refs/heads/master"},"b5a3ba47f3963edcbeeb7c4b8556d53f5995477d":{"kind":"REWORK","_number":8,"created":"2025-08-29 11:53:02.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/8"}}},"commit":{"parents":[{"commit":"4a0f64b5cbd48af7c44788e0f86e313a6e7865a5","subject":"finalize python 3.9 support removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/4a0f64b5cbd48af7c44788e0f86e313a6e7865a5"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-08-29 11:52:06.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/b5a3ba47f3963edcbeeb7c4b8556d53f5995477d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/b5a3ba47f3963edcbeeb7c4b8556d53f5995477d"}]},"branch":"refs/heads/master"},"acfb5a24ee8f2e9c9f5c2f84e91091ffad09877c":{"kind":"REWORK","_number":9,"created":"2025-08-29 14:07:21.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/9"}}},"commit":{"parents":[{"commit":"c2ad258aacef244aa4afa0baf255e00cffdd041d","subject":"Add Start and END time fields for continuous audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/c2ad258aacef244aa4afa0baf255e00cffdd041d"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-08-29 14:06:27.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/acfb5a24ee8f2e9c9f5c2f84e91091ffad09877c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/acfb5a24ee8f2e9c9f5c2f84e91091ffad09877c"}]},"branch":"refs/heads/master"},"01c5a50ff73a157463e5f00542e0e39750afce22":{"kind":"REWORK","_number":10,"created":"2025-09-01 04:46:55.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/10"}}},"commit":{"parents":[{"commit":"c2ad258aacef244aa4afa0baf255e00cffdd041d","subject":"Add Start and END time fields for continuous audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/c2ad258aacef244aa4afa0baf255e00cffdd041d"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-09-01 04:46:01.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/01c5a50ff73a157463e5f00542e0e39750afce22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/01c5a50ff73a157463e5f00542e0e39750afce22"}]},"branch":"refs/heads/master"},"c91b2de6143d8157827ed2a76c86c9323d15f777":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-09-01 04:48:15.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/11"}}},"commit":{"parents":[{"commit":"f67425d43dfa9f664530f94f0e9fb41114b04ab4","subject":"Add Start and END time fields for continuous audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/f67425d43dfa9f664530f94f0e9fb41114b04ab4"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2025-09-01 04:48:15.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/c91b2de6143d8157827ed2a76c86c9323d15f777"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/c91b2de6143d8157827ed2a76c86c9323d15f777"}]},"branch":"refs/heads/master","description":"Rebase"},"b920d065f2e37c94cdc4efafa083bc72d01f3f7d":{"kind":"REWORK","_number":12,"created":"2025-09-01 14:43:26.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/12"}}},"commit":{"parents":[{"commit":"f67425d43dfa9f664530f94f0e9fb41114b04ab4","subject":"Add Start and END time fields for continuous audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/f67425d43dfa9f664530f94f0e9fb41114b04ab4"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-09-01 14:42:22.000000000","tz":330},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/959051\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\nCo-Authored-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/b920d065f2e37c94cdc4efafa083bc72d01f3f7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/b920d065f2e37c94cdc4efafa083bc72d01f3f7d"}]},"branch":"refs/heads/master"},"1d75e29fc37baac5e6fea1ce4e896991c8a6dd77":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-09-01 14:43:47.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/13"}}},"commit":{"parents":[{"commit":"d4b9bf56e143a85c489ebbc35b72cd128de19683","subject":"Add Start and END time fields for continuous audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/d4b9bf56e143a85c489ebbc35b72cd128de19683"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2025-09-01 14:43:47.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/959051\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\nCo-Authored-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/1d75e29fc37baac5e6fea1ce4e896991c8a6dd77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/1d75e29fc37baac5e6fea1ce4e896991c8a6dd77"}]},"branch":"refs/heads/master","description":"Rebase"},"1df49b72a408fdf9d182422b28f869f62fc0aad0":{"kind":"REWORK","_number":14,"created":"2025-09-02 13:12:50.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/14"}}},"commit":{"parents":[{"commit":"c2ad258aacef244aa4afa0baf255e00cffdd041d","subject":"Add Start and END time fields for continuous audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/c2ad258aacef244aa4afa0baf255e00cffdd041d"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-09-02 13:12:10.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/1df49b72a408fdf9d182422b28f869f62fc0aad0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/1df49b72a408fdf9d182422b28f869f62fc0aad0"}]},"branch":"refs/heads/master"},"dc67371c2c2ca4e5c5883dc86a5ee53b0db3acfe":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2025-09-02 13:14:21.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/15"}}},"commit":{"parents":[{"commit":"d4b9bf56e143a85c489ebbc35b72cd128de19683","subject":"Add Start and END time fields for continuous audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/d4b9bf56e143a85c489ebbc35b72cd128de19683"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2025-09-02 13:14:21.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/dc67371c2c2ca4e5c5883dc86a5ee53b0db3acfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/dc67371c2c2ca4e5c5883dc86a5ee53b0db3acfe"}]},"branch":"refs/heads/master","description":"Rebase"},"522650212628e1d28b448fdc39ceacfefc3aa15b":{"kind":"REWORK","_number":16,"created":"2025-09-02 13:41:14.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/16"}}},"commit":{"parents":[{"commit":"c2ad258aacef244aa4afa0baf255e00cffdd041d","subject":"Add Start and END time fields for continuous audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/c2ad258aacef244aa4afa0baf255e00cffdd041d"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-09-02 13:40:59.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/522650212628e1d28b448fdc39ceacfefc3aa15b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/522650212628e1d28b448fdc39ceacfefc3aa15b"}]},"branch":"refs/heads/master"},"a2a8ae865bc6f110af4ac24708877ea66f1d52ba":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2025-09-02 13:41:36.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/17"}}},"commit":{"parents":[{"commit":"d4b9bf56e143a85c489ebbc35b72cd128de19683","subject":"Add Start and END time fields for continuous audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/d4b9bf56e143a85c489ebbc35b72cd128de19683"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2025-09-02 13:41:36.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/a2a8ae865bc6f110af4ac24708877ea66f1d52ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/a2a8ae865bc6f110af4ac24708877ea66f1d52ba"}]},"branch":"refs/heads/master","description":"Rebase"},"93c74ec2270a8da6a48eb90a737e836be3efcc49":{"kind":"REWORK","_number":18,"created":"2025-09-04 07:24:31.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/18"}}},"commit":{"parents":[{"commit":"d4b9bf56e143a85c489ebbc35b72cd128de19683","subject":"Add Start and END time fields for continuous audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/d4b9bf56e143a85c489ebbc35b72cd128de19683"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-09-04 07:21:06.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then when user uses status_message field is used. User will be\nwarned about this feature requires 1.5 watcher api version.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/93c74ec2270a8da6a48eb90a737e836be3efcc49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/93c74ec2270a8da6a48eb90a737e836be3efcc49"}]},"branch":"refs/heads/master"},"fb6e772524789adfdd5060323aaa33018d3af127":{"kind":"NO_CODE_CHANGE","_number":19,"created":"2025-09-04 07:28:39.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/19"}}},"commit":{"parents":[{"commit":"d4b9bf56e143a85c489ebbc35b72cd128de19683","subject":"Add Start and END time fields for continuous audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/d4b9bf56e143a85c489ebbc35b72cd128de19683"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-09-04 07:24:52.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then when user uses status_message field is used. User will\nreceive error about this feature requires 1.5 watcher api version.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/fb6e772524789adfdd5060323aaa33018d3af127"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/fb6e772524789adfdd5060323aaa33018d3af127"}]},"branch":"refs/heads/master"},"e807aeb1cf9fa6d145c56a5bacd72fa276645a4c":{"kind":"REWORK","_number":20,"created":"2025-09-04 07:51:55.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/20"}}},"commit":{"parents":[{"commit":"d4b9bf56e143a85c489ebbc35b72cd128de19683","subject":"Add Start and END time fields for continuous audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/d4b9bf56e143a85c489ebbc35b72cd128de19683"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-09-04 07:51:25.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then when user uses status_message field is used. User will\nreceive error about this feature requires 1.5 watcher api version.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/e807aeb1cf9fa6d145c56a5bacd72fa276645a4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/e807aeb1cf9fa6d145c56a5bacd72fa276645a4c"}]},"branch":"refs/heads/master"},"f41949cbcccbdc41de3201d8e49973d9fee818c7":{"kind":"REWORK","_number":21,"created":"2025-09-08 08:05:18.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/21"}}},"commit":{"parents":[{"commit":"d4b9bf56e143a85c489ebbc35b72cd128de19683","subject":"Add Start and END time fields for continuous audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/d4b9bf56e143a85c489ebbc35b72cd128de19683"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-08-21 14:45:42.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-09-08 08:04:56.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then skip action will not visible.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/f41949cbcccbdc41de3201d8e49973d9fee818c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/f41949cbcccbdc41de3201d8e49973d9fee818c7"}]},"branch":"refs/heads/master"},"1944c105f175703d014bf7e0ff7108ac4795e46d":{"kind":"REWORK","_number":22,"created":"2025-09-08 15:22:00.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/22"}}},"commit":{"parents":[{"commit":"41d390fb3e0b9f79804611e6040394285d586f84","subject":"Add Start and END time fields for continuous audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/41d390fb3e0b9f79804611e6040394285d586f84"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-09-08 15:21:40.000000000","tz":330},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then skip action will not visible.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/1944c105f175703d014bf7e0ff7108ac4795e46d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/1944c105f175703d014bf7e0ff7108ac4795e46d"}]},"branch":"refs/heads/master"},"a0547e10416cd0a04f615ed8f02a75925eb4ed5f":{"kind":"REWORK","_number":23,"created":"2025-09-09 05:47:47.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/23"}}},"commit":{"parents":[{"commit":"41d390fb3e0b9f79804611e6040394285d586f84","subject":"Add Start and END time fields for continuous audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/41d390fb3e0b9f79804611e6040394285d586f84"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-09-09 05:47:21.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then skip action will not visible.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/a0547e10416cd0a04f615ed8f02a75925eb4ed5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/a0547e10416cd0a04f615ed8f02a75925eb4ed5f"}]},"branch":"refs/heads/master"},"83b96948e4fd90e9afec48a16f3e46347b80f039":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2025-09-16 04:28:11.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/24"}}},"commit":{"parents":[{"commit":"9793d5ebab85cf986eca21dfab22c15dc9d44259","subject":"List strategies based on selected goal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/9793d5ebab85cf986eca21dfab22c15dc9d44259"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2025-09-16 04:28:11.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then skip action will not visible.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/83b96948e4fd90e9afec48a16f3e46347b80f039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/83b96948e4fd90e9afec48a16f3e46347b80f039"}]},"branch":"refs/heads/master","description":"Rebase"},"e7fbc5f512273f63c5094194f5ac0111ef8d8038":{"kind":"REWORK","_number":25,"created":"2025-09-17 12:01:51.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/25"}}},"commit":{"parents":[{"commit":"27f8d9d47035fdc391094d1769d55779c602ecd4","subject":"List strategies based on selected goal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/27f8d9d47035fdc391094d1769d55779c602ecd4"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-09-17 11:59:26.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then skip action will not visible.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/e7fbc5f512273f63c5094194f5ac0111ef8d8038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/e7fbc5f512273f63c5094194f5ac0111ef8d8038"}]},"branch":"refs/heads/master"},"ee871eecff5a42ad43679eac03548bae8db97795":{"kind":"REWORK","_number":26,"created":"2025-09-19 09:36:13.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/26"}}},"commit":{"parents":[{"commit":"2fe7211c5aef8b1b055020b3e06b9f3f604cdfdd","subject":"Add Start and END time fields for continuous audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/2fe7211c5aef8b1b055020b3e06b9f3f604cdfdd"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2025-09-19 09:36:13.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then skip action will not visible.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/ee871eecff5a42ad43679eac03548bae8db97795"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/ee871eecff5a42ad43679eac03548bae8db97795"}]},"branch":"refs/heads/master","description":"Rebase"},"a29b441cf4e439f775d65f187790b14a9b1ad0fb":{"kind":"REWORK","_number":27,"created":"2025-09-19 14:41:04.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/27"}}},"commit":{"parents":[{"commit":"2fe7211c5aef8b1b055020b3e06b9f3f604cdfdd","subject":"Add Start and END time fields for continuous audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/2fe7211c5aef8b1b055020b3e06b9f3f604cdfdd"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-09-19 14:39:27.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field. It will also show status_message about skipped\nactions in action plan details once it succeeded.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then skip action will not visible.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/a29b441cf4e439f775d65f187790b14a9b1ad0fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/a29b441cf4e439f775d65f187790b14a9b1ad0fb"}]},"branch":"refs/heads/master"},"5018a4bb21bf76f6742b33511591cdccba8ca101":{"kind":"REWORK","_number":28,"created":"2025-09-25 05:23:39.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/28"}}},"commit":{"parents":[{"commit":"2e6d5cac1bb4b7261873bc8415f88d7cf7caf4ff","subject":"Merge \"List strategies based on selected goal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/2e6d5cac1bb4b7261873bc8415f88d7cf7caf4ff"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-09-25 05:19:58.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field. It will also show status_message about skipped\nactions in action plan details once it succeeded.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then skip action will not visible.\n\nIt also rename wv to common_client and also fix detail\npage for strategy.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/5018a4bb21bf76f6742b33511591cdccba8ca101"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/5018a4bb21bf76f6742b33511591cdccba8ca101"}]},"branch":"refs/heads/master"},"00b0ad0ece0e7c73cb31747920bf45073d105870":{"kind":"REWORK","_number":29,"created":"2025-10-03 03:57:29.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/29"}}},"commit":{"parents":[{"commit":"2e6d5cac1bb4b7261873bc8415f88d7cf7caf4ff","subject":"Merge \"List strategies based on selected goal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/2e6d5cac1bb4b7261873bc8415f88d7cf7caf4ff"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-10-03 03:56:09.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field. It will also show status_message about skipped\nactions in action plan details once it succeeded.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then skip action will not visible.\n\nIt also rename wv to common_client.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/00b0ad0ece0e7c73cb31747920bf45073d105870"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/00b0ad0ece0e7c73cb31747920bf45073d105870"}]},"branch":"refs/heads/master"},"cde10961fe1aeb4a6e2161673f396e1200d7c720":{"kind":"REWORK","_number":30,"created":"2025-10-06 07:03:50.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/30"}}},"commit":{"parents":[{"commit":"2e6d5cac1bb4b7261873bc8415f88d7cf7caf4ff","subject":"Merge \"List strategies based on selected goal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/2e6d5cac1bb4b7261873bc8415f88d7cf7caf4ff"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-10-06 07:02:50.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field. It will also show status_message about skipped\nactions in action plan details once it succeeded.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then skip action will not visible.\n\nIt also rename wv to common_client.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/cde10961fe1aeb4a6e2161673f396e1200d7c720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/cde10961fe1aeb4a6e2161673f396e1200d7c720"}]},"branch":"refs/heads/master"},"87dd24b7864f767f66b19e6a432b1280834e2446":{"kind":"REWORK","_number":31,"created":"2025-10-10 15:03:33.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/31"}}},"commit":{"parents":[{"commit":"2e6d5cac1bb4b7261873bc8415f88d7cf7caf4ff","subject":"Merge \"List strategies based on selected goal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/2e6d5cac1bb4b7261873bc8415f88d7cf7caf4ff"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-10-10 15:02:33.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field. It will also show status_message about skipped\nactions in action plan details once it succeeded.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then skip action will not visible.\n\nIt also rename wv to common_client.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/87dd24b7864f767f66b19e6a432b1280834e2446"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/87dd24b7864f767f66b19e6a432b1280834e2446"}]},"branch":"refs/heads/master"},"56bbf21c574d9897d2ee477eeda685a58d62da15":{"kind":"REWORK","_number":32,"created":"2025-10-24 12:06:17.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/32"}}},"commit":{"parents":[{"commit":"2e6d5cac1bb4b7261873bc8415f88d7cf7caf4ff","subject":"Merge \"List strategies based on selected goal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/2e6d5cac1bb4b7261873bc8415f88d7cf7caf4ff"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-10-24 12:06:05.000000000","tz":330},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field. It will also show status_message about skipped\nactions in action plan details once it succeeded.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then skip action will not visible.\n\nIt also rename wv to common_client.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/56bbf21c574d9897d2ee477eeda685a58d62da15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/56bbf21c574d9897d2ee477eeda685a58d62da15"}]},"branch":"refs/heads/master"},"1f3a375cb983e4db9a2219247628aa16e62e05fc":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2025-10-24 13:15:27.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/33"}}},"commit":{"parents":[{"commit":"fdbd6ac68876a065e10d99761de9c2290bba7f75","subject":"Remove legacy integration test framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/fdbd6ac68876a065e10d99761de9c2290bba7f75"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2025-10-24 13:15:27.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field. It will also show status_message about skipped\nactions in action plan details once it succeeded.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then skip action will not visible.\n\nIt also rename wv to common_client.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/1f3a375cb983e4db9a2219247628aa16e62e05fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/1f3a375cb983e4db9a2219247628aa16e62e05fc"}]},"branch":"refs/heads/master","description":"Rebase"},"7275dccf0c5d99032a0ee59e70e2e0901ffc1670":{"kind":"REWORK","_number":34,"created":"2025-11-06 05:15:19.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/34"}}},"commit":{"parents":[{"commit":"ec7843cf0eb3001339a3d943c9af7cd589a37e52","subject":"Remove legacy integration test framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/ec7843cf0eb3001339a3d943c9af7cd589a37e52"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-11-06 05:13:21.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field. It will also show status_message about skipped\nactions in action plan details once it succeeded.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then skip action will not visible.\n\nIt also rename wv to common_client.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/7275dccf0c5d99032a0ee59e70e2e0901ffc1670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/7275dccf0c5d99032a0ee59e70e2e0901ffc1670"}]},"branch":"refs/heads/master"},"a7b54f1f23c48fce42f76846baae6bbd6913fcc9":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2025-11-06 05:16:03.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/35"}}},"commit":{"parents":[{"commit":"de25ea2fd2dd709e43f26186d9c4a9d3f71022f2","subject":"Merge \"Improve audit creation model and add text wrapping\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/de25ea2fd2dd709e43f26186d9c4a9d3f71022f2"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2025-11-06 05:16:03.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field. It will also show status_message about skipped\nactions in action plan details once it succeeded.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then skip action will not visible.\n\nIt also rename wv to common_client.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/a7b54f1f23c48fce42f76846baae6bbd6913fcc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/a7b54f1f23c48fce42f76846baae6bbd6913fcc9"}]},"branch":"refs/heads/master","description":"Rebase"},"18275cdb90c24a7ec060daf2c96caa3e531b40d1":{"kind":"REWORK","_number":36,"created":"2025-11-06 05:30:19.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/36"}}},"commit":{"parents":[{"commit":"de25ea2fd2dd709e43f26186d9c4a9d3f71022f2","subject":"Merge \"Improve audit creation model and add text wrapping\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/de25ea2fd2dd709e43f26186d9c4a9d3f71022f2"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-11-06 05:28:45.000000000","tz":0},"subject":"Add option to SKIP Actions","message":"Add option to SKIP Actions\n\nSupport for skipping actions via patch call has been added to Watcher\nrecently in api microversion 1.5 [1].\n\nThis patch is adding support for this feature in watcher-dashboard. It\nincludes adding an optional reason that will be added to the\nstatus_message field. It will also show status_message about skipped\nactions in action plan details once it succeeded.\n\nNote: It also make sure if older version of watcher is used in deployed\ncloud, then skip action will not visible.\n\nIt also rename wv to common_client.\n\n[1] https://review.opendev.org/c/openstack/watcher/+/955753\n\nDepends-On: https://review.opendev.org/c/openstack/python-watcherclient/+/956911\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/18275cdb90c24a7ec060daf2c96caa3e531b40d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/18275cdb90c24a7ec060daf2c96caa3e531b40d1"}]},"branch":"refs/heads/master"},"35ea4c9800470864148288cc7860b439788069ec":{"kind":"REWORK","_number":37,"created":"2025-11-20 15:41:34.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/37"}}},"commit":{"parents":[{"commit":"aaebb77ea7da21c01d8995784c2ba62b73643874","subject":"Merge \"Adopt Ruff for security; remove Bandit and fix style\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/aaebb77ea7da21c01d8995784c2ba62b73643874"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2025-11-20 15:41:34.000000000","tz":0},"subject":"Add skip action feature with microversion support","message":"Add skip action feature with microversion support\n\nThis change introduces the ability to skip actions in action plans\nthrough the watcher-dashboard UI, leveraging the API microversion 1.5\nskip action functionality introduced in python-watcherclient 4.9.0.\n\nKey features:\n- Add SkipAction table action with modal form allowing users to skip\n  actions with an optional reason (up to 200 characters)\n- Implement microversion discovery and negotiation to detect server\n  capabilities and conditionally show/hide skip controls\n- Display status_message field in action and action plan detail views\n  when supported by the backend\n- Restrict skip operation to actions in PENDING state within action\n  plans in RECOMMENDED or PENDING states\n- Support updating skip reasons for actions already in SKIPPED state\n- Gracefully degrade UI when older Watcher API versions are detected\n  (hides Actions column and skip controls for API \u003c 1.5)\n\nTechnical improvements:\n- Refactor client helper module renaming \u0027wv\u0027 to \u0027common_client\u0027 for\n  clarity and consistency\n- Add get_server_microversion_range() with per-request caching to\n  minimize repeated version discovery calls\n- Add is_supported() helper for microversion capability checking\n- Update python-watcherclient requirement to \u003e\u003d4.9.0\n- Fix PEP8 compliance issues throughout modified files\n\nTesting:\n- Add unit tests for Action.update(), Action.start(), microversion\n  helpers, and skip action permission gating\n- Verify backward compatibility with API \u003c 1.5\n\nAssisted-By: Cursor (claude-4.5-sonnet)\n\nRelated: https://review.opendev.org/c/openstack/watcher/+/955753\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/35ea4c9800470864148288cc7860b439788069ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/35ea4c9800470864148288cc7860b439788069ec"}]},"branch":"refs/heads/master"},"eb03859257e1005733fc2bee0db8432bf749e078":{"kind":"REWORK","_number":38,"created":"2026-02-20 14:49:23.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/38"}}},"commit":{"parents":[{"commit":"aaebb77ea7da21c01d8995784c2ba62b73643874","subject":"Merge \"Adopt Ruff for security; remove Bandit and fix style\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/aaebb77ea7da21c01d8995784c2ba62b73643874"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-02-20 14:49:22.000000000","tz":0},"subject":"Add skip action feature with microversion support","message":"Add skip action feature with microversion support\n\nThis change introduces the ability to skip actions in action plans\nthrough the watcher-dashboard UI, leveraging the API microversion 1.5\nskip action functionality introduced in python-watcherclient 4.9.0.\n\nKey features:\n- Add SkipAction table action with modal form allowing users to skip\n  actions with an optional reason (up to 200 characters)\n- Implement microversion discovery and negotiation to detect server\n  capabilities and conditionally show/hide skip controls\n- Display status_message field in action and action plan detail views\n  when supported by the backend\n- Restrict skip operation to actions in PENDING state within action\n  plans in RECOMMENDED or PENDING states\n- Support updating skip reasons for actions already in SKIPPED state\n- Gracefully degrade UI when older Watcher API versions are detected\n  (hides Actions column and skip controls for API \u003c 1.5)\n\nTechnical improvements:\n- Refactor client helper module renaming \u0027wv\u0027 to \u0027common_client\u0027 for\n  clarity and consistency\n- Add get_server_microversion_range() with per-request caching to\n  minimize repeated version discovery calls\n- Add is_supported() helper for microversion capability checking\n- Update python-watcherclient requirement to \u003e\u003d4.9.0\n- Fix PEP8 compliance issues throughout modified files\n\nTesting:\n- Add unit tests for Action.update(), Action.start(), microversion\n  helpers, and skip action permission gating\n- Verify backward compatibility with API \u003c 1.5\n\nAssisted-By: Cursor (claude-4.5-sonnet)\n\nRelated: https://review.opendev.org/c/openstack/watcher/+/955753\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/eb03859257e1005733fc2bee0db8432bf749e078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/eb03859257e1005733fc2bee0db8432bf749e078"}]},"branch":"refs/heads/master"},"f12d27c2e35fa97c7eff39edcff0fa7a1722cea8":{"kind":"REWORK","_number":39,"created":"2026-02-20 15:39:05.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/39"}}},"commit":{"parents":[{"commit":"aaebb77ea7da21c01d8995784c2ba62b73643874","subject":"Merge \"Adopt Ruff for security; remove Bandit and fix style\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/aaebb77ea7da21c01d8995784c2ba62b73643874"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-02-20 15:39:02.000000000","tz":0},"subject":"Add skip action feature with microversion support","message":"Add skip action feature with microversion support\n\nThis change introduces the ability to skip actions in action plans\nthrough the watcher-dashboard UI, leveraging the API microversion 1.5\nskip action functionality introduced in python-watcherclient 4.9.0.\n\nKey features:\n- Add SkipAction table action with modal form allowing users to skip\n  actions with an optional reason (up to 200 characters)\n- Implement microversion discovery and negotiation to detect server\n  capabilities and conditionally show/hide skip controls\n- Display status_message field in action and action plan detail views\n  when supported by the backend\n- Restrict skip operation to actions in PENDING state within action\n  plans in RECOMMENDED or PENDING states\n- Support updating skip reasons for actions already in SKIPPED state\n- Gracefully degrade UI when older Watcher API versions are detected\n  (hides Actions column and skip controls for API \u003c 1.5)\n\nTechnical improvements:\n- Refactor client helper module renaming \u0027wv\u0027 to \u0027common_client\u0027 for\n  clarity and consistency\n- Add get_server_microversion_range() with per-request caching to\n  minimize repeated version discovery calls\n- Add is_supported() helper for microversion capability checking\n- Update python-watcherclient requirement to \u003e\u003d4.9.0\n- Fix PEP8 compliance issues throughout modified files\n\nTesting:\n- Add unit tests for Action.update(), Action.start(), microversion\n  helpers, and skip action permission gating\n- Verify backward compatibility with API \u003c 1.5\n\nAssisted-By: Cursor (claude-4.5-sonnet)\n\nRelated: https://review.opendev.org/c/openstack/watcher/+/955753\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/f12d27c2e35fa97c7eff39edcff0fa7a1722cea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/f12d27c2e35fa97c7eff39edcff0fa7a1722cea8"}]},"branch":"refs/heads/master"},"167b2ada6ee7338bbc35d4cc1341007f29a36094":{"kind":"NO_CODE_CHANGE","_number":40,"created":"2026-02-20 15:43:48.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/40"}}},"commit":{"parents":[{"commit":"aaebb77ea7da21c01d8995784c2ba62b73643874","subject":"Merge \"Adopt Ruff for security; remove Bandit and fix style\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/aaebb77ea7da21c01d8995784c2ba62b73643874"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-02-20 15:43:09.000000000","tz":0},"subject":"Add skip action feature with microversion support","message":"Add skip action feature with microversion support\n\nThis change introduces the ability to skip actions in action plans\nthrough the watcher-dashboard UI, leveraging the API microversion 1.5\nskip action functionality introduced in python-watcherclient 4.9.0.\n\nKey features:\n- Add SkipAction table action with modal form allowing users to skip\n  actions with an optional reason (up to 200 characters)\n- Implement microversion discovery and negotiation to detect server\n  capabilities and conditionally show/hide skip controls\n- Display status_message field in action and action plan detail views\n  when supported by the backend\n- Restrict skip operation to actions in PENDING state within action\n  plans in RECOMMENDED or PENDING states\n- Support updating skip reasons for actions already in SKIPPED state\n- Gracefully degrade UI when older Watcher API versions are detected\n  (hides Actions column and skip controls for API \u003c 1.5)\n\nTechnical improvements:\n- Refactor client helper module renaming \u0027wv\u0027 to \u0027common_client\u0027 for\n  clarity and consistency\n- Add get_server_microversion_range() with per-request caching to\n  minimize repeated version discovery calls\n- Add is_supported() helper for microversion capability checking\n- Update python-watcherclient requirement to \u003e\u003d4.9.0\n- Fix PEP8 compliance issues throughout modified files\n\nTesting:\n- Add unit tests for Action.update(), microversion\n  helpers, and skip action permission gating\n- Verify backward compatibility with API \u003c 1.5\n\nAssisted-By: Cursor (claude-4.5-sonnet)\n\nRelated: https://review.opendev.org/c/openstack/watcher/+/955753\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/167b2ada6ee7338bbc35d4cc1341007f29a36094"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/167b2ada6ee7338bbc35d4cc1341007f29a36094"}]},"branch":"refs/heads/master"},"4f6c274d0ddb00cabe7d63c83971c7ff8bdfdc5f":{"kind":"REWORK","_number":41,"created":"2026-02-26 13:53:52.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/41"}}},"commit":{"parents":[{"commit":"5294b524e092f8760914d3b21ed8ea2393392745","subject":"Merge \"ruff: Enable E5 check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/5294b524e092f8760914d3b21ed8ea2393392745"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-02-26 13:53:48.000000000","tz":0},"subject":"Add skip action feature with microversion support","message":"Add skip action feature with microversion support\n\nThis change introduces the ability to skip actions in action plans\nthrough the watcher-dashboard UI, leveraging the API microversion 1.5\nskip action functionality introduced in python-watcherclient 4.9.0.\n\nKey features:\n- Add SkipAction table action with modal form allowing users to skip\n  actions with an optional reason (up to 200 characters)\n- Implement microversion discovery and negotiation to detect server\n  capabilities and conditionally show/hide skip controls\n- Display status_message field in action and action plan detail views\n  when supported by the backend\n- Restrict skip operation to actions in PENDING state; allow reason\n  updates for actions already in SKIPPED state; disable skip controls\n  when the parent action plan has SUCCEEDED\n- Gracefully degrade UI when older Watcher API versions are detected\n  (hides Actions column and skip controls for API \u003c 1.5)\n\nTechnical improvements:\n- Refactor client helper module renaming \u0027wv\u0027 to \u0027common_client\u0027 for\n  clarity and consistency\n- Add get_server_microversion_range() with per-request caching using\n  openstack_auth sessions so thar SSL, token, and proxy handling are\n  consistent with Horizon\n- Add is_supported() helper using watcherclient\u0027s api_versioning\n  APIVersion for microversion capability checking\n- Add WatcherDashboardException typed exception in common/exceptions.py,\n  replacing generic ValueError in Action.update()\n- Catch wc_exc.MethodNotAllowed specifically in Action.update() instead\n  of broad except Exception for graceful degradation on HTTP 405\n- Replace request-object mutation pattern in action_plans DetailView\n  with a detail_action_plan property and explicit kwargs to\n  RelatedActionsTable via get_tables()\n- Add get_strategy_display_name() shared helper in api/watcher.py\n  using strategy.to_dict() to access display_name (not exposed in\n  watcherclient Strategy resource fields)\n- Add get_strategies_for_goal() AJAX endpoint in audit_templates for\n  dynamic strategy filtering by goal\n- Refactor audit template CreateForm: remove unnecessary try/except\n  AttributeError around self.data (never None on Django forms),\n  dynamically populate strategy choices on POST\n- Update python-watcherclient requirement to \u003e\u003d4.9.0\n\nTesting:\n- Add unit tests for Action.update(), microversion\n  helpers, skip action permission gating, and\n  ActionPlan status_message attribute\n- Verify backward compatibility with API \u003c 1.5\n\nAssisted-By: Cursor (claude-4.5-sonnet)\nAssisted-By: Cursor (claude-4.6-opus-high)\n\nRelated: https://review.opendev.org/c/openstack/watcher/+/955753\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\nCo-Authored-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/4f6c274d0ddb00cabe7d63c83971c7ff8bdfdc5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/4f6c274d0ddb00cabe7d63c83971c7ff8bdfdc5f"}]},"branch":"refs/heads/master"},"e3375f09b61e87297e317d0e143643eb09e44385":{"kind":"REWORK","_number":42,"created":"2026-02-26 15:03:37.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/42"}}},"commit":{"parents":[{"commit":"3e1db2ba6d3ffd8b79fe0a4bc8f2650ff2f68cf9","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/3e1db2ba6d3ffd8b79fe0a4bc8f2650ff2f68cf9"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-02-26 15:03:31.000000000","tz":0},"subject":"Add skip action feature with microversion support","message":"Add skip action feature with microversion support\n\nThis change introduces the ability to skip actions in action plans\nthrough the watcher-dashboard UI, leveraging the API microversion 1.5\nskip action functionality introduced in python-watcherclient 4.9.0.\n\nKey features:\n- Add SkipAction table action with modal form allowing users to skip\n  actions with an optional reason (up to 200 characters)\n- Implement microversion discovery and negotiation to detect server\n  capabilities and conditionally show/hide skip controls\n- Display status_message field in action and action plan detail views\n  when supported by the backend\n- Restrict skip operation to actions in PENDING state; allow reason\n  updates for actions already in SKIPPED state; disable skip controls\n  when the parent action plan has SUCCEEDED\n- Gracefully degrade UI when older Watcher API versions are detected\n  (hides Actions column and skip controls for API \u003c 1.5)\n\nTechnical improvements:\n- Refactor client helper module renaming \u0027wv\u0027 to \u0027common_client\u0027 for\n  clarity and consistency\n- Add get_server_microversion_range() with per-request caching using\n  openstack_auth sessions so thar SSL, token, and proxy handling are\n  consistent with Horizon\n- Add is_supported() helper using watcherclient\u0027s api_versioning\n  APIVersion for microversion capability checking\n- Add WatcherDashboardException typed exception in common/exceptions.py,\n  replacing generic ValueError in Action.update()\n- Catch wc_exc.MethodNotAllowed specifically in Action.update() instead\n  of broad except Exception for graceful degradation on HTTP 405\n- Replace request-object mutation pattern in action_plans DetailView\n  with a detail_action_plan property and explicit kwargs to\n  RelatedActionsTable via get_tables()\n- Add get_strategy_display_name() shared helper in api/watcher.py\n  using strategy.to_dict() to access display_name (not exposed in\n  watcherclient Strategy resource fields)\n- Add get_strategies_for_goal() AJAX endpoint in audit_templates for\n  dynamic strategy filtering by goal\n- Refactor audit template CreateForm: remove unnecessary try/except\n  AttributeError around self.data (never None on Django forms),\n  dynamically populate strategy choices on POST\n- Update python-watcherclient requirement to \u003e\u003d4.9.0\n\nTesting:\n- Add unit tests for Action.update(), microversion\n  helpers, skip action permission gating, and\n  ActionPlan status_message attribute\n- Verify backward compatibility with API \u003c 1.5\n\nAssisted-By: Cursor (claude-4.5-sonnet)\nAssisted-By: Cursor (claude-4.6-opus-high)\n\nRelated: https://review.opendev.org/c/openstack/watcher/+/955753\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\nCo-Authored-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/e3375f09b61e87297e317d0e143643eb09e44385"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/e3375f09b61e87297e317d0e143643eb09e44385"}]},"branch":"refs/heads/master"},"81beba98cbc0ebbb0b8d0f0bc95c38b717339c1d":{"kind":"NO_CODE_CHANGE","_number":43,"created":"2026-02-26 15:05:04.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/43"}}},"commit":{"parents":[{"commit":"3e1db2ba6d3ffd8b79fe0a4bc8f2650ff2f68cf9","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/3e1db2ba6d3ffd8b79fe0a4bc8f2650ff2f68cf9"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-02-26 15:04:17.000000000","tz":0},"subject":"Add skip action feature with microversion support","message":"Add skip action feature with microversion support\n\nThis change introduces the ability to skip actions in action plans\nthrough the watcher-dashboard UI, leveraging the API microversion 1.5\nskip action functionality introduced in python-watcherclient 4.9.0.\n\nKey features:\n- Add SkipAction table action with modal form allowing users to skip\n  actions with an optional reason (up to 200 characters)\n- Implement microversion discovery and negotiation to detect server\n  capabilities and conditionally show/hide skip controls\n- Display status_message field in action and action plan detail views\n  when supported by the backend\n- Restrict skip operation to actions in PENDING state; allow reason\n  updates for actions already in SKIPPED state; disable skip controls\n  when the parent action plan has SUCCEEDED\n- Gracefully degrade UI when older Watcher API versions are detected\n  (hides Actions column and skip controls for API \u003c 1.5)\n\nTechnical improvements:\n- Refactor client helper module renaming \u0027wv\u0027 to \u0027common_client\u0027 for\n  clarity and consistency\n- Add get_server_microversion_range() to get microversion\n- Add is_supported() helper using watcherclient\u0027s api_versioning\n  APIVersion for microversion capability checking\n- Add WatcherDashboardException typed exception in common/exceptions.py,\n  replacing generic ValueError in Action.update()\n- Catch wc_exc.MethodNotAllowed specifically in Action.update() instead\n  of broad except Exception for graceful degradation on HTTP 405\n- Replace request-object mutation pattern in action_plans DetailView\n  with a detail_action_plan property and explicit kwargs to\n  RelatedActionsTable via get_tables()\n- Add get_strategy_display_name() shared helper in api/watcher.py\n  using strategy.to_dict() to access display_name (not exposed in\n  watcherclient Strategy resource fields)\n- Add get_strategies_for_goal() AJAX endpoint in audit_templates for\n  dynamic strategy filtering by goal\n- Refactor audit template CreateForm: remove unnecessary try/except\n  AttributeError around self.data (never None on Django forms),\n  dynamically populate strategy choices on POST\n- Update python-watcherclient requirement to \u003e\u003d4.9.0\n\nTesting:\n- Add unit tests for Action.update(), microversion\n  helpers, skip action permission gating, and\n  ActionPlan status_message attribute\n- Verify backward compatibility with API \u003c 1.5\n\nAssisted-By: Cursor (claude-4.5-sonnet)\nAssisted-By: Cursor (claude-4.6-opus-high)\n\nRelated: https://review.opendev.org/c/openstack/watcher/+/955753\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\nCo-Authored-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/81beba98cbc0ebbb0b8d0f0bc95c38b717339c1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/81beba98cbc0ebbb0b8d0f0bc95c38b717339c1d"}]},"branch":"refs/heads/master"},"721c36b0758c292498e0f6dc96fea9442d40c3f4":{"kind":"REWORK","_number":44,"created":"2026-02-26 15:18:43.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/44"}}},"commit":{"parents":[{"commit":"3e1db2ba6d3ffd8b79fe0a4bc8f2650ff2f68cf9","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/3e1db2ba6d3ffd8b79fe0a4bc8f2650ff2f68cf9"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2025-09-08 15:07:09.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-02-26 15:18:40.000000000","tz":0},"subject":"Add skip action feature with microversion support","message":"Add skip action feature with microversion support\n\nThis change introduces the ability to skip actions in action plans\nthrough the watcher-dashboard UI, leveraging the API microversion 1.5\nskip action functionality introduced in python-watcherclient 4.9.0.\n\nKey features:\n- Add SkipAction table action with modal form allowing users to skip\n  actions with an optional reason (up to 200 characters)\n- Implement microversion discovery and negotiation to detect server\n  capabilities and conditionally show/hide skip controls\n- Display status_message field in action and action plan detail views\n  when supported by the backend\n- Restrict skip operation to actions in PENDING state; allow reason\n  updates for actions already in SKIPPED state; disable skip controls\n  when the parent action plan has SUCCEEDED\n- Gracefully degrade UI when older Watcher API versions are detected\n  (hides Actions column and skip controls for API \u003c 1.5)\n\nTechnical improvements:\n- Refactor client helper module renaming \u0027wv\u0027 to \u0027common_client\u0027 for\n  clarity and consistency\n- Add get_server_microversion_range() to get microversion\n- Add is_supported() helper using watcherclient\u0027s api_versioning\n  APIVersion for microversion capability checking\n- Add WatcherDashboardException typed exception in common/exceptions.py,\n  replacing generic ValueError in Action.update()\n- Catch wc_exc.MethodNotAllowed specifically in Action.update() instead\n  of broad except Exception for graceful degradation on HTTP 405\n- Replace request-object mutation pattern in action_plans DetailView\n  with a detail_action_plan property and explicit kwargs to\n  RelatedActionsTable via get_tables()\n- Add get_strategy_display_name() shared helper in api/watcher.py\n  using strategy.to_dict() to access display_name (not exposed in\n  watcherclient Strategy resource fields)\n- Add get_strategies_for_goal() AJAX endpoint in audit_templates for\n  dynamic strategy filtering by goal\n- Refactor audit template CreateForm: remove unnecessary try/except\n  AttributeError around self.data (never None on Django forms),\n  dynamically populate strategy choices on POST\n- Update python-watcherclient requirement to \u003e\u003d4.9.0\n\nTesting:\n- Add unit tests for Action.update(), microversion\n  helpers, skip action permission gating, and\n  ActionPlan status_message attribute\n- Verify backward compatibility with API \u003c 1.5\n\nAssisted-By: Cursor (claude-4.5-sonnet)\nAssisted-By: Cursor (claude-4.6-opus-high)\n\nRelated: https://review.opendev.org/c/openstack/watcher/+/955753\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\nCo-Authored-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/721c36b0758c292498e0f6dc96fea9442d40c3f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/721c36b0758c292498e0f6dc96fea9442d40c3f4"}]},"branch":"refs/heads/master"},"37f27c514e7c513f32e0c1cc764c62d50d1e99e3":{"kind":"REWORK","_number":45,"created":"2026-03-03 07:33:38.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/45"}}},"commit":{"parents":[{"commit":"1fc78e07655f1f12bafd282f647e060415036b26","subject":"Add microversion 1.5 API support and status_message display","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/1fc78e07655f1f12bafd282f647e060415036b26"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2026-03-03 06:44:36.000000000","tz":0},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-03-03 07:33:03.000000000","tz":0},"subject":"Add skip action UI controls","message":"Add skip action UI controls\n\nBuild on the microversion 1.5 API support to let operators skip\nindividual actions from the action plan detail page. A \"Skip Action\"\nbutton appears on each PENDING or SKIPPED action row, opening a\nmodal where the user can optionally provide a reason. The button\nand Actions column are hidden when the API does not support\nmicroversion 1.5 or when the action plan has already succeeded.\n\n- Add SkipActionForm with state validation and API submission\n- Add SkipAction row button (btn-danger) on RelatedActionsTable\n- Add SkipActionView modal with redirect to parent action plan\n- Conditionally hide Actions column via has_actions property\n- Add release note for the skip action feature\n\nAssisted-By: Cursor (claude-4.6-opus-high)\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\nCo-Authored-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/37f27c514e7c513f32e0c1cc764c62d50d1e99e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/37f27c514e7c513f32e0c1cc764c62d50d1e99e3"}]},"branch":"refs/heads/master"},"260a9773dfe8d5b24e91758c63b104460d268afc":{"kind":"REWORK","_number":46,"created":"2026-03-03 09:58:10.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/46"}}},"commit":{"parents":[{"commit":"1fc78e07655f1f12bafd282f647e060415036b26","subject":"Add microversion 1.5 API support and status_message display","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/1fc78e07655f1f12bafd282f647e060415036b26"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2026-03-03 06:44:36.000000000","tz":0},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-03-03 09:57:49.000000000","tz":0},"subject":"Add skip action UI controls","message":"Add skip action UI controls\n\nWatcher API microversion 1.5 lets operators skip individual\nactions in an action plan.  This patch adds the dashboard\ncontrols for that feature.\n\nA \"Skip Action\" button appears on each PENDING or SKIPPED\naction row in the action plan detail page, opening a modal\nwhere the user can optionally provide a reason.  Submitting\nthe form calls Action.update() at microversion 1.5.  The\nbutton and Actions column are hidden when the server does\nnot support microversion 1.5 or the plan already succeeded.\n\nNew files:\n- actions/forms.py: SkipActionForm with state validation\n- actions/skip.html templates: modal dialog\n\nModified files:\n- actions/tables.py: SkipAction row button, conditional\n  Actions column via _supports_skip and _parent_succeeded\n- actions/views.py: SkipActionView modal with redirect to\n  parent action plan\n- actions/urls.py: skip URL pattern\n- action_plans/views.py: pass skip-support flags in\n  get_tables()\n- Release note for the skip action feature\n\nAssisted-By: Cursor (claude-4.6-opus-high)\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\nCo-Authored-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/260a9773dfe8d5b24e91758c63b104460d268afc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/260a9773dfe8d5b24e91758c63b104460d268afc"}]},"branch":"refs/heads/master"},"3a884013ecbca509d2379b090f2a16bb6206d2d5":{"kind":"TRIVIAL_REBASE","_number":47,"created":"2026-03-03 11:04:55.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/09/958209/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/09/958209/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/09/958209/47"}}},"commit":{"parents":[{"commit":"0561c24b445af8be5bdada78a566fe80ed48fc37","subject":"Add microversion 1.5 API support and status_message display","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/0561c24b445af8be5bdada78a566fe80ed48fc37"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2026-03-03 06:44:36.000000000","tz":0},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2026-03-03 11:04:55.000000000","tz":0},"subject":"Add skip action UI controls","message":"Add skip action UI controls\n\nWatcher API microversion 1.5 lets operators skip individual\nactions in an action plan.  This patch adds the dashboard\ncontrols for that feature.\n\nA \"Skip Action\" button appears on each PENDING or SKIPPED\naction row in the action plan detail page, opening a modal\nwhere the user can optionally provide a reason.  Submitting\nthe form calls Action.update() at microversion 1.5.  The\nbutton and Actions column are hidden when the server does\nnot support microversion 1.5 or the plan already succeeded.\n\nNew files:\n- actions/forms.py: SkipActionForm with state validation\n- actions/skip.html templates: modal dialog\n\nModified files:\n- actions/tables.py: SkipAction row button, conditional\n  Actions column via _supports_skip and _parent_succeeded\n- actions/views.py: SkipActionView modal with redirect to\n  parent action plan\n- actions/urls.py: skip URL pattern\n- action_plans/views.py: pass skip-support flags in\n  get_tables()\n- Release note for the skip action feature\n\nAssisted-By: Cursor (claude-4.6-opus-high)\n\nChange-Id: Ice3e01c214a8ce78553fe8ebc6ca9f43ffd3f273\nSigned-off-by: Alfredo Moralejo \u003camoralej@redhat.com\u003e\nCo-Authored-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/3a884013ecbca509d2379b090f2a16bb6206d2d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/3a884013ecbca509d2379b090f2a16bb6206d2d5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
