)]}'
{"id":"openstack%2Fwatcher-dashboard~964775","triplet_id":"openstack%2Fwatcher-dashboard~master~I748b55420edad1d160998f02d86cd5fc00b43435","project":"openstack/watcher-dashboard","branch":"master","topic":"integration-tests-cleanup","attention_set":{"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2025-10-24 13:19:31.000000000","reason":"\u003cGERRIT_ACCOUNT_12393\u003e replied on the change","reason_account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2025-11-03 14:25:04.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2025-10-24 16:05:53.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I748b55420edad1d160998f02d86cd5fc00b43435","subject":"Remove legacy integration test framework","status":"NEW","created":"2025-10-24 13:13:54.000000000","updated":"2025-11-06 07:58:08.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"386fa3263b7b6242c27160f24bb96b9f0cf912e0","_number":964775,"virtual_id_number":964775,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-11-06 05:36:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"disliked":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":1,"date":"2025-11-06 07:58:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":-1,"date":"2025-11-06 05:15:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}],"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-24 13:19:31.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2025-10-24 13:19:31.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2025-10-24 13:24:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-24 16:05:53.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-11-03 14:25:04.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"}],"messages":[{"id":"e8699a64639fec1be16d6026ed76e6b6c6004e41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-10-24 13:13:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d32ba2d47aa3980cfd06d57c476cef69ea3da90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 13:24:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/168fe8f141024f7ca7205fd393878230\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/154fd8fa2ce1453abcdef6e73f8c00f2 : SUCCESS in 2m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a30e8f54a0e440d91f874bb92580f1c : SUCCESS in 3m 45s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/422645992e10482e9d02bae3d9148eb4 : SUCCESS in 2m 22s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/41884899a3984d99bbaea1df77080e19 : FAILURE in 4m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89e6e391734147a1b03a88e9e3937885 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d345d327e2b4bcda46fb793db564959 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8527117d3e745b8a0150a01f4ae9a26 : SUCCESS in 9m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02bee9e9175e4ce19b20087f503afaa8 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/803f75cfe8944d958b76b2bb56a274a3 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bff1b86f962459ea57670f2b1c4f514 : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":1},{"id":"987ccc94eb814b70c681848f568b4880830f4f8f","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-10-24 16:05:53.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/e6522072e3d548649daebabce610a6aa\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5f64283e1ab04f5aaef58ae833b60476 : SUCCESS in 2h 50m 39s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/fc80c209a4ce4aba9e6a24638b602844 : SUCCESS in 1h 55m 42s","accounts_in_message":[],"_revision_number":1},{"id":"7dd7c5df1b50bcdc3e09cb8cc7b2a2e85acbb17b","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2025-10-27 15:29:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"43e328bed55ee05106f4e83e668f714f1f4a9236","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2025-10-28 14:35:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"966f3bad0151ec3bbd6c2e65ca16b822fb434b12","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-03 14:25:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49dc99ff0238d277353f7ed21bfd94106c4e2735","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-11-06 05:15:19.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"851bcd86a311f7c22577a89f4d1134232aba07b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 05:36:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac150c62a0164987aadad7e44a29ae70\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3111fb95e4644bf2a97a9996fca8b812 : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d8c88434d2642e48cf1155c8224330a : SUCCESS in 3m 29s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/c5c8cb9231fe471cafeb84a4b7efebdb : SUCCESS in 4m 16s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/ad59a06c383c487085e59f1a125f068a : FAILURE in 4m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49015c76cc2641519153ea7ec0de03a3 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e55a1067d4754be0b621521808b925bc : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c18675cf36d426c8d99713629d1e969 : SUCCESS in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8afc1167a1f14c65bdb26f649a101a85 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4fdd99128604e41af4ecce9f6079ef3 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1188f79b8d9d42e785b3589f2d58b32a : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":2},{"id":"386fa3263b7b6242c27160f24bb96b9f0cf912e0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-11-06 07:58:08.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/87600dc4b32548dca81360c921bc5347\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/556e6cf5ffb048169a209799f3c2275a : SUCCESS in 2h 41m 34s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/cfbdc471ecfd42809c908b17c2bfb2a6 : SUCCESS in 1h 52m 23s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ec7843cf0eb3001339a3d943c9af7cd589a37e52","revisions":{"fdbd6ac68876a065e10d99761de9c2290bba7f75":{"kind":"REWORK","_number":1,"created":"2025-10-24 13:13:54.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/75/964775/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/75/964775/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/75/964775/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/75/964775/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/75/964775/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/75/964775/1"}}},"commit":{"parents":[{"commit":"2e6d5cac1bb4b7261873bc8415f88d7cf7caf4ff","subject":"Merge \"List strategies based on selected goal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/2e6d5cac1bb4b7261873bc8415f88d7cf7caf4ff"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-10-24 13:12:00.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-10-24 13:12:00.000000000","tz":330},"subject":"Remove legacy integration test framework","message":"Remove legacy integration test framework\n\nFollowing the removal of integration test support in OpenStack Horizon\n(commit 49e5fe185a915ab80ccf4c130225371ade323711)i[1], this change completely\nremoves the legacy integration test framework from watcher-dashboard.\n\nChanges include:\n* Remove entire watcher_dashboard/test/integration_tests/ directory\n - Page object models for audit templates and audits\n - Selenium-based integration test cases\n - Integration test configuration files\n\n* Remove watcher_dashboard/test/selenium.py\n\n* Clean up run_tests.sh:\n - Remove selenium and integration test command-line options\n - Remove selenium configuration logic and environment variables\n - Remove run_integration_tests function\n - Simplify test exclusion logic\n\n* Update tox.ini to remove integration test exclusion flags\n\n* Remove selenium and xvfbwrapper dependencies from test-requirements.txt\n\n* Add release note documenting the removal\n\nLinks:\n[1]. https://review.opendev.org/c/openstack/horizon/+/960204\n\n🤖 Generated with [Claude Code](https://claude.com/claude-code)\n\nGenerated-By: Claude \u003cnoreply@anthropic.com\u003e\n\nChange-Id: I748b55420edad1d160998f02d86cd5fc00b43435\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/fdbd6ac68876a065e10d99761de9c2290bba7f75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/fdbd6ac68876a065e10d99761de9c2290bba7f75"}]},"branch":"refs/heads/master"},"ec7843cf0eb3001339a3d943c9af7cd589a37e52":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-11-06 05:15:19.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/75/964775/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/75/964775/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/75/964775/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/75/964775/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/75/964775/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/75/964775/2"}}},"commit":{"parents":[{"commit":"de25ea2fd2dd709e43f26186d9c4a9d3f71022f2","subject":"Merge \"Improve audit creation model and add text wrapping\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/de25ea2fd2dd709e43f26186d9c4a9d3f71022f2"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-10-24 13:12:00.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2025-11-06 04:56:09.000000000","tz":0},"subject":"Remove legacy integration test framework","message":"Remove legacy integration test framework\n\nFollowing the removal of integration test support in OpenStack Horizon\n(commit 49e5fe185a915ab80ccf4c130225371ade323711)i[1], this change completely\nremoves the legacy integration test framework from watcher-dashboard.\n\nChanges include:\n* Remove entire watcher_dashboard/test/integration_tests/ directory\n - Page object models for audit templates and audits\n - Selenium-based integration test cases\n - Integration test configuration files\n\n* Remove watcher_dashboard/test/selenium.py\n\n* Clean up run_tests.sh:\n - Remove selenium and integration test command-line options\n - Remove selenium configuration logic and environment variables\n - Remove run_integration_tests function\n - Simplify test exclusion logic\n\n* Update tox.ini to remove integration test exclusion flags\n\n* Remove selenium and xvfbwrapper dependencies from test-requirements.txt\n\n* Add release note documenting the removal\n\nLinks:\n[1]. https://review.opendev.org/c/openstack/horizon/+/960204\n\n🤖 Generated with [Claude Code](https://claude.com/claude-code)\n\nGenerated-By: Claude \u003cnoreply@anthropic.com\u003e\n\nChange-Id: I748b55420edad1d160998f02d86cd5fc00b43435\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/ec7843cf0eb3001339a3d943c9af7cd589a37e52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/ec7843cf0eb3001339a3d943c9af7cd589a37e52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
