)]}'
{"id":"openstack%2Fwatcher-dashboard~978171","triplet_id":"openstack%2Fwatcher-dashboard~master~Ia82cf23e67a22f03d13b6aa705a0d83c75b8cb34","project":"openstack/watcher-dashboard","branch":"master","attention_set":{},"removed_from_attention_set":{"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2026-03-03 00:19:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia82cf23e67a22f03d13b6aa705a0d83c75b8cb34","subject":"Simplify selected goal retrieval in Audit template CreateForm","status":"MERGED","created":"2026-02-27 08:45:13.000000000","updated":"2026-03-03 00:20:21.000000000","submitted":"2026-03-03 00:19:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978171","meta_rev_id":"6573735cdc3c24f147e50fa4f87dddbfd0c3a870","_number":978171,"virtual_id_number":978171,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-03 00:19:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2026-03-03 00:01:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2026-03-02 05:53:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2026-03-02 05:53:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":1,"date":"2026-03-02 08:33:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2026-03-02 16:49:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2026-03-03 00:01:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}],"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-27 08:45:22.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2026-02-27 09:01:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-27 11:18:27.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-02-27 12:03:00.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-02-27 13:59:57.000000000","updated_by":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"reviewer":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"state":"REVIEWER"},{"updated":"2026-03-02 16:49:10.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2026-03-03 00:01:40.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"}],"messages":[{"id":"5ce56543c4df12904e407dd5317e61ce9dbcb72c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-02-27 08:45:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a03abcb90900748450fa9a0bcc175d40b9536078","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-27 08:45:22.000000000","message":"Patch Set 1:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41d43933c8bea7b404acb8ffd39f9e964934db15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 09:01:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5a02745224a46ba8479412c7e33342c\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca4fa53fcae34d39a53c0d73e64a766c : SUCCESS in 2m 23s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/09601c43144345378f0068f1cf3cbfd4 : SUCCESS in 4m 59s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/32fa1bdf8c324d9596abc25a6ff4319f : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12549e1a9b1a4a2bb0854da4a6399103 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddddd7ddea4f40f1ae4492c0947d11f2 : SUCCESS in 2m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ea8af53365ac42619d83da703a16742d : FAILURE in 8m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95b527a71a1e48b9ab4f66d2f6b0c7c9 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31ae8dafa82c430697fd1762a949702f : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"019f75fa150ef8671a65762b65bedf3016abb0f4","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-27 09:02:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/0c42bb55f8024b55a8bd59437f28e4e6\n\n- teim-code-review https://zuul.teim.app/t/main/build/1926bf72119642159e26bb2d3f098608 : SUCCESS in 16m 36s\n\nWarning:\n  Comments left for invalid file watcher_dashboard/content/audit_templates/tests.py","accounts_in_message":[],"_revision_number":1},{"id":"9022a8eeef5ab5b0a5e05aae614575c166077139","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-27 11:18:27.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"38971fad6c2e43de32c1571668c3f73f5aa5a1e1","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-02-27 12:03:00.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/1d1e4a896d944517a07214c1ea78af2b\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/4d9fdaee5cc343368e69322c3fbb0df8 : SUCCESS in 2h 38m 12s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/aa0b3461309e4a70aa43b55c77833e79 : SUCCESS in 1h 59m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c2c8c1fb9b3cb25a45196868fbb4f3436ae23621","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-02-27 13:05:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b5df644151574ca6b54306042f578c2055b511bc","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-27 13:05:46.000000000","message":"Patch Set 2:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":2},{"id":"467ada6090769aa3167b40969b3eaf96e5b151b3","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-27 13:13:37.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/cab24acf13f14db886504e666143de49\n\n- teim-code-review https://zuul.teim.app/t/main/build/f7d2f0ffa55c423e89a1513fd1a90302 : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":2},{"id":"06a8cbeb4735033c54f1e928353a695dc1c0a93b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 13:18:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fe5ba0924814e3e89f8fbab17a71ad1\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0aa9ba5e8d7e4175bea74b08ac232b8d : SUCCESS in 4m 30s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/7e3dc34461c048c8a5f3709802079ddd : SUCCESS in 6m 48s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/ee054ee7362d4a69a864a56f35e7156c : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52fe6c52b08c4026a24739a432a3b272 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/645618437a8f48819ccc36bbd92d3d3a : SUCCESS in 3m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/68dd6b597d7c4f958087574508a9da4f : FAILURE in 6m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7327e198de1f47c4933e336babd0f5de : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/849612eced0b4415943cfa2890c6ebf4 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":2},{"id":"effb3028a82cb1c1d280a0b4fff9f29aac2b1052","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-02-27 13:59:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bb4c7695cde88a283accde9449254f4bf34a596a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-02-27 16:06:25.000000000","message":"Patch Set 2: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/718aa659f67c40808a2da958d7b543e5\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/2d45228e83154a52a162fb342aaaf5fb : SUCCESS in 2h 59m 16s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/fd5839eebfa8412bac1e8f8c1dabf553 : RETRY_LIMIT in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"4078f9cc411e3e0ac7bff9f764bca7a47ae511f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-03-02 05:53:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"af0cbeef8d32d62525df744b130da16bbc34ea5a","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-03-02 05:53:56.000000000","message":"Patch Set 3:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":3},{"id":"768a35a0073d2bedcbde92676065e75f4b0f3f36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 06:04:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6694cc20e0e84d23b24a0e63438d3b93\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/908b2226350e41b7934e36dfa9cd8e31 : SUCCESS in 4m 15s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/145c6f7991314df196c2785cfa40bfe2 : SUCCESS in 4m 00s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/3d07d68da94e4081a5bf30095c2a508f : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ae79036cbf44ba9b5bca0b4b6f46f68 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7dac171d5dfc42ddbdd6b884d25fe87f : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e413db04e94c472b9d88945d1900ae89 : FAILURE in 6m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55ef8937fd124a528044f3a8aef0e4c6 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/297d60032e0d46da9676ec356d788c91 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":3},{"id":"b3f50dc45b84853e33d2cdfb6852ee50dfebbf21","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-03-02 06:21:53.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/a60272a0c86c493dad1be2c830ee1a77\n\n- teim-code-review https://zuul.teim.app/t/main/build/8c9fd691927f4accb34654c8a8ba7588 : SUCCESS in 27m 48s","accounts_in_message":[],"_revision_number":3},{"id":"920b6c81937b707bed7ae7865ab594eee3043739","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-03-02 08:33:54.000000000","message":"Patch Set 3: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/b70f21524af04e8490eb6d3bb3abadda\n\n- openstack-meta-content-provider-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/f4bbdf6b7dd0494395ffd2120e7a956a : SUCCESS in 2h 38m 57s\n- watcher-operator-validation-master https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5b5cc9a082064563884dc77dbd7eaa6c : SUCCESS in 1h 59m 18s","accounts_in_message":[],"_revision_number":3},{"id":"736de623f1424ac97c88b8aefbbf89ab35c21990","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2026-03-02 16:49:10.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a94a4955dac184de3f34671d5054f78a979086c4","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2026-03-03 00:01:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"21469e4542861261e0eb100f5ba68609c7f02dcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 00:01:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6b87aa4175de6e187c8d7d0e9dcfaab71e0c3634","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 00:19:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a88343b45554a358fe2437ba9d6838b\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/254e4b40fd6f479fadfabed49d7a4cb0 : SUCCESS in 3m 03s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/cc8beaff8efc4601a91c13d444da34f7 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64904a0840e94d4d8f70bb3eccf9efdd : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aaa42bbf5e794b11a4ab7934a5c7cbcd : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2456eb241c8f482c98922a22ce989308 : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":3},{"id":"98dfa4e676216cfeef3d7e4ad9ab9c5c4d7bc933","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 00:19:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6573735cdc3c24f147e50fa4f87dddbfd0c3a870","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 00:20:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a183e858f64f4dc894ec1d728605f3bf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d04f16804d0b4f688dcd1ad35ead6f3e : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e5d4b16788c12eb02360f1cd487a8e457f958e1f","revisions":{"ec6e7e80d9c5d2759b9c262da6b2d5c8096f0d1d":{"kind":"REWORK","_number":1,"created":"2026-02-27 08:45:13.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/71/978171/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/71/978171/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/978171/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/978171/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/978171/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/978171/1"}}},"commit":{"parents":[{"commit":"0aaae03acc8126881dd87eb33ee3ef072f95a3bc","subject":"Centralize strategy display_name access into a shared helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/0aaae03acc8126881dd87eb33ee3ef072f95a3bc"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-02-27 08:38:24.000000000","tz":0},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-02-27 08:38:24.000000000","tz":0},"subject":"Simplify selected goal retrieval in Audit template CreateForm","message":"Simplify selected goal retrieval in Audit template CreateForm\n\nDjango forms always have self.data[1] available after __init__,\nwhich makes hasattr(self, \u0027data\u0027) guard redundant.\n\nIt simplifies to a direct self.data.get(\u0027goal\u0027) call with no\nbehavior change.\n\nLinks:\n[1].https://github.com/django/django/blob/stable/6.0.x/django/forms/forms.py#L89\n\nChange-Id: Ia82cf23e67a22f03d13b6aa705a0d83c75b8cb34\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/ec6e7e80d9c5d2759b9c262da6b2d5c8096f0d1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/ec6e7e80d9c5d2759b9c262da6b2d5c8096f0d1d"}]},"branch":"refs/heads/master"},"f2bc92cfa61c27d9e6b56f8e09459ac67615f930":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-27 13:05:35.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/71/978171/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/71/978171/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/978171/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/978171/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/978171/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/978171/2"}}},"commit":{"parents":[{"commit":"d6d1add8f5a47411700e695452059fbcfb557359","subject":"Centralize strategy display_name access into a shared helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/d6d1add8f5a47411700e695452059fbcfb557359"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-02-27 08:38:24.000000000","tz":0},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2026-02-27 13:05:36.000000000","tz":0},"subject":"Simplify selected goal retrieval in Audit template CreateForm","message":"Simplify selected goal retrieval in Audit template CreateForm\n\nDjango forms always have self.data[1] available after __init__,\nwhich makes hasattr(self, \u0027data\u0027) guard redundant.\n\nIt simplifies to a direct self.data.get(\u0027goal\u0027) call with no\nbehavior change.\n\nLinks:\n[1].https://github.com/django/django/blob/stable/6.0.x/django/forms/forms.py#L89\n\nChange-Id: Ia82cf23e67a22f03d13b6aa705a0d83c75b8cb34\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/f2bc92cfa61c27d9e6b56f8e09459ac67615f930"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/f2bc92cfa61c27d9e6b56f8e09459ac67615f930"}]},"branch":"refs/heads/master","description":"Rebase"},"e5d4b16788c12eb02360f1cd487a8e457f958e1f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-02 05:53:48.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/71/978171/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/71/978171/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/978171/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/978171/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/978171/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/71/978171/3"}}},"commit":{"parents":[{"commit":"2c8d13ccd3e09083098d487b81b173e18b3522e7","subject":"Centralize strategy display_name access into a shared helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/2c8d13ccd3e09083098d487b81b173e18b3522e7"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-02-27 08:38:24.000000000","tz":0},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2026-03-02 05:53:48.000000000","tz":0},"subject":"Simplify selected goal retrieval in Audit template CreateForm","message":"Simplify selected goal retrieval in Audit template CreateForm\n\nDjango forms always have self.data[1] available after __init__,\nwhich makes hasattr(self, \u0027data\u0027) guard redundant.\n\nIt simplifies to a direct self.data.get(\u0027goal\u0027) call with no\nbehavior change.\n\nLinks:\n[1].https://github.com/django/django/blob/stable/6.0.x/django/forms/forms.py#L89\n\nChange-Id: Ia82cf23e67a22f03d13b6aa705a0d83c75b8cb34\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/e5d4b16788c12eb02360f1cd487a8e457f958e1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/e5d4b16788c12eb02360f1cd487a8e457f958e1f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
