)]}'
{"id":"openstack%2Fwatcher-dashboard~980414","triplet_id":"openstack%2Fwatcher-dashboard~stable%2F2026.1~Ia847412f32d4f6a3696a577fe773e0cd47a5791e","project":"openstack/watcher-dashboard","branch":"stable/2026.1","topic":"create-2026.1","hashtags":[],"change_id":"Ia847412f32d4f6a3696a577fe773e0cd47a5791e","subject":"Update .gitreview for stable/2026.1","status":"MERGED","created":"2026-03-13 11:58:23.000000000","updated":"2026-03-19 14:19:55.000000000","submitted":"2026-03-19 14:18:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980414-create-2026.1","meta_rev_id":"a6f23800a9eb3be43685c1333fa61fa1a73793e0","_number":980414,"virtual_id_number":980414,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-19 14:18:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2026-03-19 13:28:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2026-03-19 14:04:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2026-03-19 14:04:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-13 12:16:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-19 13:28:58.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2026-03-19 14:04:32.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"a469ef2aa3e5447720844cc7b8ca05faa37063aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-03-13 11:58:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c955ca6cc7d8eef051032ac8015a8dea9508f4ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 12:16:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5729a817c3c04b6d91245e106a1a86d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1b76379eb424058a27f5a299bb1d86f : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23bb855bc8e44fa8a8e948b9bcb637fc : FAILURE in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4daba26629434540b2aa4188459ad29a : FAILURE in 5m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81189e47ee1c4e9383f9d28ff40a9784 : FAILURE in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ee8441c6bee4677a6449cb70ac6ecf4 : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7c289b2969b39c6e1dc64be1e61b3a27c7ce10e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"real_author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-03-19 13:15:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_22816\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"3144128034dd9a296cc398039d2634bdf6c64df5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 13:25:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc924fc668e34a94addff37c9aea3ce7\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50d96d3d53d64877ba4151b7ae08a42f : SUCCESS in 4m 02s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/b2113cb30e5048178c56e1bb659a680e : SUCCESS in 2m 30s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/8706009328a340b9ad4857c8fefbaf07 : SUCCESS in 2m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/870cda0845a34f138617f53062afbb0c : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b403e299e494169adc1dea70475245b : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23a1295b7cae4a5bb9656436f7fd0944 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/241261e7257840af941f753cf0ef99b0 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"14376806b9db381c8f3f7036dfcdd9e559943fed","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2026-03-19 13:28:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"155509b78a3c97e3c5b7fea856da3e873e0b69f3","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2026-03-19 14:04:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"27cccb0945f21adfc613e63f8f0461255081a39f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 14:04:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6772d0f0430bad7829980e4c53069e4ccec21344","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 14:18:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f74501c30ba74d2a862b8aac6a33d341\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81fcca91eb51456d85f58df36779fa71 : SUCCESS in 4m 10s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/7653569971334febad58566df5d595e1 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f71c5e4ab654a73a21581ebadb324cb : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b44c6c1b15446569046e232ec0481e5 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7752e471d7004b168770fc2ff7761c1e : SUCCESS in 7m 51s","accounts_in_message":[],"_revision_number":2},{"id":"7e4a6e821be915c6b6c8fb13b0f9c2b012cee3ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 14:18:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a6f23800a9eb3be43685c1333fa61fa1a73793e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 14:19:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3bf093f2d494bbea49ac5c36ca969eb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6654709eb1d4fe998f842c8a18b591f : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3e8bc95d41b35c175f7da3992baabee41fa2e603","revisions":{"bdb21075bfb166e606e6508d7082b686b8520724":{"kind":"REWORK","_number":1,"created":"2026-03-13 11:58:23.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/14/980414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/14/980414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/14/980414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/14/980414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/14/980414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/14/980414/1"}}},"commit":{"parents":[{"commit":"d03dbda70d8ca090756a03dd6408303bf5d4479b","subject":"Merge \"Add skip action UI controls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/d03dbda70d8ca090756a03dd6408303bf5d4479b"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-13 11:58:22.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-13 11:58:22.000000000","tz":0},"subject":"Update .gitreview for stable/2026.1","message":"Update .gitreview for stable/2026.1\n\nChange-Id: Ia847412f32d4f6a3696a577fe773e0cd47a5791e\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/bdb21075bfb166e606e6508d7082b686b8520724"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/bdb21075bfb166e606e6508d7082b686b8520724"}]},"branch":"refs/heads/stable/2026.1"},"3e8bc95d41b35c175f7da3992baabee41fa2e603":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-19 13:15:26.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/14/980414/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-dashboard","ref":"refs/changes/14/980414/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/14/980414/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/14/980414/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-dashboard refs/changes/14/980414/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-dashboard refs/changes/14/980414/2"}}},"commit":{"parents":[{"commit":"8a6bf6555d0dcf807f7a81fe31357492d8d62f0e","subject":"Fix fontawesome.css patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/8a6bf6555d0dcf807f7a81fe31357492d8d62f0e"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-13 11:58:22.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-03-19 13:15:26.000000000","tz":0},"subject":"Update .gitreview for stable/2026.1","message":"Update .gitreview for stable/2026.1\n\nChange-Id: Ia847412f32d4f6a3696a577fe773e0cd47a5791e\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/3e8bc95d41b35c175f7da3992baabee41fa2e603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-dashboard/commit/3e8bc95d41b35c175f7da3992baabee41fa2e603"}]},"branch":"refs/heads/stable/2026.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
