)]}'
{"id":"openstack%2Fwatcher-specs~683268","triplet_id":"openstack%2Fwatcher-specs~master~I501545c4c3c3caa9329e62da6cab99db4464fa44","project":"openstack/watcher-specs","branch":"master","hashtags":[],"change_id":"I501545c4c3c3caa9329e62da6cab99db4464fa44","subject":"Fix the page display","status":"MERGED","created":"2019-09-20 02:46:37.000000000","updated":"2019-09-20 06:25:18.000000000","submitted":"2019-09-20 06:24:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683268-1568960649179-805a2d6b","meta_rev_id":"bdd27f3e4a0107fa0e542ee4a78260761b558c32","_number":683268,"virtual_id_number":683268,"owner":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"value":2,"date":"2019-09-20 06:24:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"all":[{"value":2,"date":"2019-09-20 06:14:37.000000000","_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"all":[{"value":1,"date":"2019-09-20 06:14:37.000000000","_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 06:24:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7abb59a019a1d05fb0d842ce32df7656f7663167","author":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"date":"2019-09-20 02:46:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"111d33ac4642589ff96e1af5cb992c0e0d460f02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 02:57:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7cbfed1ec30456f8743d4d5cae20e34 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a045eb9f3fd42e8b3b8ed9a32dfc37e : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c1bfe53c5b83e7a8fd00bf3eb3e268e67b9d94a1","author":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"date":"2019-09-20 06:14:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3902b7c37d0ec87221559c99551d720dded597e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 06:14:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cda918cb048ca516c5748a96100169ae84795af7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 06:24:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa40fd63ac90431b88acb932fd1407f2 : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05ee964a19234944928da0f6bc0a9b5f : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"455046ccc65d6576643647d7358af5321676b8bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 06:24:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bdd27f3e4a0107fa0e542ee4a78260761b558c32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 06:25:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/f0b0d2cf7c72415a99244be018df6289 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9de0fa56ae080bc07a9f65a17e02ab2abe643a7e","revisions":{"9de0fa56ae080bc07a9f65a17e02ab2abe643a7e":{"kind":"REWORK","_number":1,"created":"2019-09-20 02:46:37.000000000","uploader":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"ref":"refs/changes/68/683268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/68/683268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/68/683268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/68/683268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/68/683268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/68/683268/1"}}},"commit":{"parents":[{"commit":"0366bb38d72fc37923288b7c975ac954b085d47f","subject":"Merge \"General purpose threadpool for the decision engine\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/0366bb38d72fc37923288b7c975ac954b085d47f"}]}],"author":{"name":"licanwei","email":"li.canwei2@zte.com.cn","date":"2019-09-20 02:46:13.000000000","tz":-420},"committer":{"name":"licanwei","email":"li.canwei2@zte.com.cn","date":"2019-09-20 02:46:13.000000000","tz":-420},"subject":"Fix the page display","message":"Fix the page display\n\nChange-Id: I501545c4c3c3caa9329e62da6cab99db4464fa44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/9de0fa56ae080bc07a9f65a17e02ab2abe643a7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/9de0fa56ae080bc07a9f65a17e02ab2abe643a7e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
