)]}'
{"id":"openstack%2Fwatcher-specs~899026","triplet_id":"openstack%2Fwatcher-specs~master~I54b3c578f677ad7d554732a5018163e4780f9457","project":"openstack/watcher-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"28748":{"account":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"last_update":"2023-10-31 05:27:51.000000000","reason":"\u003cGERRIT_ACCOUNT_28748\u003e replied on the change","reason_account":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}},"8543":{"account":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"last_update":"2023-10-31 05:43:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I54b3c578f677ad7d554732a5018163e4780f9457","subject":"Add MAAS spec","status":"MERGED","created":"2023-10-23 09:06:00.000000000","updated":"2023-10-31 05:44:22.000000000","submitted":"2023-10-31 05:43:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899026","meta_rev_id":"8d66d47d2d2698e19d6de79293e5f517b1389c24","_number":899026,"virtual_id_number":899026,"owner":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-31 05:43:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"all":[{"value":2,"date":"2023-10-31 05:27:51.000000000","_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"all":[{"value":1,"date":"2023-10-31 05:27:51.000000000","_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-23 09:10:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-23 11:52:32.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"CC"},{"updated":"2023-10-31 05:27:51.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"}],"messages":[{"id":"9ed437eae40efcee5aba5ee1470532a56ba63632","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2023-10-23 09:06:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac6a1b55bc4aae4eb3d95784e86af679596cc11c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 09:10:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/acdd95c93ab44dce9fa0dd7803fb48cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acc4b308805d473d90fa2317dabeca94 : FAILURE in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45957dbbfadd49d7870da6e75d58ea81 : FAILURE in 2m 25s","accounts_in_message":[],"_revision_number":1},{"id":"5eac704557f8bbeb8c9c11264260fb09eca9e34c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2023-10-23 09:26:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7992a8204a5d4086fec40559f3c121e10fc7406a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 09:35:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/31652886d3d9494a9b06d06cfda42fbd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa7509a9007a4892ac1a202bcd46c590 : FAILURE in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a6817396ab14566b84759480b1a5008 : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c21c35643565ceff4aadd1d98de8252a9f0eaee4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2023-10-23 11:25:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ff0de598f407e445f9016dc92fd6c0d36dec8f4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 11:36:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/245c7db48bb04aed9c530f3cb7812dd6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8f3769f9cf2448f93f39b8c367aa7fb : FAILURE in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db0cc79f0f0b45bc966c078e8ac4d6f4 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":3},{"id":"69039cdcec6ab196ca7bba275243b1843b37b8b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2023-10-23 11:48:26.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9b2131a19306be4b129e556854ab02564cca2c2e","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2023-10-23 11:52:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7ecaa059dc3e0b02ee56aae1f27532c23243a142","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2023-10-23 11:56:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d9543f4203c6efc2697af7010165b1febc17980c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 11:58:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddebee701a9b440ba05c952797622f58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/373717080b844f859278fb207fba042a : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d222b21fab249e7ace4de6845bad2c1 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":4},{"id":"e9bb3db6b85869b29d3d383a4f65817477f370eb","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2023-10-31 05:27:51.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"16a0b1eed6fc247366c63bee9e4baf5ed9f57166","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 05:28:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f9481c67c238b7d01a1f67de131317b1209a9980","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 05:43:17.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5617ea56ad3d40908b51caccfb5a9419\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d7202ad4d26470d8732e05e7cf7bd80 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b9d498671c640d6b9cf76cfdfb29384 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":4},{"id":"13d30f0347bc8ea6ec751cc07b2d3901a5c73225","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 05:43:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"8d66d47d2d2698e19d6de79293e5f517b1389c24","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 05:44:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/255e182e37d74975b05c50db41f44aef\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/6e2a2ded96ac4e1099af3d269ff804c2 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ea19cfd56e362f106dbf6e66851337df4cbbdd3d","revisions":{"86cfa8ae5cb0afb88318e0ad2a4e4a52c2a8d2d3":{"kind":"REWORK","_number":1,"created":"2023-10-23 09:06:00.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/26/899026/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/26/899026/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/26/899026/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/26/899026/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/26/899026/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/26/899026/1"}}},"commit":{"parents":[{"commit":"86df397e3e6bd00756314dede7b50426724a1a45","subject":"Merge \"Specificiation for time series framework and analysis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/86df397e3e6bd00756314dede7b50426724a1a45"}]}],"author":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2023-10-23 09:01:19.000000000","tz":180},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2023-10-23 09:05:58.000000000","tz":180},"subject":"Add MAAS spec","message":"Add MAAS spec\n\nWe intend to add support for MAAS, another bare metal provisioning\nand management service.\n\nChange-Id: I54b3c578f677ad7d554732a5018163e4780f9457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/86cfa8ae5cb0afb88318e0ad2a4e4a52c2a8d2d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/86cfa8ae5cb0afb88318e0ad2a4e4a52c2a8d2d3"}]},"branch":"refs/heads/master"},"e5d4278a6f4a2dcad752bae434d26f8ff124e4ca":{"kind":"REWORK","_number":2,"created":"2023-10-23 09:26:54.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/26/899026/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/26/899026/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/26/899026/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/26/899026/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/26/899026/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/26/899026/2"}}},"commit":{"parents":[{"commit":"86df397e3e6bd00756314dede7b50426724a1a45","subject":"Merge \"Specificiation for time series framework and analysis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/86df397e3e6bd00756314dede7b50426724a1a45"}]}],"author":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2023-10-23 09:01:19.000000000","tz":180},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2023-10-23 09:26:11.000000000","tz":180},"subject":"Add MAAS spec","message":"Add MAAS spec\n\nWe intend to add support for MAAS, another bare metal provisioning\nand management service.\n\nNote that the tox job fails at the moment as \"whitelist_externals\"\nhas been deprecated in favor of \"allowlist_externals\", so we\u0027ll\nneed to address that as well.\n\nChange-Id: I54b3c578f677ad7d554732a5018163e4780f9457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/e5d4278a6f4a2dcad752bae434d26f8ff124e4ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/e5d4278a6f4a2dcad752bae434d26f8ff124e4ca"}]},"branch":"refs/heads/master"},"55e63dee9bbfee5b3d3344f54e1f7396f0be4c53":{"kind":"REWORK","_number":3,"created":"2023-10-23 11:25:34.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/26/899026/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/26/899026/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/26/899026/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/26/899026/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/26/899026/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/26/899026/3"}}},"commit":{"parents":[{"commit":"86df397e3e6bd00756314dede7b50426724a1a45","subject":"Merge \"Specificiation for time series framework and analysis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/86df397e3e6bd00756314dede7b50426724a1a45"}]}],"author":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2023-10-23 09:01:19.000000000","tz":180},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2023-10-23 11:22:52.000000000","tz":180},"subject":"Add MAAS spec","message":"Add MAAS spec\n\nWe intend to add support for MAAS, another bare metal provisioning\nand management service.\n\nNote that the tox job fails at the moment as \"whitelist_externals\"\nhas been deprecated in favor of \"allowlist_externals\", so we\u0027ll\nneed to address that as well.\n\nAnother issue that we need to address is that sphinx is unable to\nlocate some image files:\n\n  /home/zuul/src/opendev.org/openstack/watcher-specs/doc/source/\n  specs/newton/implemented/scoring-module.rst:232: WARNING: image\n\n  file not readable: doc/source/images/scoring-module-deployment.png\n\nTo address this, we\u0027ll use paths relative to the doc/source folder.\n\nChange-Id: I54b3c578f677ad7d554732a5018163e4780f9457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/55e63dee9bbfee5b3d3344f54e1f7396f0be4c53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/55e63dee9bbfee5b3d3344f54e1f7396f0be4c53"}]},"branch":"refs/heads/master"},"ea19cfd56e362f106dbf6e66851337df4cbbdd3d":{"kind":"REWORK","_number":4,"created":"2023-10-23 11:48:26.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/26/899026/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/26/899026/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/26/899026/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/26/899026/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/26/899026/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/26/899026/4"}}},"commit":{"parents":[{"commit":"86df397e3e6bd00756314dede7b50426724a1a45","subject":"Merge \"Specificiation for time series framework and analysis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/86df397e3e6bd00756314dede7b50426724a1a45"}]}],"author":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2023-10-23 09:01:19.000000000","tz":180},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2023-10-23 11:47:27.000000000","tz":180},"subject":"Add MAAS spec","message":"Add MAAS spec\n\nWe intend to add support for MAAS, another bare metal provisioning\nand management service.\n\nNote that the tox job fails at the moment as \"whitelist_externals\"\nhas been deprecated in favor of \"allowlist_externals\", so we\u0027ll\nneed to address that as well.\n\nAnother issue that we need to fix is that sphinx is unable to\nlocate some image files:\n\n  /home/zuul/src/opendev.org/openstack/watcher-specs/doc/source/\n  specs/newton/implemented/scoring-module.rst:232: WARNING: image\n\n  file not readable: doc/source/images/scoring-module-deployment.png\n\nChange-Id: I54b3c578f677ad7d554732a5018163e4780f9457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/ea19cfd56e362f106dbf6e66851337df4cbbdd3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/ea19cfd56e362f106dbf6e66851337df4cbbdd3d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
