)]}'
{"id":"openstack%2Fwatcher-specs~947271","triplet_id":"openstack%2Fwatcher-specs~master~I18738993092eb9ca9b8e185c6e3bcaba417d0e52","project":"openstack/watcher-specs","branch":"master","topic":"2025_2_specs","attention_set":{},"removed_from_attention_set":{"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2025-05-06 12:17:16.000000000","reason":"Change was submitted"},"34452":{"account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"last_update":"2025-05-06 12:17:16.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-05-06 12:10:26.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2025-05-06 12:17:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I18738993092eb9ca9b8e185c6e3bcaba417d0e52","subject":"Update 2025.2 spec directory and move spec implemented in 2025.1","status":"MERGED","created":"2025-04-15 17:56:24.000000000","updated":"2025-05-06 12:18:07.000000000","submitted":"2025-05-06 12:17:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947271-2025_2_specs","meta_rev_id":"9594574e52b42383e0518da3392fd847d4aa7481","_number":947271,"virtual_id_number":947271,"owner":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-06 12:17:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":2,"date":"2025-05-06 12:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":1,"date":"2025-05-06 12:10:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-15 18:07:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-16 13:33:33.000000000","updated_by":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"reviewer":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"state":"REVIEWER"},{"updated":"2025-04-17 08:52:33.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2025-05-02 18:07:56.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"08e4f7b1d37123517af21746009a07a861a257f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-04-15 17:56:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f299440ef7c55133e5256168a4c83eaf20171aee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 18:07:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d123d4596a048aaab26d9f6c6981937\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4d4b07a67cd42d898ed86885ecd1828 : FAILURE in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a34f05bfce4e42739cee2b76757907d4 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8b9da02a0eaa967e1d66706975c5c29414753b87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-04-15 18:14:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"434b45f5055c976132b2159fe0a2bb6e2ac50825","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 18:18:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d7a12c349fa4008b3d3de5793ec30ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2289e1af57f44788e747bd8fe6331bb : FAILURE in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24d6128f915f4cb3b493c6d1d2e3c971 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":2},{"id":"73cd605a90df085ce52367ae8e3a829307946451","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-04-15 19:31:02.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d6a2bf5c8bd5044cecd6a010cf560336f1f99d36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-04-15 19:33:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4628924dc96550b980919e8f4d7ef82dc282e2ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 19:39:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ea2c6fc494b43f6a2e2c15302eab6ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da786bd474974791a5cb16e3d3c77bcc : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54ea09f348c1424b92adb867b1c128cd : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":4},{"id":"a80c5f01bdf10ed4de9edc069dac847ece821f19","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2025-04-16 13:33:33.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e3f62cf3fbdd29114e1af0c5f99d06514546ce8e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-04-17 08:52:33.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"da6ab7d5154e89f8c7215a99b4a6e5e8137eddd5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-02 18:07:56.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4dde1d0e2dafb22d63e38138ad92d58782d6f41d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-05-06 11:34:24.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"cfe46f513b059fd0db0fa57d35603d76b7258e26","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-05-06 11:35:30.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"42b847b357816ec725e1e9126981af785d5821c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 11:45:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73aa2ec16a144e9dbeba60912728b55c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd2afea30fa54c6b9e37ee13bb1d671c : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1a877f44c3d4b61a6cfd49f3ff0f015 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":5},{"id":"4e26a8f368fb60a18f5a6c572b790c16ea50735b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-06 12:10:26.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"df9e91feba1411758434f1f4a492234c77cb63b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 12:11:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f2d9d6adccbbb5a615c99529a54047b23e360b01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 12:17:16.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c760b8db02944998b96317555b92eb80\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98b0361b8fbb4b0ebaa99b4a36cacdd7 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5070368f92bb4cd19804d60960276b91 : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":5},{"id":"07f2c0cbe76da9d52dbad9fb2d1a3e4ed0867ccb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 12:17:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"9594574e52b42383e0518da3392fd847d4aa7481","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 12:18:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f64f685f62c14743b2cd5b8a960880e5\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/990efaa203a74625a447a68f579bf977 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f936a2d97679fad40cf48b40e2bf6365c3b2ce0b","revisions":{"fb07b7b691616b19d4ab2876f30b9d24b3fdc414":{"kind":"REWORK","_number":1,"created":"2025-04-15 17:56:24.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/71/947271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/71/947271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/1"}}},"commit":{"parents":[{"commit":"ab703ff1cba72b2ab71976dc9399e1943bf24124","subject":"tox: Drop envdir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/ab703ff1cba72b2ab71976dc9399e1943bf24124"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 17:53:58.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 17:55:51.000000000","tz":-180},"subject":"Update 2025.2 spec directory and move spec implemented in 2025.1","message":"Update 2025.2 spec directory and move spec implemented in 2025.1\n\nUpdate this repository to include 2025.2 directory for specs.\nIncludes a spec template for 2025.2 that is based on the previous one.\nMoves the implemented spec in 2025.1 to the implemented directory.\n\nChange-Id: I18738993092eb9ca9b8e185c6e3bcaba417d0e52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/fb07b7b691616b19d4ab2876f30b9d24b3fdc414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/fb07b7b691616b19d4ab2876f30b9d24b3fdc414"}]},"branch":"refs/heads/master"},"842ecad213befd7dc8c6729ed12268084a3fc343":{"kind":"REWORK","_number":2,"created":"2025-04-15 18:14:50.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/71/947271/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/71/947271/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/2"}}},"commit":{"parents":[{"commit":"ab703ff1cba72b2ab71976dc9399e1943bf24124","subject":"tox: Drop envdir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/ab703ff1cba72b2ab71976dc9399e1943bf24124"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 17:53:58.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 18:14:32.000000000","tz":-180},"subject":"Update 2025.2 spec directory and move spec implemented in 2025.1","message":"Update 2025.2 spec directory and move spec implemented in 2025.1\n\nUpdate this repository to include 2025.2 directory for specs.\nIncludes a spec template for 2025.2 that is based on the previous one.\nMoves the implemented spec in 2025.1 to the implemented directory.\n\nChange-Id: I18738993092eb9ca9b8e185c6e3bcaba417d0e52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/842ecad213befd7dc8c6729ed12268084a3fc343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/842ecad213befd7dc8c6729ed12268084a3fc343"}]},"branch":"refs/heads/master"},"cb657bf06ced4414f633a198faa7e1a640dc13e4":{"kind":"REWORK","_number":3,"created":"2025-04-15 19:31:02.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/71/947271/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/71/947271/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/3"}}},"commit":{"parents":[{"commit":"ab703ff1cba72b2ab71976dc9399e1943bf24124","subject":"tox: Drop envdir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/ab703ff1cba72b2ab71976dc9399e1943bf24124"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 17:53:58.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 19:30:26.000000000","tz":-180},"subject":"Update 2025.2 spec directory and move spec implemented in 2025.1","message":"Update 2025.2 spec directory and move spec implemented in 2025.1\n\nUpdate this repository to include 2025.2 directory for specs.\nIncludes a spec template for 2025.2 that is based on the previous one.\nMoves the implemented spec in 2025.1 to the implemented directory.\n\nChange-Id: I18738993092eb9ca9b8e185c6e3bcaba417d0e52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/cb657bf06ced4414f633a198faa7e1a640dc13e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/cb657bf06ced4414f633a198faa7e1a640dc13e4"}]},"branch":"refs/heads/master"},"a73b939f67e09963817582288a5e12296736a89b":{"kind":"REWORK","_number":4,"created":"2025-04-15 19:33:44.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/71/947271/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/71/947271/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/4"}}},"commit":{"parents":[{"commit":"ab703ff1cba72b2ab71976dc9399e1943bf24124","subject":"tox: Drop envdir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/ab703ff1cba72b2ab71976dc9399e1943bf24124"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 17:53:58.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 19:33:33.000000000","tz":-180},"subject":"Update 2025.2 spec directory and move spec implemented in 2025.1","message":"Update 2025.2 spec directory and move spec implemented in 2025.1\n\nUpdate this repository to include 2025.2 directory for specs.\nIncludes a spec template for 2025.2 that is based on the previous one.\nMoves the implemented spec in 2025.1 to the implemented directory.\n\nChange-Id: I18738993092eb9ca9b8e185c6e3bcaba417d0e52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/a73b939f67e09963817582288a5e12296736a89b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/a73b939f67e09963817582288a5e12296736a89b"}]},"branch":"refs/heads/master"},"f936a2d97679fad40cf48b40e2bf6365c3b2ce0b":{"kind":"REWORK","_number":5,"created":"2025-05-06 11:34:24.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/71/947271/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/71/947271/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/71/947271/5"}}},"commit":{"parents":[{"commit":"ab703ff1cba72b2ab71976dc9399e1943bf24124","subject":"tox: Drop envdir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/ab703ff1cba72b2ab71976dc9399e1943bf24124"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 17:53:58.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-05-06 11:29:29.000000000","tz":-180},"subject":"Update 2025.2 spec directory and move spec implemented in 2025.1","message":"Update 2025.2 spec directory and move spec implemented in 2025.1\n\nUpdate this repository to include 2025.2 directory for specs.\nIncludes a spec template for 2025.2 that is based on the previous one.\nMoves the implemented spec in 2025.1 to the implemented directory.\n\nChange-Id: I18738993092eb9ca9b8e185c6e3bcaba417d0e52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/f936a2d97679fad40cf48b40e2bf6365c3b2ce0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/f936a2d97679fad40cf48b40e2bf6365c3b2ce0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
