)]}'
{"id":"openstack%2Fwatcher-specs~947282","triplet_id":"openstack%2Fwatcher-specs~master~I3dcf4f9d24d326951594db122a2978b40466a081","project":"openstack/watcher-specs","branch":"master","topic":"bp/extend-compute-model-attributes","attention_set":{},"removed_from_attention_set":{"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2025-07-03 18:51:36.000000000","reason":"\u003cGERRIT_ACCOUNT_30002\u003e replied on the change","reason_account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},"34452":{"account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"last_update":"2025-07-02 10:59:54.000000000","reason":"\u003cGERRIT_ACCOUNT_34452\u003e replied on the change","reason_account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-07-01 12:32:18.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2025-07-03 18:59:09.000000000","reason":"Change was submitted"},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2025-07-03 03:34:09.000000000","reason":"\u003cGERRIT_ACCOUNT_12393\u003e replied on the change","reason_account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}}},"hashtags":[],"change_id":"I3dcf4f9d24d326951594db122a2978b40466a081","subject":"Adds spec for extend compute model attributes","status":"MERGED","created":"2025-04-15 19:51:39.000000000","updated":"2025-07-03 19:00:14.000000000","submitted":"2025-07-03 18:59:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":71,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"947282-bp/extend-compute-model-attributes","meta_rev_id":"4867667c10e50d8270e952787bc4676efbb8b7f7","_number":947282,"virtual_id_number":947282,"owner":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-03 18:59:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2025-07-03 18:51:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2025-07-01 12:32:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2025-07-02 10:59:54.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2025-07-03 03:34:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2025-07-03 18:51:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-15 19:58:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-16 14:35:56.000000000","updated_by":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"reviewer":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"state":"CC"},{"updated":"2025-06-03 15:56:05.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-06-04 10:04:55.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"CC"},{"updated":"2025-06-26 06:54:35.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2025-06-27 04:34:14.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2025-06-27 08:46:02.000000000","updated_by":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"reviewer":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"state":"REVIEWER"}],"messages":[{"id":"65f8b3c20aa022c069c5379b97c6101dc295ee24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-04-15 19:51:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45c2643fa3888ec24d2c0626b1a120c0d214d812","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 19:58:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/30163dc054d2440289251194472979ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c5a7a1f23a545d09f3ffc56db744930 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c7f7576e6eb4a6eae2f11c10d97eba8 : FAILURE in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5e079e97e8df5869bc1ff51cfe7d285f44e46dac","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-04-16 11:32:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"02be30cb4824870e818f411c89b67067a38c2964","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2025-04-16 14:35:56.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"743b2a8e1c6a83f8902e0ffb4ab3d51ee6c780ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-05-06 11:34:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"55bb434c7c4521481ce3266927d37f6775e2021e","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-05-06 11:36:06.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4a79e19128d57f8f18584b3a9323a864fb5513a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 11:46:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5ae757dd37345e3a2f3e54b00c90704\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf0f20d42cc4263b356ebc4d974b904 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b7055d77fde478d960e85cb051afcff : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":2},{"id":"df6a55c52667d41f93158d815baeef90f8e2c066","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-05-21 20:51:06.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b455c44a6241d5ff79538eded0d4eb5e6fed67cd","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-05-22 19:21:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eee623738a9973e072377fa012c601e3571afad8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-05-28 12:37:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b414a2051c6c55e8132544cce15ac9cbcd1670aa","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-05-28 12:38:29.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e9192fbda39fd76be42a6ce5d3d61fc06b91d003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 12:51:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/832871f78bb24f4e94e084ec0195d3a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c15e3229c414daf8afd52a823d7363b : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f682a7b226343899d4e4d8182e5e52d : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":3},{"id":"e666a2ab4cdddfc0fc85c53a22b7b958b65c3131","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-03 15:56:05.000000000","message":"Patch Set 3: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0d54d2b38e49c4a73b5ff9e4e9a64f420e3d010b","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-06-03 19:28:40.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"dea610b720e24c709712b82edb0cb445c5abc440","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2025-06-04 10:04:55.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"12ca422a837e41d21f6a456118aa7b102dfe2756","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-06-05 19:11:28.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ef54aceede71bb94160ab63c5c93145d4e0f746d","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-06-05 19:21:38.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"10577880251c054c2b2d49b2dcdd40ae95e8984d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 19:30:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da3b159e34b942cd94e15abf93863377\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df4485efb13040e6be75848d003e98ee : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f4140e7040b488db43e0c2f2fdfb4df : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":4},{"id":"b95d3035979fbb9d0638e1fced5bf52b1fe5b815","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-24 11:41:37.000000000","message":"Patch Set 4: Code-Review+1\n\n(13 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e8ad6d23b9bba0fce45e3c7e27d745d8f39388f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-06-24 17:41:57.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"2d4ee7d56f1ed033f3c9a7dec1c31a01bb477fe5","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-06-24 17:42:22.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"67d48f33d20e5c8a11ab1ac3d6a05fb76f9bf67b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 17:51:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6961c1504ee24653aee9595931e30ff9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b303b26111a54b6a9a666c6d44c3a3ed : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/875646921f7944ba85305049a04cdf85 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":5},{"id":"8eec468de31ddfc2f04e51961c486562a34d949b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-25 16:39:06.000000000","message":"Patch Set 5: Code-Review+2\n\n(8 comments)","accounts_in_message":[],"_revision_number":5},{"id":"82445327f0a2c1574cf625c861c749b15a6b38a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-06-25 18:27:25.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"d232519c321db0977f003ec56611eb39482d2cb8","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-06-25 18:28:11.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"70e95db1fe1b167e4d733e23fd70263b40abb81f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 18:33:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7de5ddbd3eb644088b4992ebcb216e9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66110ebd833340918854ef3af59b2057 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2158070d89764b86991ebc1f82d73876 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":6},{"id":"b9b37cdbea4ddd895e95adc6bea81ee1a766629e","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2025-06-26 06:54:35.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"bb4f8c54241df92eabed257568488300653fd8cf","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-06-27 04:34:14.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a796809d942ed620047c3caa5857e4b6a8c4901c","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2025-06-27 08:46:02.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"86f8c1f31d947ce458107d6846f25cdfebd5096f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-30 13:34:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"22c9573417145835f6730c4cac265cf2a7907dde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-06-30 18:42:51.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"6bfbe5cffe8b6f3a5f27dc58fb318ce0f0936ff0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 18:50:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a50909a3cf04786bae2ddcc0e10b08b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a05efc5fff7540dea716d5a363085a6f : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e996ddfd6934319ba252d72fd4a951a : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":7},{"id":"ed7e26f3f38c92125dcc4aa2219671fe9559bfd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-06-30 18:56:54.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"5f0be1417a5af567eaf686af7a374cf8a8c42990","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-06-30 18:57:53.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"6827edc3e6f0930164c09fb49bcdac4bd0e83bb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 19:04:17.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/936ebec2c68446dfb227d285e2b2ec18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30b9780e38cd44c5870a2fe9bb663540 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f85d2e2953824968afbeb64c11ae48d7 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":8},{"id":"8ea6a4072bd151483ea7fe14d185b4a749abbc5e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-01 12:32:18.000000000","message":"Patch Set 8: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"4b76a9f5acbe84d0215e3a5d384d46fc9942652b","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2025-07-02 10:59:54.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"12201d6dbcc3bfdd0769b63ad1badef1e63ad3d3","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-07-03 03:34:09.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5231306b578a07360aa2722ed516f7b1a5227428","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-07-03 18:51:36.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"018ddb5abb60a55157022e3553ae7428886c24f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 18:51:46.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"626eb881a0d205eab2125ce08df0a8719db88871","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 18:59:09.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/324185c05e2a4290bd48c34392a20e60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/636ce586de4e4a51a3919b55fcd2c19b : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5e8c3831bf1419db5d4f7bf58583a0e : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":8},{"id":"a896dfc7e647015f0b64e05112a2aac7cc0787f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 18:59:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"4867667c10e50d8270e952787bc4676efbb8b7f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 19:00:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3aba5a0e193e494bae2f4cf86f8d920e\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/42bcfecf094a4036a02b410a2cac16cb : SUCCESS in 58s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"1e06f0ae5bc3a9d55e2411e3e5a1da1352b379db","revisions":{"c4659d4bdfc48eceaa2c6e27c73cf1859adfba78":{"kind":"REWORK","_number":1,"created":"2025-04-15 19:51:39.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/82/947282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/82/947282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/1"}}},"commit":{"parents":[{"commit":"a73b939f67e09963817582288a5e12296736a89b","subject":"Update 2025.2 spec directory and move spec implemented in 2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/a73b939f67e09963817582288a5e12296736a89b"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 19:49:49.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 19:49:49.000000000","tz":-180},"subject":"Adds spec for extend compute model attributes","message":"Adds spec for extend compute model attributes\n\nThis spec provide more details about the blueprint to\nextend the compute model to include new instance attributes\nand improve watcher awareness about Nova\u0027s contraints\n\nChange-Id: I3dcf4f9d24d326951594db122a2978b40466a081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/c4659d4bdfc48eceaa2c6e27c73cf1859adfba78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/c4659d4bdfc48eceaa2c6e27c73cf1859adfba78"}]},"branch":"refs/heads/master"},"8a48696d7715269566dc5e7fa28b3c617dc3e2e1":{"kind":"REWORK","_number":2,"created":"2025-05-06 11:34:24.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/82/947282/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/82/947282/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/2"}}},"commit":{"parents":[{"commit":"f936a2d97679fad40cf48b40e2bf6365c3b2ce0b","subject":"Update 2025.2 spec directory and move spec implemented in 2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/f936a2d97679fad40cf48b40e2bf6365c3b2ce0b"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 19:49:49.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-05-06 11:33:44.000000000","tz":-180},"subject":"Adds spec for extend compute model attributes","message":"Adds spec for extend compute model attributes\n\nThis spec provide more details about the blueprint to\nextend the compute model to include new instance attributes\nand improve watcher awareness about Nova\u0027s contraints\n\nChange-Id: I3dcf4f9d24d326951594db122a2978b40466a081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/8a48696d7715269566dc5e7fa28b3c617dc3e2e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/8a48696d7715269566dc5e7fa28b3c617dc3e2e1"}]},"branch":"refs/heads/master"},"13dae9f6646a9f01d8259bafc5939e9210422bec":{"kind":"REWORK","_number":3,"created":"2025-05-28 12:37:49.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/82/947282/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/82/947282/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/3"}}},"commit":{"parents":[{"commit":"b5883e981ef3b45e6bfd6e2834f251e7be873e30","subject":"Merge \"Replace UPPER_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/b5883e981ef3b45e6bfd6e2834f251e7be873e30"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 19:49:49.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-05-28 12:37:34.000000000","tz":-180},"subject":"Adds spec for extend compute model attributes","message":"Adds spec for extend compute model attributes\n\nThis spec provide more details about the blueprint to\nextend the compute model to include new instance attributes\nand improve watcher awareness about Nova\u0027s contraints\n\nChange-Id: I3dcf4f9d24d326951594db122a2978b40466a081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/13dae9f6646a9f01d8259bafc5939e9210422bec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/13dae9f6646a9f01d8259bafc5939e9210422bec"}]},"branch":"refs/heads/master"},"9a8d8119a466f069bd6084764b4e37c50fef270d":{"kind":"REWORK","_number":4,"created":"2025-06-05 19:11:28.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/82/947282/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/82/947282/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/4"}}},"commit":{"parents":[{"commit":"b5883e981ef3b45e6bfd6e2834f251e7be873e30","subject":"Merge \"Replace UPPER_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/b5883e981ef3b45e6bfd6e2834f251e7be873e30"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 19:49:49.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-06-05 19:08:40.000000000","tz":-180},"subject":"Adds spec for extend compute model attributes","message":"Adds spec for extend compute model attributes\n\nThis spec provide more details about the blueprint to\nextend the compute model to include new instance attributes\nand improve watcher awareness about Nova\u0027s contraints\n\nChange-Id: I3dcf4f9d24d326951594db122a2978b40466a081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/9a8d8119a466f069bd6084764b4e37c50fef270d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/9a8d8119a466f069bd6084764b4e37c50fef270d"}]},"branch":"refs/heads/master"},"c66c8e07f5121367be93092481f7ca7d78205e05":{"kind":"REWORK","_number":5,"created":"2025-06-24 17:41:57.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/82/947282/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/82/947282/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/5"}}},"commit":{"parents":[{"commit":"b5883e981ef3b45e6bfd6e2834f251e7be873e30","subject":"Merge \"Replace UPPER_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/b5883e981ef3b45e6bfd6e2834f251e7be873e30"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 19:49:49.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-06-24 17:41:21.000000000","tz":-180},"subject":"Adds spec for extend compute model attributes","message":"Adds spec for extend compute model attributes\n\nThis spec provide more details about the blueprint to\nextend the compute model to include new instance attributes\nand improve watcher awareness about Nova\u0027s contraints\n\nChange-Id: I3dcf4f9d24d326951594db122a2978b40466a081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/c66c8e07f5121367be93092481f7ca7d78205e05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/c66c8e07f5121367be93092481f7ca7d78205e05"}]},"branch":"refs/heads/master"},"0eeb1024372ae99b07d8cf4cf1fc70255cf99310":{"kind":"REWORK","_number":6,"created":"2025-06-25 18:27:25.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/82/947282/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/82/947282/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/6"}}},"commit":{"parents":[{"commit":"b5883e981ef3b45e6bfd6e2834f251e7be873e30","subject":"Merge \"Replace UPPER_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/b5883e981ef3b45e6bfd6e2834f251e7be873e30"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 19:49:49.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-06-25 16:57:32.000000000","tz":-180},"subject":"Adds spec for extend compute model attributes","message":"Adds spec for extend compute model attributes\n\nThis spec provide more details about the blueprint to\nextend the compute model to include new instance attributes\nand improve watcher awareness about Nova\u0027s contraints\n\nChange-Id: I3dcf4f9d24d326951594db122a2978b40466a081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/0eeb1024372ae99b07d8cf4cf1fc70255cf99310"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/0eeb1024372ae99b07d8cf4cf1fc70255cf99310"}]},"branch":"refs/heads/master"},"e1139268bb7218144e6911978c1ab7c78db2d931":{"kind":"REWORK","_number":7,"created":"2025-06-30 18:42:51.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/82/947282/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/82/947282/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/7"}}},"commit":{"parents":[{"commit":"b5883e981ef3b45e6bfd6e2834f251e7be873e30","subject":"Merge \"Replace UPPER_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/b5883e981ef3b45e6bfd6e2834f251e7be873e30"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 19:49:49.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-06-30 18:42:37.000000000","tz":-180},"subject":"Adds spec for extend compute model attributes","message":"Adds spec for extend compute model attributes\n\nThis spec provide more details about the blueprint to\nextend the compute model to include new instance attributes\nand improve watcher awareness about Nova\u0027s contraints\n\nChange-Id: I3dcf4f9d24d326951594db122a2978b40466a081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/e1139268bb7218144e6911978c1ab7c78db2d931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/e1139268bb7218144e6911978c1ab7c78db2d931"}]},"branch":"refs/heads/master"},"1e06f0ae5bc3a9d55e2411e3e5a1da1352b379db":{"kind":"REWORK","_number":8,"created":"2025-06-30 18:56:54.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/82/947282/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/82/947282/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/82/947282/8"}}},"commit":{"parents":[{"commit":"b5883e981ef3b45e6bfd6e2834f251e7be873e30","subject":"Merge \"Replace UPPER_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/b5883e981ef3b45e6bfd6e2834f251e7be873e30"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-04-15 19:49:49.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2025-06-30 18:56:46.000000000","tz":-180},"subject":"Adds spec for extend compute model attributes","message":"Adds spec for extend compute model attributes\n\nThis spec provide more details about the blueprint to\nextend the compute model to include new instance attributes\nand improve watcher awareness about Nova\u0027s contraints\n\nChange-Id: I3dcf4f9d24d326951594db122a2978b40466a081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/1e06f0ae5bc3a9d55e2411e3e5a1da1352b379db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/1e06f0ae5bc3a9d55e2411e3e5a1da1352b379db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
