)]}'
{"id":"openstack%2Fwatcher-specs~952729","triplet_id":"openstack%2Fwatcher-specs~master~I3c807f7e6b0240e3e4988103f0b2c6763d6d1839","project":"openstack/watcher-specs","branch":"master","topic":"bp//host-maintenance-strategy-disable-migration","attention_set":{},"removed_from_attention_set":{"34720":{"account":{"_account_id":34720,"name":"JamesLin","display_name":"jneo8","email":"james.lin@canonical.com","username":"jneo8"},"last_update":"2025-06-17 14:44:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3c807f7e6b0240e3e4988103f0b2c6763d6d1839","subject":"fix: Update the launchpad blueprint link of host-maintenance-strategy-disable-migration spec","status":"MERGED","created":"2025-06-17 01:13:01.000000000","updated":"2025-06-17 14:44:59.000000000","submitted":"2025-06-17 14:44:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952729-bp//host-maintenance-strategy-disable-migration","meta_rev_id":"907e4359f17a9c5d872c73b5b63579cb2aff1141","_number":952729,"virtual_id_number":952729,"owner":{"_account_id":34720,"name":"JamesLin","display_name":"jneo8","email":"james.lin@canonical.com","username":"jneo8"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-17 14:44:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2025-06-17 14:34:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2025-06-17 10:12:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2025-06-17 14:34:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-17 01:21:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-17 10:12:59.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-06-17 14:34:09.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"}],"messages":[{"id":"42d64f6fa5a1416d7a1b5e9332741b73296ca462","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34720,"name":"JamesLin","display_name":"jneo8","email":"james.lin@canonical.com","username":"jneo8"},"date":"2025-06-17 01:13:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7ca2124a21ad662ed3d69d9363797afa8163b17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 01:21:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/290bf6421cda41bfac07e40ae3014839\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3c86a3ab91c4f379d463922d5670705 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff69203319474064a6d2845e48e60594 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1},{"id":"7da2bf309d6b0f80ed8860e05eee833085a0f1ee","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-17 10:12:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5f85357ee4bef2f89e891400a6ab3da7f29a3894","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-17 10:13:34.000000000","message":"Topic fix/wrong-blueprint-launchpad-link-of-disable-cold-or-live-migration-for-host-maintenance-strategy-spec removed","accounts_in_message":[],"_revision_number":1},{"id":"1be9c9cb50f53fe2002ccef8744bb93757f49a7d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-17 10:13:49.000000000","message":"Topic set to bp//host-maintenance-strategy-disable-migration","accounts_in_message":[],"_revision_number":1},{"id":"8464e62435fff2536ca4a5bcdaad2bdce6b8893e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-06-17 10:14:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52be853601c69d80634f2147bef399d45f436109","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-06-17 14:34:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06bc837fd41c82ae1f745e5c051527d32c765677","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 14:34:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eab04e8c78fd1383a3bc3862ecc993a5493450ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 14:44:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e34edd34933642289dcb6c17e25fee70\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60071cd2243146dab3d673cbab5c0d11 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28465dc0a7a941c3b4a722114b2fd282 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":1},{"id":"93f41b264d14bcf8bef1230a8cb965a7f29ed046","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 14:44:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"907e4359f17a9c5d872c73b5b63579cb2aff1141","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 14:44:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8615983735f34d6da4c156c0a2ddaf23\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/e3efa090391f4eb79c22efb63bba5abd : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79cc3c5878d987fddce88c9eaced1cb3a0c490b1","revisions":{"79cc3c5878d987fddce88c9eaced1cb3a0c490b1":{"kind":"REWORK","_number":1,"created":"2025-06-17 01:13:01.000000000","uploader":{"_account_id":34720,"name":"JamesLin","display_name":"jneo8","email":"james.lin@canonical.com","username":"jneo8"},"ref":"refs/changes/29/952729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/29/952729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/29/952729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/29/952729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/29/952729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/29/952729/1"}}},"commit":{"parents":[{"commit":"68eab6d25f5c14e7ab9d89418b4d3e74228b3adf","subject":"feat(spec): disable cold/live migration for host maintenance strategy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/68eab6d25f5c14e7ab9d89418b4d3e74228b3adf"}]}],"author":{"name":"jneo8","email":"james.lin@canonical.com","date":"2025-06-17 01:11:08.000000000","tz":480},"committer":{"name":"jneo8","email":"james.lin@canonical.com","date":"2025-06-17 01:11:41.000000000","tz":480},"subject":"fix: Update the launchpad blueprint link of host-maintenance-strategy-disable-migration spec","message":"fix: Update the launchpad blueprint link of host-maintenance-strategy-disable-migration spec\n\nChange-Id: I3c807f7e6b0240e3e4988103f0b2c6763d6d1839\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/79cc3c5878d987fddce88c9eaced1cb3a0c490b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/79cc3c5878d987fddce88c9eaced1cb3a0c490b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
