)]}'
{"id":"openstack%2Fwatcher-specs~953912","triplet_id":"openstack%2Fwatcher-specs~master~Ia5331649b4b0294bb852906e06cf4661c5f4fa02","project":"openstack/watcher-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2025-07-02 16:54:45.000000000","reason":"Change was abandoned"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-07-02 16:54:45.000000000","reason":"Change was abandoned"},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2025-07-02 16:54:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia5331649b4b0294bb852906e06cf4661c5f4fa02","subject":"Move content to correct sections in Aetos spec","status":"ABANDONED","created":"2025-07-01 22:37:59.000000000","updated":"2025-07-02 16:54:45.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7763769a9d2ae90032955d0553968c31219df87a","_number":953912,"virtual_id_number":953912,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-02 07:51:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-01 22:41:45.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2025-07-01 22:41:45.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-07-01 22:47:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6db01b99c3ad0bcc364505fba90c7c7c1c507343","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-07-01 22:37:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca48646cf97b7194e27d55efa68e678972ea19af","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-07-01 22:41:28.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"10d25d8e6572f9f8ef0be2db3d82bd6d4b10a424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 22:47:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7e318c5195b417685c8be4742b63ccd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e3cee1dd76b4459a581d40e8c84f8ed : FAILURE in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df2b5b5b860e4cc9a5fd706f6b3123f4 : SUCCESS in 2m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e2c6089fa7eeafee18fa355fb36afbee92b7172b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-07-02 07:40:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c6071f0eb4c6af93e624ea00d9cdc2f1d997fff5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 07:51:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/584f793503a7433aa0adb75f348277a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/063374d4dfc3459eb098bdd19172ecf7 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dab735f305f94e33ba8c80249c4895d0 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"7763769a9d2ae90032955d0553968c31219df87a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-02 16:54:45.000000000","message":"Abandoned\n\ni merged this into the base commit as part of resolving the merge conflict","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"94cc8ef11444f37c6ba8a9b3e06071116fa44e52","revisions":{"ee83de1533da5c2fbd2376fc897cf148b9d9a6b0":{"kind":"REWORK","_number":1,"created":"2025-07-01 22:37:59.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/12/953912/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/12/953912/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/12/953912/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/12/953912/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/12/953912/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/12/953912/1"}}},"commit":{"parents":[{"commit":"81513f081e408ac45bdc85e68baa99a526ee8cf4","subject":"Support multitenancy with Prometheus datasource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/81513f081e408ac45bdc85e68baa99a526ee8cf4"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-07-01 22:18:14.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-07-01 22:22:02.000000000","tz":-240},"subject":"Move content to correct sections in Aetos spec","message":"Move content to correct sections in Aetos spec\n\nAs discussed in the previous change, some of the content in the\nspec to add support for Prometheus multitenancy by adding the\nAetos datasource needs to move inte the appropriate sections.\n\nChange-Id: Ia5331649b4b0294bb852906e06cf4661c5f4fa02\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/ee83de1533da5c2fbd2376fc897cf148b9d9a6b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/ee83de1533da5c2fbd2376fc897cf148b9d9a6b0"}]},"branch":"refs/heads/master"},"94cc8ef11444f37c6ba8a9b3e06071116fa44e52":{"kind":"REWORK","_number":2,"created":"2025-07-02 07:40:40.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/12/953912/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-specs","ref":"refs/changes/12/953912/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/12/953912/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/12/953912/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-specs refs/changes/12/953912/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-specs refs/changes/12/953912/2"}}},"commit":{"parents":[{"commit":"81513f081e408ac45bdc85e68baa99a526ee8cf4","subject":"Support multitenancy with Prometheus datasource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/81513f081e408ac45bdc85e68baa99a526ee8cf4"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-07-01 22:18:14.000000000","tz":-240},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2025-07-02 07:24:47.000000000","tz":-240},"subject":"Move content to correct sections in Aetos spec","message":"Move content to correct sections in Aetos spec\n\nAs discussed in the previous change, some of the content in the\nspec to add support for Prometheus multitenancy by adding the\nAetos datasource needs to move inte the appropriate sections.\n\nChange-Id: Ia5331649b4b0294bb852906e06cf4661c5f4fa02\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/94cc8ef11444f37c6ba8a9b3e06071116fa44e52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-specs/commit/94cc8ef11444f37c6ba8a9b3e06071116fa44e52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
