)]}'
{"id":"openstack%2Fwatcher-tempest-plugin~1005381","triplet_id":"openstack%2Fwatcher-tempest-plugin~master~Ide5378ae08c62653d02db0b8b9ada8a143531272","project":"openstack/watcher-tempest-plugin","branch":"master","attention_set":{"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2026-09-12 16:02:06.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-09-18 17:16:41.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-18 17:16:51.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-09-18 22:49:37.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ide5378ae08c62653d02db0b8b9ada8a143531272","subject":"Fix typos","status":"NEW","created":"2026-09-12 12:57:20.000000000","updated":"2026-09-18 22:49:37.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c33de9a44a4567994fefad0b045952d899ea17db","_number":1005381,"virtual_id_number":1005381,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-18 18:51:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2026-09-18 22:49:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}],"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-12 12:57:45.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2026-09-12 14:30:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-12 15:56:55.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-09-12 16:02:06.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2026-09-18 17:16:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"95fa8c19583108a7d0c0a75f4ed18009c48f614c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-12 12:57:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dca5cd342eadef70c84b29242efc406dcd420a22","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-12 12:57:45.000000000","message":"Patch Set 1:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d08a547671213776950ccee369e2569b8fe4b984","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-12 12:58:19.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4d6352c5688ee241f418eb3e7459bcb0dd9369a0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-12 12:58:22.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"acf1655a13fc5fc26c1c5589dd5562e3c8006e8b","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-12 12:58:26.000000000","message":"Patch Set 2:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":2},{"id":"27a478f44b5b1192845b292f3b3f409671d96a83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-12 12:59:12.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f225df4cbfb956979d9c27418bf8e9fac5629c6c","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-12 12:59:20.000000000","message":"Patch Set 3:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":3},{"id":"831d24346b167f5562260a645d53cd94955bd8d8","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-12 13:07:01.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/5b4c62a7094148d09d9cff3d759e7fe9\n\n- teim-code-review https://zuul.teim.app/t/main/build/c5a83c100f914e8398324f05dd43719d : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":3},{"id":"d01146c26338dfda09a79fa9bb17bea1ef5b9b64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 14:30:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/82fe946747ed49deb68e65d878137a3c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d171ab00c80843f58a34080f16332097 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93dbd48f55e44db4b0a21e94e54b4eff : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2b2fd88ea7c444bb627fe8bd20544cd : SUCCESS in 3m 02s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/fc23367fe0d34855ab6e3c2edd3e1933 : SUCCESS in 1h 19m 06s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/c4ab6e53c4794820a0886390ea239099 : SUCCESS in 1h 13m 36s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/b38c978fa87445b3af027cdacda3c844 : FAILURE in 13m 02s\n- watcher-tempest-aetos-2026-1 https://zuul.opendev.org/t/openstack/build/7d1bfff6ee2341b0ac32580b8a3d7f54 : SUCCESS in 57m 47s\n- watcher-tempest-gnocchi-2026-1 https://zuul.opendev.org/t/openstack/build/5725683a9f56498a8fe84e0cd9b3a85d : SUCCESS in 1h 14m 15s\n- watcher-tempest-aetos-2025-2 https://zuul.opendev.org/t/openstack/build/1c798285e3d7431f861f7b8f478f1a4e : SUCCESS in 1h 15m 37s\n- watcher-tempest-gnocchi-2025-2 https://zuul.opendev.org/t/openstack/build/3c3d76bd7e794c57b0190cfad1392f4b : SUCCESS in 1h 25m 31s\n- watcher-tempest-gnocchi-2025-1 https://zuul.opendev.org/t/openstack/build/f6dbff1c53784a7ab588204d16217b50 : SUCCESS in 1h 26m 52s\n- watcher-tempest-prometheus-2025-1 https://zuul.opendev.org/t/openstack/build/4c9c4a30a4d84185a4026e89ee441d2f : SUCCESS in 46m 27s","accounts_in_message":[],"_revision_number":3},{"id":"56904c31c0e72d2a10b3029f350bb78aca0c531e","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-12 15:56:55.000000000","message":"Patch Set 3: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/2e722ffab109471bbf782ef308745f66\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/a7416035e99842ba9daaf6d60b68ffde : SUCCESS in 2h 56m 42s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/404e6b99d4ec4c87b92b56af61e339a4 : SUCCESS in 2h 17m 35s","accounts_in_message":[],"_revision_number":3},{"id":"86847f6a6914069c9cb3b34bc11813c1353b11ba","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-12 16:01:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d97be42148ba6cf1b615e5350b6eb44b101339a9","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-12 16:01:39.000000000","message":"Patch Set 3:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9b3a106bf3f10945cacecb5e75f1179daeaa17ae","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-12 16:08:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/d91d3c3eb7584e46a6ecac10310fcae5\n\n- teim-code-review https://zuul.teim.app/t/main/build/6ed848a0db4448c59681535f0439b089 : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":3},{"id":"ab0e7518dfac11771b8e33847dfcc84d3ae8cea4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 17:49:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a09ceae717e4e8781a85fd13358a3aa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11f2c73ba4574b3f9097453716822108 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66a9fbd9524c47da9e7beddf02075dc5 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aec3e8742e00418c8906baccb463948c : SUCCESS in 2m 02s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/d14ca45f92c4447fb8e17c9af32bc54c : SUCCESS in 1h 40m 59s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/2a39d62659b748f3b913ad25cedd917e : SUCCESS in 1h 30m 44s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/63a16af29f8946108ec033205304b640 : SUCCESS in 1h 32m 08s\n- watcher-tempest-aetos-2026-1 https://zuul.opendev.org/t/openstack/build/0145085389b04fba8c14c1a324e6a53c : SUCCESS in 1h 37m 16s\n- watcher-tempest-gnocchi-2026-1 https://zuul.opendev.org/t/openstack/build/cb797b1c26ea44739f30aa7a466b94e5 : SUCCESS in 1h 25m 07s\n- watcher-tempest-aetos-2025-2 https://zuul.opendev.org/t/openstack/build/b3bb5f3490ea40ac87c4b492e4f302f2 : SUCCESS in 1h 32m 38s\n- watcher-tempest-gnocchi-2025-2 https://zuul.opendev.org/t/openstack/build/dc19c71c17494f67acceb09414edc614 : SUCCESS in 48m 01s\n- watcher-tempest-gnocchi-2025-1 https://zuul.opendev.org/t/openstack/build/9787a0f2249a47d38ddfb8814707a298 : SUCCESS in 57m 28s\n- watcher-tempest-prometheus-2025-1 https://zuul.opendev.org/t/openstack/build/a57345352d58407b92994b40934d7c71 : SUCCESS in 1h 15m 35s","accounts_in_message":[],"_revision_number":3},{"id":"fecd6ad05757f4e8988ff84172dfd6cec29f8be5","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-12 19:05:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/f769a227a9754909b9e538ce5fb5f051\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/8a5c488f200d45d9b5efd2c23a08394a : SUCCESS in 3h 02m 49s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/92fcfdcf207646ffbdd0158781a622c5 : SUCCESS in 2h 23m 24s","accounts_in_message":[],"_revision_number":3},{"id":"7f2a1573eb57f8c2d9818a2b7d5172b37b10fca2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-18 17:16:29.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"16b27f453a190399fb9e5fcd09ab494d996c8728","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-18 17:16:33.000000000","message":"Patch Set 4:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e1a690f7d59ab1ce4749b80c165bdc8aca631f41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-18 17:16:47.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"040eec2a2ea44150ba677ab9ab1f58b7130f7769","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-18 17:16:51.000000000","message":"Patch Set 5:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":5},{"id":"df41ca2bf4901acced5b8c734e622cca4c90e0e5","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-18 17:52:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/453b7bd8c176433c960b9ca5e740f974\n\n- teim-code-review https://zuul.teim.app/t/main/build/993c2fae60b945bda4d3021448e7a916 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":5},{"id":"f0d4942798f0c6090635d99a5d2ba08153ac87c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 18:51:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20dfc98aa58645778c694601d9e460ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a81612518974d9785f130d2c544d925 : SUCCESS in 2m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dca976683d5d4d9783edd93f27a66c33 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18062826739a4be69ff25348af6f8c46 : SUCCESS in 2m 08s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/9a5560f4c760463fb64d63b970c3b154 : SUCCESS in 1h 29m 28s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/217c73c2cc9a488f880f722bef6af6b3 : SUCCESS in 47m 49s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/7234258540a0487caed03a7ccf2e857d : SUCCESS in 56m 59s\n- watcher-tempest-aetos-2026-1 https://zuul.opendev.org/t/openstack/build/a9aedb116c01469ca4697a9082ca5e29 : SUCCESS in 1h 24m 50s\n- watcher-tempest-gnocchi-2026-1 https://zuul.opendev.org/t/openstack/build/8979b9d517144e5b90ea4978d7b8db04 : SUCCESS in 1h 27m 54s\n- watcher-tempest-aetos-2025-2 https://zuul.opendev.org/t/openstack/build/076bbe6864a9438d9e2d420f6b855a45 : SUCCESS in 54m 18s\n- watcher-tempest-gnocchi-2025-2 https://zuul.opendev.org/t/openstack/build/94863ccc01aa4e06afe10eeb04525ceb : SUCCESS in 1h 14m 35s\n- watcher-tempest-gnocchi-2025-1 https://zuul.opendev.org/t/openstack/build/ef0a133e1fa4498bbf4eee6f42570870 : SUCCESS in 1h 11m 28s\n- watcher-tempest-prometheus-2025-1 https://zuul.opendev.org/t/openstack/build/5d307941e4994c1abdd8ccd0db02df56 : SUCCESS in 44m 09s","accounts_in_message":[],"_revision_number":5},{"id":"c33de9a44a4567994fefad0b045952d899ea17db","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-18 22:49:37.000000000","message":"Patch Set 5: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/3e69b5f080314a8ca1f0c02efbb3f431\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/a07cb1dcf1074e9d9e5ca74e8f5d247c : SUCCESS in 2h 59m 43s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/6765e934a7ba446db81c438cbe1004ca : SUCCESS in 2h 15m 58s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6a1b628d748daf7290bca72321f17f38a27ecff7","revisions":{"ee2478e22876fe828323ef265b9350cabf620109":{"kind":"REWORK","_number":1,"created":"2026-09-12 12:57:20.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/81/1005381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-tempest-plugin","ref":"refs/changes/81/1005381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/1"}}},"commit":{"parents":[{"commit":"227584af2ccd27333d71bd5005b698806a54d1c0","subject":"Handle multi-cell deployments in live-migration scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/227584af2ccd27333d71bd5005b698806a54d1c0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-12 12:57:20.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-12 12:57:20.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Ide5378ae08c62653d02db0b8b9ada8a143531272\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/ee2478e22876fe828323ef265b9350cabf620109"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/ee2478e22876fe828323ef265b9350cabf620109"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"ddf2c294b5fc0c8cd93744e135af33aafe1fa74b":{"kind":"REWORK","_number":2,"created":"2026-09-12 12:58:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/81/1005381/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-tempest-plugin","ref":"refs/changes/81/1005381/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/2"}}},"commit":{"parents":[{"commit":"227584af2ccd27333d71bd5005b698806a54d1c0","subject":"Handle multi-cell deployments in live-migration scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/227584af2ccd27333d71bd5005b698806a54d1c0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-12 12:57:20.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-12 12:58:19.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Ide5378ae08c62653d02db0b8b9ada8a143531272\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/ddf2c294b5fc0c8cd93744e135af33aafe1fa74b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/ddf2c294b5fc0c8cd93744e135af33aafe1fa74b"}]},"branch":"refs/heads/master"},"10cb41c9300b88b3e8b5de1fa2d9933d2f965b48":{"kind":"REWORK","_number":3,"created":"2026-09-12 12:59:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/81/1005381/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-tempest-plugin","ref":"refs/changes/81/1005381/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/3"}}},"commit":{"parents":[{"commit":"227584af2ccd27333d71bd5005b698806a54d1c0","subject":"Handle multi-cell deployments in live-migration scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/227584af2ccd27333d71bd5005b698806a54d1c0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-12 12:57:20.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-12 12:59:12.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Ide5378ae08c62653d02db0b8b9ada8a143531272\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/10cb41c9300b88b3e8b5de1fa2d9933d2f965b48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/10cb41c9300b88b3e8b5de1fa2d9933d2f965b48"}]},"branch":"refs/heads/master"},"bcba40108c7d114273c47c9b06bb6eebe37772d7":{"kind":"REWORK","_number":4,"created":"2026-09-18 17:16:29.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/81/1005381/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-tempest-plugin","ref":"refs/changes/81/1005381/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/4"}}},"commit":{"parents":[{"commit":"227584af2ccd27333d71bd5005b698806a54d1c0","subject":"Handle multi-cell deployments in live-migration scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/227584af2ccd27333d71bd5005b698806a54d1c0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-12 12:57:20.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-18 17:16:29.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Ide5378ae08c62653d02db0b8b9ada8a143531272\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/bcba40108c7d114273c47c9b06bb6eebe37772d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/bcba40108c7d114273c47c9b06bb6eebe37772d7"}]},"branch":"refs/heads/master"},"6a1b628d748daf7290bca72321f17f38a27ecff7":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-09-18 17:16:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/81/1005381/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-tempest-plugin","ref":"refs/changes/81/1005381/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/81/1005381/5"}}},"commit":{"parents":[{"commit":"227584af2ccd27333d71bd5005b698806a54d1c0","subject":"Handle multi-cell deployments in live-migration scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/227584af2ccd27333d71bd5005b698806a54d1c0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-12 12:57:20.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-18 17:16:47.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ide5378ae08c62653d02db0b8b9ada8a143531272\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/6a1b628d748daf7290bca72321f17f38a27ecff7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/6a1b628d748daf7290bca72321f17f38a27ecff7"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
