)]}'
{"id":"openstack%2Fwatcher-tempest-plugin~941192","triplet_id":"openstack%2Fwatcher-tempest-plugin~master~I57f434e8d77f3e22bf2b322e75b1c35c315bdc90","project":"openstack/watcher-tempest-plugin","branch":"master","topic":"tox-envdir","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-12 14:52:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I57f434e8d77f3e22bf2b322e75b1c35c315bdc90","subject":"tox: Drop envdir","status":"MERGED","created":"2025-02-11 04:32:40.000000000","updated":"2025-02-12 14:53:26.000000000","submitted":"2025-02-12 14:52:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"941192-tox-envdir","meta_rev_id":"05992fb25c5c16446f44f39e63342745fa0e66ca","_number":941192,"virtual_id_number":941192,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-12 14:52:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2025-02-12 11:20:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2025-02-12 14:20:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2025-02-12 14:20:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-11 05:33:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-12 11:20:42.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2025-02-12 14:20:20.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"}],"messages":[{"id":"fe1c4671c9dc87fe39284d2448d2a932a1cf8ffd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-11 04:32:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"111a66dcb0f5f3ad1aae334b7a9143ff531bfac9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 05:33:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dacf11cd77145f6bfaa6f578ad3dc81\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58c58ce09ffc4574988140f0a4d14219 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea0c74f1e449463a8bf016edae6e6706 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c1b0e91e4484997b0e088673ec99483 : SUCCESS in 2m 22s\n- watcherclient-tempest-functional https://zuul.opendev.org/t/openstack/build/ba4eaa9ec59a4c2999efcd76a5c8d219 : SUCCESS in 43m 26s\n- watcher-tempest-functional https://zuul.opendev.org/t/openstack/build/b63bc9d6d0934445be7e6dca207c4ac0 : SUCCESS in 38m 51s\n- watcher-tempest-actuator https://zuul.opendev.org/t/openstack/build/9695783344a54a82ab8d72b6c7bd6f5d : SUCCESS in 53m 11s\n- watcher-tempest-strategies https://zuul.opendev.org/t/openstack/build/2d4ce5b3cf5142bdaf5d6eb53b741045 : SUCCESS in 52m 46s\n- watcher-tempest-functional-2024-2 https://zuul.opendev.org/t/openstack/build/565cfc09d8364d3d8f5adacbc5dbc8f9 : SUCCESS in 28m 20s\n- watcher-tempest-functional-2024-1 https://zuul.opendev.org/t/openstack/build/90b0e2b96ac6498cbdb665626fe286d4 : SUCCESS in 26m 31s\n- watcher-tempest-functional-2023-2 https://zuul.opendev.org/t/openstack/build/71f4450490764da0b5737d3e63eb4577 : SUCCESS in 25m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a4f92409f7bda44a708f8d4ebd566959f2c85398","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2025-02-12 11:20:42.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ecd1de25d9e95004daeef710d6cd61f1d0e566d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-02-12 14:20:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"48fa180618f9d0a6ac6b2bc7500150c2d558f374","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 14:20:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bfa3c8387d0bac922cc4c7bdcf115fda04f03cb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 14:52:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afaf48f1c1074d3992cd7b2f261c1282\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c3168affd4d421786ecfe72c6ee9292 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5ab732ced96439e8d306e1823e0d4fb : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a86539907e34474289f327341f8fc533 : SUCCESS in 2m 16s\n- watcher-tempest-functional https://zuul.opendev.org/t/openstack/build/490cba90b97949fdbac335bd52fa5f6d : SUCCESS in 29m 37s","accounts_in_message":[],"_revision_number":1},{"id":"0625110d36161cdfdb22e3d5e7092485a29f0e24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 14:52:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"05992fb25c5c16446f44f39e63342745fa0e66ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 14:53:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a06868d56e004c46a30878dc3023c3f9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7690b3225c334f8db0c6210f9a7b5207 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d2e9fb40e6143ce822af836c617ea32 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ffeee483ccfca1285b2208bbe4b756bd85b33ddd","revisions":{"ffeee483ccfca1285b2208bbe4b756bd85b33ddd":{"kind":"REWORK","_number":1,"created":"2025-02-11 04:32:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/941192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher-tempest-plugin","ref":"refs/changes/92/941192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/92/941192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/92/941192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/92/941192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher-tempest-plugin refs/changes/92/941192/1"}}},"commit":{"parents":[{"commit":"ad10537df98b65b0d14ae14d08c29469b123b316","subject":"Merge \"Remove scenario manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/ad10537df98b65b0d14ae14d08c29469b123b316"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-11 04:32:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-11 04:32:32.000000000","tz":540},"subject":"tox: Drop envdir","message":"tox: Drop envdir\n\ntox now always recreates an env although the env is shared using envdir\noptions.\n~~~\n$ tox -e genpolicy\ngenpolicy: recreate env because env type changed from\n{\u0027name\u0027: \u0027genconfig\u0027, \u0027type\u0027: \u0027VirtualEnvRunner\u0027} to\n{\u0027name\u0027: \u0027genpolicy\u0027, \u0027type\u0027: \u0027VirtualEnvRunner\u0027}\n~~~\n\nAccording to the maintainer of tox, this functionality is not intended\nto be supported.\nhttps://github.com/tox-dev/tox/issues/425#issuecomment-1011944293\n\nChange-Id: I57f434e8d77f3e22bf2b322e75b1c35c315bdc90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/ffeee483ccfca1285b2208bbe4b756bd85b33ddd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher-tempest-plugin/commit/ffeee483ccfca1285b2208bbe4b756bd85b33ddd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
