)]}'
{"id":"openstack%2Fwatcher~1000170","triplet_id":"openstack%2Fwatcher~master~I5325c559b09ba2dd7604669aa33b1403f52f1d3d","project":"openstack/watcher","branch":"master","attention_set":{"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2026-08-27 03:01:38.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"28006":{"account":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"last_update":"2026-09-03 01:19:42.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"36630":{"account":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"last_update":"2026-08-27 17:59:39.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-08-27 18:09:48.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I5325c559b09ba2dd7604669aa33b1403f52f1d3d","subject":"Add Workload Optimization goal and Workload Preemption strategy","status":"NEW","created":"2026-08-08 02:52:54.000000000","updated":"2026-09-03 01:19:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":39,"unresolved_comment_count":13,"has_review_started":true,"meta_rev_id":"58cc7a77f85791f63f9123c212af19a6091a1eb7","_number":1000170,"virtual_id_number":1000170,"owner":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-27 20:00:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":-1,"date":"2026-09-03 01:19:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":-1,"date":"2026-08-27 18:09:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"CC":[{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-08 02:53:08.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2026-08-08 03:03:42.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-08-08 04:43:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-17 19:37:26.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2026-09-03 01:19:42.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"034cb1e48a1d48762b230f6f776f226b9a1adae1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"date":"2026-08-08 02:52:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"267dd64006233d9b09d61c6a01819a500406febf","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-08 02:53:08.000000000","message":"Patch Set 1:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d63a6e65887b98d6d71321b6e9cf66653e9d8702","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-08 03:03:42.000000000","message":"Patch Set 1: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/61072b64f5ea488ca1bd91dc651d2b9e\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/d35529e65b324856abc7f432307e9918 : FAILURE in 9m 04s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/08aefade196a4c6dbb82d41c105c6137 : SKIPPED Skipped due to failed job openstack-meta-content-provider-master","accounts_in_message":[],"_revision_number":1},{"id":"a673a7b1d99149263a97286c20042dc6442c7735","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-08 03:17:45.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/7391400b63bd41e79a89603f208f2b44\n\n- teim-code-review https://zuul.teim.app/t/main/build/9c195b8fc9dd4139a7dc9450bb699803 : SUCCESS in 10m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d42ee82f2ba72527c3c46992161b17cfe40e3d52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 04:43:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/05dca51f21ac4661b74cee7f09badab9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5278076b91f24d0396f80cd958020933 : SUCCESS in 3m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24406adb3e9440d38218530919d0ec84 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6eef55d5e2f41ea8d1a7ec455401403 : FAILURE in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/421925328ff14f87acbf63630bfbce97 : SUCCESS in 5m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be0e9f56456d4cd1bf4a995d309a4d6d : SUCCESS in 7m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/34d45178c97c4ad49f799f0df7a43ecb : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90cbc947b01a4b259fee5233631b23ee : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32e49c2445fc40249fecb95c1283a432 : SUCCESS in 4m 01s\n- openstack-tox-py313-eventlet https://zuul.opendev.org/t/openstack/build/10591dafcf8944f69fd6780e0d3d5479 : SUCCESS in 4m 52s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/3be63aaec6bf4673a3a33e00e9b1a790 : SUCCESS in 24m 18s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6f8d02d6a733457d9d0619c5fe09f21d : SUCCESS in 43m 44s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/f2209baf3c5e49d5bd9eb6f08d3892ae : SUCCESS in 30m 49s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/4752fb18f3124b24aedcb32d80442a05 : SUCCESS in 1h 25m 25s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/98b42783dad941b7b692ce509f8a9142 : SUCCESS in 1h 37m 17s\n- watcher-tempest-prometheus-eventlet https://zuul.opendev.org/t/openstack/build/508fbcd0d0134b4db3c03e87211cdc94 : SUCCESS in 1h 19m 21s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/b80116f18ee84539aa39b04d1487f8ed : SUCCESS in 46m 53s","accounts_in_message":[],"_revision_number":1},{"id":"11baf48c81d6adece9de6db6f989c8c0b3f843bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"date":"2026-08-13 15:20:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"57a0e62641cb9cae06e6ca13edff159448491ee1","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-13 15:20:16.000000000","message":"Patch Set 2:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9c0c2022d06658d1dacbae7f1ef90bab8edd5b05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 17:07:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2d8c648d2b54ff7930390f2ae2aa4d9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/321da52efdad4b479fe166e0131a2440 : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0155083fa9334b8bb99a573e94575946 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da78031066d14b738d4d7226100cb510 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fcdb8ec971104db1a82dc4801005dfd0 : SUCCESS in 5m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7973ee3cb73479c854135c6782d2423 : SUCCESS in 5m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d691984ccf184df395baaf89da8da634 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/452898b4d1344660b86428999e0f827e : SUCCESS in 9m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0611b9efef6477eb6732777136d823f : SUCCESS in 4m 39s\n- openstack-tox-py313-eventlet https://zuul.opendev.org/t/openstack/build/83b35233c5184db9be4eb4f91d696c27 : SUCCESS in 4m 06s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/7e2664652cb34a62b16ea76a3dfd339a : SUCCESS in 52m 55s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d494072f81ec4f7b948e458596513706 : SUCCESS in 50m 48s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/9bde1036ea5a47ddac0e1f2f2259a7e0 : SUCCESS in 38m 15s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/154e401dfd4445b1bfe27720932cf948 : SUCCESS in 50m 21s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/08e9b3c6de36482386d38b53cde76eb5 : FAILURE in 1h 30m 51s\n- watcher-tempest-prometheus-eventlet https://zuul.opendev.org/t/openstack/build/6204d489c77c4304935268ce5247cfb1 : SUCCESS in 1h 33m 31s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/e9249a54b4854122b60a295887a2f244 : SUCCESS in 1h 43m 46s","accounts_in_message":[],"_revision_number":2},{"id":"c04b8c19615c6ff6b3393a8b85c5e81b270f434c","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-13 17:25:02.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/6232c4db912147c0bfbfcf73c8222e85\n\n- teim-code-review https://zuul.teim.app/t/main/build/0e94e14f928d48a587127de1a32c5011 : SUCCESS in 13m 18s\n\nWarning:\n  File .kube/cache/http/4cbf3e74a0cf251e76b79c730e27aa78 found in build home directory during playbook github.com/SeanMooney/ci-sean-mooney/playbooks/base/pre.yaml for job teim-code-review.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .kube/cache/http/0a771f2540bee8272a9133c8d531842b found in build home directory during playbook github.com/SeanMooney/ci-sean-mooney/playbooks/base/pre.yaml for job teim-code-review.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .kube/cache/http/829f255153eb3bac5924c1395599a64a found in build home directory during playbook github.com/SeanMooney/ci-sean-mooney/playbooks/base/pre.yaml for job teim-code-review.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .kube/cache/http/f6c297faaa59bf3b224705307298722f found in build home directory during playbook github.com/SeanMooney/ci-sean-mooney/playbooks/base/pre.yaml for job teim-code-review.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .kube/cache/http/5d6ec245f855ced60908a544f340db02 found in build home directory during playbook github.com/SeanMooney/ci-sean-mooney/playbooks/base/pre.yaml for job teim-code-review.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"55848c22af712390db6357d6adfbb80858ba09b3","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-13 19:09:15.000000000","message":"Patch Set 2: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/421b2f80c52a496a986c8144cafeb4a2\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/4cb9649c0080473d8666a2ee64cbc3b1 : FAILURE in 9m 37s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/9da6dbdb27cf4f9baefd65ba9f0dc889 : SKIPPED Skipped due to failed job openstack-meta-content-provider-master","accounts_in_message":[],"_revision_number":2},{"id":"e0db4a39a495201f048d0ee1af2a232eceff7668","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2026-08-17 19:37:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e1ff70dc05b66064f6b803b50ab58561d22458b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"date":"2026-08-26 01:39:25.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ee5ec3b9d0424fbef775bad45eff0ad84cd63d70","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-26 01:39:35.000000000","message":"Patch Set 3:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":3},{"id":"090c8e60b2fe45804a4c0cb5da0fc5bed32bc8c8","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 01:49:59.000000000","message":"Patch Set 3: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/97dcac9e3fed46dfa4f0559ab7d48510\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/c0f5e4971d204312a759421653f9e765 : FAILURE in 8m 43s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/a1fb8f83bda846a1af6adf49316a732c : SKIPPED Skipped due to failed job openstack-meta-content-provider-master\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/f6a020d48aa64d2d84b633710ec2a10f : SUCCESS in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"31ac8266e70ae498abe44a73a2f5266f81981dde","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-26 01:54:06.000000000","message":"Patch Set 3:\n\n(3 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/b441219ea6db4a458014bf1a14c28891\n\n- teim-code-review https://zuul.teim.app/t/main/build/4481ddc6f1b94f1c8158c90fbd8d6aa4 : SUCCESS in 14m 21s","accounts_in_message":[],"_revision_number":3},{"id":"b4a166aca4a9072015a2244718c0e21a4b38b3eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 03:47:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb76f57cac1848fa91dfe6a69b272fb9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d4cecb013ed7426eaf25f6e0b8af1625 : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8be55cf05eb044f2886afa1ea768e436 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa52eed217014048b099a093e14a08db : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f9605b600ca400fb02533a858435554 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ddb4318a44f4cb793bf5b14c7efa48a : SUCCESS in 5m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4fb4647c098a439da9eefbd21ac480fc : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2603d78c5bf4cd6908498d653ef4b0a : SUCCESS in 9m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8dca0e966acc4f4b80ea60b1b4ec4405 : SUCCESS in 4m 19s\n- openstack-tox-py313-eventlet https://zuul.opendev.org/t/openstack/build/7a51854a1fd240d6aabcaaa1433a05d1 : SUCCESS in 5m 02s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/a2fee766c9ae4e58b5ce3e2cc0e66ba5 : SUCCESS in 53m 55s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/758d298e5a3440c186892e33b6cf90f6 : SUCCESS in 59m 17s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/ce1a60ffe3ea45618aba593b3aa3c1e0 : SUCCESS in 37m 33s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/55597b270b42421e8fe5a679e0422f1a : SUCCESS in 1h 53m 03s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/25817cf54e5940b1a9797552033e739f : SUCCESS in 1h 23m 27s\n- watcher-tempest-prometheus-eventlet https://zuul.opendev.org/t/openstack/build/04967169e8b54bdba9ed9d0a75949210 : SUCCESS in 1h 42m 05s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/9e82905894e04821ab60f19fb3f13c78 : SUCCESS in 1h 34m 20s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/61b4e8da6bed4ab7ba708a4b609b1a1d : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":3},{"id":"59e02cee0f730c75ce0ebe71fe884780d4b1c272","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2026-08-26 13:57:30.000000000","message":"Patch Set 3: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"93c6eb4492a201831459996c14afe9188f8cffcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"date":"2026-08-26 21:29:40.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"c6b12c00c4769a54695b10ebf3c1175e31eec769","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-26 21:29:51.000000000","message":"Patch Set 4:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":4},{"id":"37e640592172dd3cc030974f286d144995ffb960","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 21:47:46.000000000","message":"Patch Set 4: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/ca2eca95ba634b8194134cb94955f0ff\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/f9bce7e4a6304dbd92729cafd2c814e8 : FAILURE in 7m 56s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/70ec33c08cfc4caca520ee7851f1f9f8 : SKIPPED Skipped due to failed job openstack-meta-content-provider-master\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/ae45428e1996478bb1affcd28a5cf5d5 : SUCCESS in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"009808ef51113169e0687ef362bee9cba40345f6","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-26 21:51:23.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/00cd5876bd0347f5816f91a727b63f33\n\n- teim-code-review https://zuul.teim.app/t/main/build/213800e10d804749b27c8cec0361fa01 : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":4},{"id":"0bb96d612727b1f958f30d2512deaa9c717f5e6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 23:25:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92b9428738f043f38b9dde82942caf5d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/320529a7a00644e982e02d76062c6777 : SUCCESS in 3m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/482bcf3a64c446579e46ceef3fef14ea : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d55ac3ab2984e4d89fc1c736358acd7 : FAILURE in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65c29722e27c4211b12c7fd644bdd25e : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ac7aa67e04e4f2db45f770ff1c120ea : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2c6f4a53d22a408ca60769b671115b30 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22a75159cd404ed88c549ffa207d20d9 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35d35357854f4e9090685641c8b15731 : SUCCESS in 4m 13s\n- openstack-tox-py313-eventlet https://zuul.opendev.org/t/openstack/build/c5f7cf4bdfec432f8124708377f89fa0 : SUCCESS in 3m 29s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/6679759e755645298a61179f19a49d79 : SUCCESS in 41m 09s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ef7d206bd4d4ac78fa4617eaee9a6e0 : SUCCESS in 54m 01s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/e6be3960c69e4a9c909ba79c27e8f3de : SUCCESS in 18m 28s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/8c6991f3e26d43519f07d79b6e51d4fc : SUCCESS in 1h 33m 59s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/d2bd97c25c2f48d6b04722cb505ebdd2 : SUCCESS in 1h 22m 29s\n- watcher-tempest-prometheus-eventlet https://zuul.opendev.org/t/openstack/build/674fcde7ca894a62a88dedf770cec619 : SUCCESS in 1h 40m 32s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/e083c0bd2dd4421c98cb91b65fe04a80 : SUCCESS in 51m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/48d6c2239081454183bac9a83bd272cc : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":4},{"id":"ad8b2bd6010f8f126906aa8cec550b90d5d4bd6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"date":"2026-08-27 03:01:38.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"5e4b52883a8558bf753ba27755d96c4be01f85af","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-27 03:01:48.000000000","message":"Patch Set 5:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":5},{"id":"70b55a54cbfcc5fdf0bf00f6008ec2a064b760f4","author":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"date":"2026-08-27 03:01:50.000000000","message":"Patch Set 5:\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"107c78e76840e51fd031b5ccc2c3364c7a966e9f","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 03:11:26.000000000","message":"Patch Set 5: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/a183f906eeeb4b9a8f8aebba78d1a746\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/07e5bfd3efc34a0faa31c50a40353db9 : FAILURE in 8m 02s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/cd6d899835a94d978210adde3396a08d : SKIPPED Skipped due to failed job openstack-meta-content-provider-master\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/5c16c20a14864707bc2339aed9e08a9d : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7d881a3d51fda27fba6ccc657522605e09d005c4","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-27 03:16:42.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/f413564f63ab4c819d016fb376739c60\n\n- teim-code-review https://zuul.teim.app/t/main/build/8ec345b0fd8e425f94d9254849650c80 : SUCCESS in 14m 45s","accounts_in_message":[],"_revision_number":5},{"id":"e4f2a539fbe6b9b1c73726ad202793cb9f750978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 04:58:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/98b1ff35eae241e7ae2b338a7b3ef720\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0d4b59e43356411cabd722f17e38bf67 : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e2da48b34ab4637b320973349015cb3 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae64bf5ec0e844f5900df1ecd779a27a : FAILURE in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/473dd3a69b16439b95fdf0c3869bfa67 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/590917879ccd48b7b27d00400e749ae8 : SUCCESS in 6m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c14cedd2f4884a9e839006287a112f51 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/762104d28f7049c4a7785568d08bae78 : SUCCESS in 9m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84ed6e96f933431d95618a42efdd4caa : SUCCESS in 7m 06s\n- openstack-tox-py313-eventlet https://zuul.opendev.org/t/openstack/build/48748fc786444bb29c3140ff7499fdd6 : SUCCESS in 3m 54s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/ba5141099e3d4b7d8ad107af915735cb : SUCCESS in 53m 27s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/23f3d3911b3f47099da80aa071b459e2 : SUCCESS in 1h 00m 09s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/5cc8d0937945490190c18a1cb1b2f6c8 : SUCCESS in 39m 16s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/6de5ef3759204e89b92cb2aef552a657 : SUCCESS in 1h 32m 28s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/37faa6a2c9b14f659d3f8c1153029361 : SUCCESS in 1h 41m 13s\n- watcher-tempest-prometheus-eventlet https://zuul.opendev.org/t/openstack/build/c096b29aea2f4f69ac453f69a52331d9 : SUCCESS in 1h 32m 36s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/397a04a4a91547df9ac62b662cff7bbe : SUCCESS in 1h 20m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8ab61bcb4f7446b191f734a3919dac30 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":5},{"id":"5ebd0983fcbf272e93eafefe5e1ab86c15cbed7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"date":"2026-08-27 14:47:53.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"9b0127006526d04b67da6d9e3ecb635720ce5d24","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-27 14:48:03.000000000","message":"Patch Set 6:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8f12878126af80c15adb89657c50f33e15dfd161","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 14:57:30.000000000","message":"Patch Set 6: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/f6e8fec0e69548ebb27ef0241524a4da\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/8847a14ef999468091d561c79036a466 : FAILURE in 8m 25s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/79181978dea34d25b842db907008ed9b : SKIPPED Skipped due to failed job openstack-meta-content-provider-master\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/47f2975ec667426ea4dbe8e54b4d1a03 : SUCCESS in 5m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"89c6bc2ccf1d9bd0ae5f5908dec3f160a70a83e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"date":"2026-08-27 15:07:37.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"196293f40643cb6bd071ff2632e3d4055c7a08af","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-27 15:07:56.000000000","message":"Patch Set 7:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":7},{"id":"5b2283103d23d2dcf13d4e8b366327f53db39b8a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 15:17:28.000000000","message":"Patch Set 7: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/5c86cbed2a6a41709ab538597538f1f9\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/135074965efc4b29af12e9cdf6282257 : FAILURE in 8m 28s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/99d44b2eb39c46e39eae7acd249c0ae0 : SKIPPED Skipped due to failed job openstack-meta-content-provider-master\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/0f00999505ed4c248ebc2efc5f4f3ee4 : SUCCESS in 5m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f7d7e3f9f895b9486d5d134275580511e82ab502","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-27 15:38:41.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/48bae9fe289747aea061cb52ced88289\n\n- teim-code-review https://zuul.teim.app/t/main/build/bf555befaea04394ab94fb9daedaf6fa : SUCCESS in 10m 06s","accounts_in_message":[],"_revision_number":7},{"id":"6725fc3ca109cb140531e103f83a844db753092f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 17:01:36.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fa1bae2188a40078399a9cb98d73123\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/95d7dd3766b14c4db38647b646306a0d : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8f675b3e34a466da67d0885c44fff04 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/002959a8ea52467a9933edd63291f3d8 : FAILURE in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ed529ca8d9064ab88664502dd1d6288a : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/021443d8d9d9458a8a7e3d9ecb0b90f8 : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/84048ba0d7a54175a53e3f8cda9c12f9 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c166cd7a41f41e883cb397118ba155b : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1edcb4c5459e45639d8e6647da0f15c1 : SUCCESS in 3m 46s\n- openstack-tox-py313-eventlet https://zuul.opendev.org/t/openstack/build/c1b1283f50574bec8a715393eca4b5a8 : SUCCESS in 3m 41s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/365fb13031a347768a9d443282a8820d : SUCCESS in 1h 01m 43s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d581a2c256a469a8f151a22d5c21f88 : SUCCESS in 41m 52s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/648b4859c7f24bfca33f46b6244cfc4e : SUCCESS in 33m 35s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/0f9340f547cc4759ad7282fb39d3f614 : SUCCESS in 1h 40m 24s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/cde051ce12564e4d9ea74fb4fa02008f : SUCCESS in 1h 24m 28s\n- watcher-tempest-prometheus-eventlet https://zuul.opendev.org/t/openstack/build/4c1c659ad0474898818716e3af987ecc : SUCCESS in 47m 41s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/9f578799a2784c70b32c9ad4449ee359 : SUCCESS in 1h 37m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4d33b0f9a7c54fed8dfeab342beab191 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":7},{"id":"c7c9ca26c206a97f9fd8308beaacf8b77593012e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"date":"2026-08-27 17:59:26.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"8926d354be4ba25cffd21cd929af40609e394852","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-27 17:59:39.000000000","message":"Patch Set 8:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":8},{"id":"3ca42d298d585a35afc77bddd856c4d75a2e186c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 18:09:48.000000000","message":"Patch Set 8: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/4141dc1cab35494a975672f65cddb132\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/fbe4f927780149d780791cb522948ced : FAILURE in 8m 32s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/3779b32cd46b4d069aea6bf3f687c792 : SKIPPED Skipped due to failed job openstack-meta-content-provider-master\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/aff73d48b6cb4141b52bcc6f490de693 : SUCCESS in 4m 55s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2fbf51a21c3a819b8814a2afac3aa262249a8df4","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-08-27 18:13:07.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/f7a78de90e9e42418a39d7bf4024b2e7\n\n- teim-code-review https://zuul.teim.app/t/main/build/83e5e593d5f94cd49b3d43258eb6dc04 : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":8},{"id":"1569798b09056067349465810e937034eefe28d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 20:00:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5e33ca8c4474ed68a669410b5a52938\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dc654c3af7604de0a730a75a0924ffd2 : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15287fadfbf64c728beb394f04a9895a : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6f120f0bc034fee884e100f7a6333ca : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1df31056cd9f48e2b62ce44244355f83 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/087863d231604e02af00f090868e5017 : SUCCESS in 5m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f193303f7c53462ab34e2bc18c6c115a : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/331c32dad1a2402b8227049a60d9273a : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49a243e749394560a84abe3f46e74295 : SUCCESS in 5m 09s\n- openstack-tox-py313-eventlet https://zuul.opendev.org/t/openstack/build/e24093e16dee41af80b6ce9d2cf7770d : SUCCESS in 5m 20s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/1bede78b96a34a8f8e3914db2ece779c : SUCCESS in 54m 17s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/809f1e189c6c4d46a1a03d0f29e52759 : SUCCESS in 27m 05s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/2d097682bc214aeab4f61d4b4d51bb7c : SUCCESS in 40m 15s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/6d26a49011f440ad8e2c75c55890fe43 : SUCCESS in 1h 34m 08s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/04c8f54e5e42472d84a08ce678497e86 : SUCCESS in 1h 33m 20s\n- watcher-tempest-prometheus-eventlet https://zuul.opendev.org/t/openstack/build/e64522bbf2264b0e9dc9a480a1c8db26 : SUCCESS in 1h 31m 36s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/81ce60581e3a4d9795701944dccde2c0 : SUCCESS in 1h 43m 45s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/774dc4403c154b48836ef309f155ce81 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":8},{"id":"58cc7a77f85791f63f9123c212af19a6091a1eb7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-03 01:19:42.000000000","message":"Patch Set 8: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"6c3f4f93c2e97eca16fe1ea621e7ff6e3a9dfd15","revisions":{"ca7aaf96ba531eb6543d38ed30eb28c9505f6035":{"kind":"REWORK","_number":1,"created":"2026-08-08 02:52:54.000000000","uploader":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"ref":"refs/changes/70/1000170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/70/1000170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/70/1000170/1"}}},"commit":{"parents":[{"commit":"19e9f76aa18d194999cdc433f7ce5b33d3e17081","subject":"Add new shelve action","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/19e9f76aa18d194999cdc433f7ce5b33d3e17081"}]}],"author":{"name":"Winicius Allan","email":"wdasilva@redhat.com","date":"2026-08-08 02:21:10.000000000","tz":-180},"committer":{"name":"Winicius Allan","email":"wdasilva@redhat.com","date":"2026-08-08 02:52:33.000000000","tz":-180},"subject":"Add Workload Optimization goal and Workload Preemption strategy","message":"Add Workload Optimization goal and Workload Preemption strategy\n\nAdd the Workload Optimization goal and Workload Preemption strategy for\npreemptible instances management. This implements the core feature\ndescribed in the preemptible instances specification.\n\nThe strategy preempts (deletes or shelves) low-priority instances marked\nas preemptable via their flavor extra specs to free allocated resources\nand ensure a target percentage of free capacity is available for\non-demand workloads.\n\nAssisted-By: claude-opus-4.6\nImplements: blueprint-preemptible-instances\nChange-Id: I5325c559b09ba2dd7604669aa33b1403f52f1d3d\nSigned-off-by: Winicius Allan \u003cwdasilva@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/ca7aaf96ba531eb6543d38ed30eb28c9505f6035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/ca7aaf96ba531eb6543d38ed30eb28c9505f6035"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"19e9f76aa18d194999cdc433f7ce5b33d3e17081","is_merged_in_target_branch":false,"change_id":"Ie5bfa9532c1befd0f8236f31c766d0148ff8d070","change_number":1000169,"patch_set_number":1,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"4a1f1a6a839ebafd3e40db49309ec960aead5bf0":{"kind":"REWORK","_number":2,"created":"2026-08-13 15:20:04.000000000","uploader":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"ref":"refs/changes/70/1000170/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/70/1000170/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/70/1000170/2"}}},"commit":{"parents":[{"commit":"eb07a5bf09adcf55f3719c59559a643ad4f1fba7","subject":"Add new delete and shelve actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/eb07a5bf09adcf55f3719c59559a643ad4f1fba7"}]}],"author":{"name":"Winicius Allan","email":"wdasilva@redhat.com","date":"2026-08-08 02:21:10.000000000","tz":-180},"committer":{"name":"Winicius Allan","email":"wdasilva@redhat.com","date":"2026-08-13 15:19:40.000000000","tz":-180},"subject":"Add Workload Optimization goal and Workload Preemption strategy","message":"Add Workload Optimization goal and Workload Preemption strategy\n\nAdd the Workload Optimization goal and Workload Preemption strategy for\npreemptible instances management. This implements the core feature\ndescribed in the preemptible instances specification.\n\nThe strategy preempts (deletes or shelves) low-priority instances marked\nas preemptable via their flavor extra specs to free allocated resources\nand ensure a target percentage of free capacity is available for\non-demand workloads.\n\nAssisted-By: claude-opus-4.6\nImplements: blueprint-preemptible-instances\nChange-Id: I5325c559b09ba2dd7604669aa33b1403f52f1d3d\nSigned-off-by: Winicius Allan \u003cwdasilva@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/4a1f1a6a839ebafd3e40db49309ec960aead5bf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/4a1f1a6a839ebafd3e40db49309ec960aead5bf0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eb07a5bf09adcf55f3719c59559a643ad4f1fba7","is_merged_in_target_branch":false,"change_id":"I858767982822162cffa56ad76bce0c32206d30a8","change_number":1000168,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"231988760117ee3f73284839b002ea25955c7a00":{"kind":"REWORK","_number":3,"created":"2026-08-26 01:39:25.000000000","uploader":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"ref":"refs/changes/70/1000170/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/70/1000170/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/70/1000170/3"}}},"commit":{"parents":[{"commit":"eb07a5bf09adcf55f3719c59559a643ad4f1fba7","subject":"Add new delete and shelve actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/eb07a5bf09adcf55f3719c59559a643ad4f1fba7"}]}],"author":{"name":"Winicius Allan","email":"wdasilva@redhat.com","date":"2026-08-08 02:21:10.000000000","tz":-180},"committer":{"name":"Winicius Allan","email":"wdasilva@redhat.com","date":"2026-08-26 01:02:24.000000000","tz":-180},"subject":"Add Workload Optimization goal and Workload Preemption strategy","message":"Add Workload Optimization goal and Workload Preemption strategy\n\nAdd the Workload Optimization goal and Workload Preemption strategy for\npreemptible instances management. This implements the core feature\ndescribed in the preemptible instances specification.\n\nThe strategy preempts (deletes or shelves) low-priority instances marked\nas preemptable via their flavor extra specs to free allocated resources\nand ensure a target percentage of free capacity is available for\non-demand workloads.\n\nAssisted-By: claude-opus-4.6\nImplements: blueprint-preemptible-instances\nChange-Id: I5325c559b09ba2dd7604669aa33b1403f52f1d3d\nSigned-off-by: Winicius Allan \u003cwdasilva@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/231988760117ee3f73284839b002ea25955c7a00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/231988760117ee3f73284839b002ea25955c7a00"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eb07a5bf09adcf55f3719c59559a643ad4f1fba7","is_merged_in_target_branch":false,"change_id":"I858767982822162cffa56ad76bce0c32206d30a8","change_number":1000168,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"19d3332769497a2a02e63e4afa2c47fd77b50474":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-08-26 21:29:40.000000000","uploader":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"ref":"refs/changes/70/1000170/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/70/1000170/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/70/1000170/4"}}},"commit":{"parents":[{"commit":"4c1c54f2b9fb0e2a01a9a905c0b7fa3d8de9a686","subject":"Add new delete and shelve actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/4c1c54f2b9fb0e2a01a9a905c0b7fa3d8de9a686"}]}],"author":{"name":"Winicius Allan","email":"wdasilva@redhat.com","date":"2026-08-08 02:21:10.000000000","tz":-180},"committer":{"name":"Winicius Allan","email":"wdasilva@redhat.com","date":"2026-08-26 21:29:31.000000000","tz":-180},"subject":"Add Workload Optimization goal and Workload Preemption strategy","message":"Add Workload Optimization goal and Workload Preemption strategy\n\nAdd the Workload Optimization goal and Workload Preemption strategy for\npreemptible instances management. This implements the core feature\ndescribed in the preemptible instances specification.\n\nThe strategy preempts (deletes or shelves) low-priority instances marked\nas preemptable via their flavor extra specs to free allocated resources\nand ensure a target percentage of free capacity is available for\non-demand workloads.\n\nAssisted-By: claude-opus-4.6\nImplements: blueprint-preemptible-instances\nChange-Id: I5325c559b09ba2dd7604669aa33b1403f52f1d3d\nSigned-off-by: Winicius Allan \u003cwdasilva@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/19d3332769497a2a02e63e4afa2c47fd77b50474"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/19d3332769497a2a02e63e4afa2c47fd77b50474"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4c1c54f2b9fb0e2a01a9a905c0b7fa3d8de9a686","is_merged_in_target_branch":false,"change_id":"I858767982822162cffa56ad76bce0c32206d30a8","change_number":1000168,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8c713b785430238121b232a83978e10ebb614a05":{"kind":"REWORK","_number":5,"created":"2026-08-27 03:01:38.000000000","uploader":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"ref":"refs/changes/70/1000170/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/70/1000170/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/70/1000170/5"}}},"commit":{"parents":[{"commit":"4c1c54f2b9fb0e2a01a9a905c0b7fa3d8de9a686","subject":"Add new delete and shelve actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/4c1c54f2b9fb0e2a01a9a905c0b7fa3d8de9a686"}]}],"author":{"name":"Winicius Allan","email":"wdasilva@redhat.com","date":"2026-08-08 02:21:10.000000000","tz":-180},"committer":{"name":"Winicius Allan","email":"wdasilva@redhat.com","date":"2026-08-27 03:01:29.000000000","tz":-180},"subject":"Add Workload Optimization goal and Workload Preemption strategy","message":"Add Workload Optimization goal and Workload Preemption strategy\n\nAdd the Workload Optimization goal and Workload Preemption strategy for\npreemptible instances management. This implements the core feature\ndescribed in the preemptible instances specification.\n\nThe strategy preempts (deletes or shelves) low-priority instances marked\nas preemptable via their flavor extra specs to free allocated resources\nand ensure a target percentage of free capacity is available for\non-demand workloads.\n\nAssisted-By: claude-opus-4.6\nImplements: blueprint-preemptible-instances\nChange-Id: I5325c559b09ba2dd7604669aa33b1403f52f1d3d\nSigned-off-by: Winicius Allan \u003cwdasilva@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/8c713b785430238121b232a83978e10ebb614a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/8c713b785430238121b232a83978e10ebb614a05"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4c1c54f2b9fb0e2a01a9a905c0b7fa3d8de9a686","is_merged_in_target_branch":false,"change_id":"I858767982822162cffa56ad76bce0c32206d30a8","change_number":1000168,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"4d56a353e46f482ad51fd2e4069d6fb9f9ecfb9d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-08-27 14:47:53.000000000","uploader":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"ref":"refs/changes/70/1000170/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/70/1000170/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/70/1000170/6"}}},"commit":{"parents":[{"commit":"cbd7d7fe885cede6ed5f2eb0796e33bea8e5b89f","subject":"Add new delete and shelve actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/cbd7d7fe885cede6ed5f2eb0796e33bea8e5b89f"}]}],"author":{"name":"Winicius Allan","email":"wdasilva@redhat.com","date":"2026-08-08 02:21:10.000000000","tz":-180},"committer":{"name":"Winicius Allan","email":"wdasilva@redhat.com","date":"2026-08-27 14:32:18.000000000","tz":-180},"subject":"Add Workload Optimization goal and Workload Preemption strategy","message":"Add Workload Optimization goal and Workload Preemption strategy\n\nAdd the Workload Optimization goal and Workload Preemption strategy for\npreemptible instances management. This implements the core feature\ndescribed in the preemptible instances specification.\n\nThe strategy preempts (deletes or shelves) low-priority instances marked\nas preemptable via their flavor extra specs to free allocated resources\nand ensure a target percentage of free capacity is available for\non-demand workloads.\n\nAssisted-By: claude-opus-4.6\nImplements: blueprint-preemptible-instances\nChange-Id: I5325c559b09ba2dd7604669aa33b1403f52f1d3d\nSigned-off-by: Winicius Allan \u003cwdasilva@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/4d56a353e46f482ad51fd2e4069d6fb9f9ecfb9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/4d56a353e46f482ad51fd2e4069d6fb9f9ecfb9d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cbd7d7fe885cede6ed5f2eb0796e33bea8e5b89f","is_merged_in_target_branch":false,"change_id":"I858767982822162cffa56ad76bce0c32206d30a8","change_number":1000168,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"1ffbe7a54b8b2a1393a3173f0762efa819bf1b01":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-08-27 15:07:37.000000000","uploader":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"ref":"refs/changes/70/1000170/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/70/1000170/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/70/1000170/7"}}},"commit":{"parents":[{"commit":"a22a4b7c1b696900f6196a0f872131b1f4f378bf","subject":"Add new delete and shelve actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/a22a4b7c1b696900f6196a0f872131b1f4f378bf"}]}],"author":{"name":"Winicius Allan","email":"wdasilva@redhat.com","date":"2026-08-08 02:21:10.000000000","tz":-180},"committer":{"name":"Winicius Allan","email":"wdasilva@redhat.com","date":"2026-08-27 15:07:26.000000000","tz":-180},"subject":"Add Workload Optimization goal and Workload Preemption strategy","message":"Add Workload Optimization goal and Workload Preemption strategy\n\nAdd the Workload Optimization goal and Workload Preemption strategy for\npreemptible instances management. This implements the core feature\ndescribed in the preemptible instances specification.\n\nThe strategy preempts (deletes or shelves) low-priority instances marked\nas preemptable via their flavor extra specs to free allocated resources\nand ensure a target percentage of free capacity is available for\non-demand workloads.\n\nAssisted-By: claude-opus-4.6\nImplements: blueprint-preemptible-instances\nChange-Id: I5325c559b09ba2dd7604669aa33b1403f52f1d3d\nSigned-off-by: Winicius Allan \u003cwdasilva@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/1ffbe7a54b8b2a1393a3173f0762efa819bf1b01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/1ffbe7a54b8b2a1393a3173f0762efa819bf1b01"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a22a4b7c1b696900f6196a0f872131b1f4f378bf","is_merged_in_target_branch":false,"change_id":"I858767982822162cffa56ad76bce0c32206d30a8","change_number":1000168,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"6c3f4f93c2e97eca16fe1ea621e7ff6e3a9dfd15":{"kind":"REWORK","_number":8,"created":"2026-08-27 17:59:26.000000000","uploader":{"_account_id":36630,"name":"Winicius Allan Bezerra da Silva","display_name":"Winicius Allan","email":"winiciusab12@gmail.com","username":"winiciusallan"},"ref":"refs/changes/70/1000170/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/70/1000170/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/70/1000170/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/70/1000170/8"}}},"commit":{"parents":[{"commit":"a22a4b7c1b696900f6196a0f872131b1f4f378bf","subject":"Add new delete and shelve actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/a22a4b7c1b696900f6196a0f872131b1f4f378bf"}]}],"author":{"name":"Winicius Allan","email":"wdasilva@redhat.com","date":"2026-08-08 02:21:10.000000000","tz":-180},"committer":{"name":"Winicius Allan","email":"wdasilva@redhat.com","date":"2026-08-27 17:59:21.000000000","tz":-180},"subject":"Add Workload Optimization goal and Workload Preemption strategy","message":"Add Workload Optimization goal and Workload Preemption strategy\n\nAdd the Workload Optimization goal and Workload Preemption strategy for\npreemptible instances management. This implements the core feature\ndescribed in the preemptible instances specification.\n\nThe strategy preempts (deletes or shelves) low-priority instances marked\nas preemptable via their flavor extra specs to free allocated resources\nand ensure a target percentage of free capacity is available for\non-demand workloads.\n\nAssisted-By: claude-opus-4.6\nImplements: blueprint-preemptible-instances\nChange-Id: I5325c559b09ba2dd7604669aa33b1403f52f1d3d\nSigned-off-by: Winicius Allan \u003cwdasilva@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/6c3f4f93c2e97eca16fe1ea621e7ff6e3a9dfd15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/6c3f4f93c2e97eca16fe1ea621e7ff6e3a9dfd15"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a22a4b7c1b696900f6196a0f872131b1f4f378bf","is_merged_in_target_branch":false,"change_id":"I858767982822162cffa56ad76bce0c32206d30a8","change_number":1000168,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
