)]}'
{"id":"openstack%2Fwatcher~1006113","triplet_id":"openstack%2Fwatcher~master~Ib8fc8d9ca297238701ee8c1e7295ac46f1e8115f","project":"openstack/watcher","branch":"master","topic":"indri-removals","attention_set":{"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2026-09-17 20:35:55.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib8fc8d9ca297238701ee8c1e7295ac46f1e8115f","subject":"Remove the deprecated WSGI scripts","status":"NEW","created":"2026-09-17 20:35:46.000000000","updated":"2026-09-18 12:00:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9a01ad2959b1eb540105ad76b65952f2f51033fb","_number":1006113,"virtual_id_number":1006113,"owner":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-17 22:42:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"disliked":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":2,"date":"2026-09-18 12:00:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":-1,"date":"2026-09-17 21:06:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}],"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-17 20:35:55.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2026-09-17 20:45:04.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-09-17 22:42:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-18 08:44:12.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"CC"},{"updated":"2026-09-18 12:00:47.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"cda5ad10b85f706ed909906a537e092ca726eb73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2026-09-17 20:35:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"739016527bd57430ac137712bb5676477c618e53","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-17 20:35:55.000000000","message":"Patch Set 1:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d435d41939a195d3049edb545cead55933d88d51","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-17 20:45:04.000000000","message":"Patch Set 1: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/e659c090567f4b70b77d64c4151dae68\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/206e5da9e5af42709e9821bb3732a67b : FAILURE in 8m 09s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/47df88a5091544d68c33b49fbfb8723c : SKIPPED Skipped due to failed job openstack-meta-content-provider-master\n- watcher-s2i-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/afd7dfb0705f4fd9bbd815e45d48ee9b : FAILURE in 2m 11s (non-voting)\n- watcher-s2i-tempest https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/95e9b6555e024bc8baae7c7928d42c53 : SKIPPED Skipped due to failed job watcher-s2i-content-provider (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"effa2824ed819fe0ae5439e1fbeba6088c41313f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2026-09-17 20:52:11.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"8a2f0aa2798c8af1d99628b330c2a43c652ebce7","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-17 20:52:19.000000000","message":"Patch Set 2:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":2},{"id":"81786fa22094482a212f77bfdea1f2a0324f1842","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-17 21:06:57.000000000","message":"Patch Set 2: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/c78ea120fc0e4fa3b44b51ef8b74d494\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/1f9778b1ad2e40f8a8ce0e02acb4648f : FAILURE in 13m 38s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/f1eef1986eb844f59c2377287b368e64 : SKIPPED Skipped due to failed job openstack-meta-content-provider-master\n- watcher-s2i-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/03596a57d9da473991567ca012c78353 : FAILURE in 2m 23s (non-voting)\n- watcher-s2i-tempest https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/c6ae42c7577540df816afcb0b7b1d212 : SKIPPED Skipped due to failed job watcher-s2i-content-provider (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad222afe54140e5479da10144085097301975c6c","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-17 21:13:12.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/3d43696f1ac24ddf89d11a3014906e4e\n\n- teim-code-review https://zuul.teim.app/t/main/build/f1b44ec01c5e49bd88c0a6d0b123e860 : SUCCESS in 9m 22s","accounts_in_message":[],"_revision_number":2},{"id":"bdd64a5cca5f6330aeba14ff28d6bc4cd3c89f06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 22:42:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91c020f9a12b4cc6b7895b4e8a152d51\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/72312889de96415991b6933bdb6a8264 : SUCCESS in 2m 30s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dfb1d1b4a11247df9174bfc5e9b3961b : SUCCESS in 4m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cb33c35797b9468d9327bdd0f9504bd5 : SUCCESS in 2m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a451ff0659043f39b5cb8180da0e42b : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d18263b6d014675ba61011b27c728a3 : SUCCESS in 2m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c2ec410f472496fa96812caeefe6bae : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/03045579560940b2b6c3855419dad775 : SUCCESS in 5m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/53f2fd1c0f59409e859b6968243ed312 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b97a5c94f8a04c77ad3295f7cd37c0b7 : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f859632b6274331ab2386ec30e9e271 : SUCCESS in 5m 10s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/520c9749a2d545afbe71134adbacf24e : SUCCESS in 44m 57s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d77174c4850d4ac5a303057331ef7cb6 : SUCCESS in 53m 05s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/c776353e563e4a88b26ce80ec790458b : SUCCESS in 17m 06s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/8b83af197d0d4ae1be54476ad73c4afc : SUCCESS in 1h 21m 53s\n- python-watcherclient-functional https://zuul.opendev.org/t/openstack/build/0ac770367a0f4975949c999cd3aace54 : SUCCESS in 38m 42s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/541885fb0e97478f8ae7610005222bb4 : SUCCESS in 1h 38m 00s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/5565cce02038428fb2071016b401145b : SUCCESS in 1h 25m 59s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1c4ee265b91e4746b9394223b94e37c0 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":2},{"id":"6e5b353f63ca2b3cad9535e055f938a558f8541a","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2026-09-18 08:44:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9a01ad2959b1eb540105ad76b65952f2f51033fb","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-18 12:00:47.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cf5bb467123aaaa5756ec776eb66b3803ebe00a6","revisions":{"bcc9352e7661be9459e379cc0b3b50e9b4e4ee4c":{"kind":"REWORK","_number":1,"created":"2026-09-17 20:35:46.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/13/1006113/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/13/1006113/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/13/1006113/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/13/1006113/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/13/1006113/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/13/1006113/1"}}},"commit":{"parents":[{"commit":"b07eb0cb7367d3c1b43d286218c9581b22a9dd49","subject":"Remove eventlet support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/b07eb0cb7367d3c1b43d286218c9581b22a9dd49"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2026-09-17 20:28:44.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2026-09-17 20:33:02.000000000","tz":-180},"subject":"Remove the deprecated WSGI scripts","message":"Remove the deprecated WSGI scripts\n\nThe watcher/api/app.wsgi script has been deprecated since watcher-api-wsgi\nwas introduced, and watcher-api-wsgi is itself generated by the pbr\nwsgi_scripts extension, which is deprecated and planned for removal from\npbr. Both are removed here, leaving the watcher.wsgi.api Python module as\nthe only way to load the API application.\n\nDeployment tooling should reference watcher.wsgi.api:application, as\nDevStack already does through uwsgi. Servers that require a file on disk,\nsuch as Apache mod_wsgi, can be pointed at the installed\nwatcher/wsgi/api.py module.\n\nAssisted-By: Claude Code (Opus 5)\nChange-Id: Ib8fc8d9ca297238701ee8c1e7295ac46f1e8115f\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/bcc9352e7661be9459e379cc0b3b50e9b4e4ee4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/bcc9352e7661be9459e379cc0b3b50e9b4e4ee4c"}]},"branch":"refs/heads/master"},"cf5bb467123aaaa5756ec776eb66b3803ebe00a6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-17 20:52:11.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/13/1006113/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/13/1006113/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/13/1006113/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/13/1006113/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/13/1006113/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/13/1006113/2"}}},"commit":{"parents":[{"commit":"4ab3720c750adb73b6dd197504da3707d040ba4b","subject":"Remove eventlet support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/4ab3720c750adb73b6dd197504da3707d040ba4b"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2026-09-17 20:28:44.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2026-09-17 20:51:12.000000000","tz":-180},"subject":"Remove the deprecated WSGI scripts","message":"Remove the deprecated WSGI scripts\n\nThe watcher/api/app.wsgi script has been deprecated since watcher-api-wsgi\nwas introduced, and watcher-api-wsgi is itself generated by the pbr\nwsgi_scripts extension, which is deprecated and planned for removal from\npbr. Both are removed here, leaving the watcher.wsgi.api Python module as\nthe only way to load the API application.\n\nDeployment tooling should reference watcher.wsgi.api:application, as\nDevStack already does through uwsgi. Servers that require a file on disk,\nsuch as Apache mod_wsgi, can be pointed at the installed\nwatcher/wsgi/api.py module.\n\nAssisted-By: Claude Code (Opus 5)\nChange-Id: Ib8fc8d9ca297238701ee8c1e7295ac46f1e8115f\nSigned-off-by: Douglas Viroel \u003cviroel@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/cf5bb467123aaaa5756ec776eb66b3803ebe00a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/cf5bb467123aaaa5756ec776eb66b3803ebe00a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
