)]}'
{"id":"openstack%2Fwatcher~1007420","triplet_id":"openstack%2Fwatcher~master~I6c254a8a826aa8c6d12922f6c0cd239d6b9b92db","project":"openstack/watcher","branch":"master","attention_set":{},"removed_from_attention_set":{"34452":{"account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"last_update":"2026-09-28 18:26:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6c254a8a826aa8c6d12922f6c0cd239d6b9b92db","subject":"Skip unscheduled volumes in Cinder collector","status":"MERGED","created":"2026-09-25 06:51:25.000000000","updated":"2026-09-28 18:30:08.000000000","submitted":"2026-09-28 18:26:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007420","meta_rev_id":"866f7fc8f2e6d2e4e154c1c323741a2fed1fab71","_number":1007420,"virtual_id_number":1007420,"owner":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-28 18:26:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2026-09-28 17:08:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":-1,"date":"2026-09-25 22:53:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-09-25 09:33:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2026-09-28 17:08:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"CC":[{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-25 06:51:29.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2026-09-25 07:52:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-25 09:33:53.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2026-09-25 22:53:57.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-09-28 17:08:44.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"}],"messages":[{"id":"8f2cbe5a3a161a690ed19e7af3b2c5658ed1d6c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-09-25 06:51:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffac143546aaf83869939b982ad18b03e68508f2","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-25 06:51:29.000000000","message":"Patch Set 1:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4adfc9e68da12bcf6d7cd1267f734bffc66fb7e1","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-09-25 07:00:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/339b2471f5be4eb9b3c56b86a6e1b918\n\n- teim-code-review https://zuul.teim.app/t/main/build/cfd632696de24997a2d48bd6ebfe003a : SUCCESS in 8m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d5e4ae81963aaa668cbf8dbae872ca3255a7dd6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 07:52:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1d28da61c454123b1264a168dcfc4bb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee71e484eb7842f7916ca2ca4b58e41f : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de36710248dc4b08a29f5582f19f5af5 : SUCCESS in 6m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c26bf82d2064bc68118d1e341aac7f2 : SUCCESS in 4m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1293b4dd70a748108549ada29a230642 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/597c0a648c284383a130f7b36d24766e : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c99f4dce4b6146538fbacbb31d171319 : SUCCESS in 4m 34s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/8d97372112b24625a4ecf68bc11d150c : SUCCESS in 23m 51s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3b31db6a2f9d4dea9db3f7d18f2444da : SUCCESS in 55m 38s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/728aec6c84da4c5a9ff10ad9f6906f6a : SUCCESS in 24m 16s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/53c519a27b0b46c99e0dc50687ac2551 : SUCCESS in 50m 55s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/84a202c5747346e49670ee0a09b3d348 : SUCCESS in 53m 00s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/3790c157f87140e2b53e8e159c03c548 : SUCCESS in 53m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8e6cdf87d3844505bc793a1df40b460c : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6b02a3ad62911b5646ef10d10a10d8b478f3cb4c","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2026-09-25 09:33:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9cf4292c8dfee26758a2ddee28f0d63c1a6b8c60","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-25 22:53:57.000000000","message":"Patch Set 1: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/ed9bb20b73f4484dbcbacb6648302809\n\n- openstack-meta-content-provider-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/4f1576bc52c740b19c46ca19912b4c6a : FAILURE in 8m 15s\n- watcher-operator-validation-master https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/b9023a62ac3d4374ba4e93dffac5693d : SKIPPED Skipped due to failed job openstack-meta-content-provider-master\n- watcher-s2i-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/2dd7ab48a79a4fd1acc9f6f697202fbf : FAILURE in 2m 11s (non-voting)\n- watcher-s2i-tempest https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/3fdc4936ef7e41fb976bc79dd9c24594 : SKIPPED Skipped due to failed job watcher-s2i-content-provider (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"728523f4ba0a5f6c148537bb322389c0a472e763","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2026-09-28 17:08:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d2ebd669a80c55817574998a2745eb83aae3a308","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 17:29:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77dff1e8309671a4d4a0f80436a62dc81cd745b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 18:26:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1a2d6a52058465381f4ccf619248298\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49991b5723824aa4be2d5d26e2ee9a0d : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d6ebc6e933d64f6eba6ff3fe0c91cc93 : SUCCESS in 4m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5a9f483f86f54dfcb41702c197f4021a : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1467105dc9546a1bb3a31145b80ed78 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e57ce0d7ca644f29d4efde70b77d24b : SUCCESS in 3m 33s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/f993f9c78bfd4500918a61d752110a2f : SUCCESS in 38m 03s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/62ed3d3548244f0b8439916251eeab6e : SUCCESS in 27m 16s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/1b96e440b3db46ecbed288b31887a048 : SUCCESS in 17m 42s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/2afd9dacc3b742338e4b58e51e0052cc : SUCCESS in 49m 08s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/e1517b2771f84739ad2eb57498d212e8 : SUCCESS in 53m 17s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/54762c6ff9ab479f91b4e175e8742335 : SUCCESS in 49m 18s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bf9f9bb936ea4d5ba9092c50e2b29520 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"75339fbcf38145576c2f32ec948aecd70911a3ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 18:26:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"866f7fc8f2e6d2e4e154c1c323741a2fed1fab71","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 18:30:08.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/24120413241d4b69abe1a1e471fd28b5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c37bf1c220274f43a4987d99ef920003 : FAILURE in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f0123e58bba4330b3b6f111e36d978c : FAILURE in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b5ab3b9474a6fe424f14e09db99523f3abc17a2","revisions":{"5b5ab3b9474a6fe424f14e09db99523f3abc17a2":{"kind":"REWORK","_number":1,"created":"2026-09-25 06:51:25.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/20/1007420/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/20/1007420/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/20/1007420/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/20/1007420/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/20/1007420/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/20/1007420/1"}}},"commit":{"parents":[{"commit":"67d38cb6c615029805d080f463e7b87568a5bcf1","subject":"Merge \"Do not skip tempest tests for requirements update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/67d38cb6c615029805d080f463e7b87568a5bcf1"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-09-25 06:49:08.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-09-25 06:49:08.000000000","tz":120},"subject":"Skip unscheduled volumes in Cinder collector","message":"Skip unscheduled volumes in Cinder collector\n\nVolumes in the \"creating\" state have not yet been assigned a\nbackend host by the Cinder scheduler. Passing host\u003dNone to\nelement.Volume(), which has a non-nullable StringField, caused\na ValueError that aborted the entire storage data model build.\n\nSkip any volume whose host is None before attempting to build\nits node. Such volumes will be picked up in a later collection\ncycle or via storage notifications once creation completes.\n\nCloses-Bug: #2167852\nGenerated-By: claude-code (claude-sonnet-4-6)\nChange-Id: I6c254a8a826aa8c6d12922f6c0cd239d6b9b92db\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/5b5ab3b9474a6fe424f14e09db99523f3abc17a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/5b5ab3b9474a6fe424f14e09db99523f3abc17a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
