)]}'
{"id":"openstack%2Fwatcher~1007693","triplet_id":"openstack%2Fwatcher~stable%2F2025.2~I4e3586b325c17155a0fee43c6657890130aa670d","project":"openstack/watcher","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"34452":{"account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"last_update":"2026-10-01 14:26:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4e3586b325c17155a0fee43c6657890130aa670d","subject":"Fix baremetal collector notification_endpoints returns None","status":"MERGED","created":"2026-09-28 11:00:53.000000000","updated":"2026-10-01 14:27:48.000000000","submitted":"2026-10-01 14:26:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007693","meta_rev_id":"f9385ddf91c3ecdd2a24d298704a121278c6e9cd","_number":1007693,"virtual_id_number":1007693,"owner":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-01 14:26:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2026-09-28 11:23:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2026-10-01 12:42:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2026-10-01 12:42:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-28 11:23:52.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2026-09-28 13:14:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-10-01 12:42:52.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"}],"messages":[{"id":"c9e788bed462e42b6cc74d39f580a3bb24cba6d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-09-28 11:00:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3723f1a8f2bc54c1f6a4dd16fe09af8b6969ddb0","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2026-09-28 11:23:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e76af57124ca54081151c37a4f23940e8ec70c95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 13:14:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90caf3968ea74dfb8428ea4fb25091f5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2162737304de4aaaa945faf48dd8ea07 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f94905a61d6b45718ccc01e0129f6a63 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7705a09feb4046679d0826f7d2d85f1a : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4f70addfe6954c008a07777c660112ee : SUCCESS in 7m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/300df3b0c58b4e3196ee0d8af99afcc7 : SUCCESS in 12m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab9c18fc99034ac89d62878976099a72 : SUCCESS in 9m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04d2e528675346f7b4cf8788ba40c369 : SUCCESS in 5m 20s\n- openstack-tox-py312-threading https://zuul.opendev.org/t/openstack/build/c9d547ece5704363a99b18dafa8a741f : SUCCESS in 6m 30s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/83a0ef2e401a47bea3f8b679ba5082cc : SUCCESS in 31m 24s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/c93600955c7c45c29cfc11311f72b959 : SUCCESS in 22m 21s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/860cc4895c044ed5b50682bf357bf05d : SUCCESS in 1h 36m 42s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/73b9cec4463445b88e706cf5fd0fb3a2 : SUCCESS in 1h 27m 38s\n- watcher-tempest-prometheus-threading https://zuul.opendev.org/t/openstack/build/5d240fce44c049ecbe261d0ca5cb05f3 : SUCCESS in 1h 15m 01s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/824bdac7cf9a45e8abcbc8aa6ed3c54d : SUCCESS in 1h 29m 08s","accounts_in_message":[],"_revision_number":1},{"id":"8725c9f8a301f70bb8221ebb0e315840953700fb","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2026-10-01 12:42:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"610b18d7da84ca53f9ac3e6cfd7d08440d4f1eeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 12:43:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"05c96f2f388b2a52e9240d8ba132dce10bc1e013","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 14:26:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55d3c74c6e0c403695f628e59b3bf2ce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/449aa7a41ad3479794186caa9eede9c9 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35b1bf3ad180452891feca4f9c98e921 : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ef6b6904da340beb1b0330a5cd88ec8 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/294edc0569df47a88a4fbc358658cd69 : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff9f0a121f7541419bcb6ff28cc0cf40 : SUCCESS in 4m 24s\n- openstack-tox-py312-threading https://zuul.opendev.org/t/openstack/build/92a36478486c4aab8b2a7f6c60b189bf : SUCCESS in 3m 36s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/951ce99cf36f4045b5a261b9ea6eb183 : SUCCESS in 51m 12s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/03478354fb904854a066dddccd0313cd : SUCCESS in 36m 44s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/55ddb01fd74547d4adf3c37b711c0f30 : SUCCESS in 1h 29m 59s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/2d1348025c5a4cd5b15b9b28e6c43d53 : SUCCESS in 1h 24m 11s\n- watcher-tempest-prometheus-threading https://zuul.opendev.org/t/openstack/build/64e46577133540e9af774670dc1c853a : SUCCESS in 1h 36m 46s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/c7dcfb34d90d4727af798e299ebcffef : SUCCESS in 1h 35m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a765fadedd2221a0d3c8eee2d1a2792099d43492","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 14:26:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f9385ddf91c3ecdd2a24d298704a121278c6e9cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 14:27:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/238db03f3d524e1ab45ed9912ae68d34\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e8d2fa3e56341eea25203c6e68f1617 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cb1f278e66b30100dae1354dcdd46ad3385fc415","revisions":{"cb1f278e66b30100dae1354dcdd46ad3385fc415":{"kind":"REWORK","_number":1,"created":"2026-09-28 11:00:53.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/93/1007693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/93/1007693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/93/1007693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/93/1007693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/93/1007693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/93/1007693/1"}}},"commit":{"parents":[{"commit":"001f455a931f9a344c958cf83185e11040051970","subject":"Merge \"Fix allowed_nodes corruption in ComputeScope\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/001f455a931f9a344c958cf83185e11040051970"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-09-16 13:18:17.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-09-28 10:52:25.000000000","tz":120},"subject":"Fix baremetal collector notification_endpoints returns None","message":"Fix baremetal collector notification_endpoints returns None\n\nBaremetalClusterDataModelCollector.notification_endpoints returned\nNone instead of an empty list, causing a TypeError when\nCollectorManager.get_notification_endpoints() called\nendpoints.extend() on the None value. This crashed the\nwatcher-decision-engine on startup when baremetal collector was\nconfigured.\n\nReturn [] to match the documented return type and allow iteration.\n\nCloses-Bug: #2167461\nGenerated-By: claude-code (claude-sonnet-4-6)\nChange-Id: I4e3586b325c17155a0fee43c6657890130aa670d\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n(cherry picked from commit 551d950035d4cce5fee7b261b4ca08754e8db1a6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/cb1f278e66b30100dae1354dcdd46ad3385fc415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/cb1f278e66b30100dae1354dcdd46ad3385fc415"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
