)]}'
{"id":"openstack%2Fwatcher~1008475","triplet_id":"openstack%2Fwatcher~stable%2F2026.2~I178dfcceb998704cb635f4ce3b6616c3fb2fe3e7","project":"openstack/watcher","branch":"stable/2026.2","attention_set":{},"removed_from_attention_set":{"34452":{"account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"last_update":"2026-10-05 13:21:48.000000000","reason":"\u003cGERRIT_ACCOUNT_34452\u003e replied on the change","reason_account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}},"28647":{"account":{"_account_id":28647,"name":"David","display_name":"morenod","email":"dsanzmor@redhat.com","username":"morenod"},"last_update":"2026-10-05 20:00:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I178dfcceb998704cb635f4ce3b6616c3fb2fe3e7","subject":"Rename multinode base job to three-node and use three-node nodeset","status":"MERGED","created":"2026-10-02 14:08:03.000000000","updated":"2026-10-05 20:02:27.000000000","submitted":"2026-10-05 20:00:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1008475","cherry_pick_of_change":995797,"cherry_pick_of_patch_set":8,"meta_rev_id":"66c82acea7a0593e28ec5e25f5a75de4ba0e00f0","_number":1008475,"virtual_id_number":1008475,"owner":{"_account_id":28647,"name":"David","display_name":"morenod","email":"dsanzmor@redhat.com","username":"morenod"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-05 20:00:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2026-10-05 13:55:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2026-10-05 13:21:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2026-10-05 13:55:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-10-02 17:08:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-10-05 10:14:59.000000000","updated_by":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"reviewer":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"state":"REVIEWER"},{"updated":"2026-10-05 13:55:27.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"}],"messages":[{"id":"5904d59a9bd465b3d27effa0e205bcdede4f1564","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28647,"name":"David","display_name":"morenod","email":"dsanzmor@redhat.com","username":"morenod"},"date":"2026-10-02 14:08:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9d64a4f0b4717d93a14db60420d9e73705939382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 17:08:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ca59adaae3147898143a00756462c96\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a3c5d66dd0d41e18ac924907f8e4651 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d32295d13a784ff3850da0066b5bfd9a : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ef23cecf50648c680da67784f2e9b6c : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec30ea28776a4294a03169c4e00e61f6 : SUCCESS in 4m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0068dc89714848bf94f0bbb395bce6b3 : SUCCESS in 11m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c4498da83b54940a872f15c5537d780 : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58b1b7b9788149c1be2613814e4795ad : SUCCESS in 5m 37s\n- openstack-tox-py313-eventlet https://zuul.opendev.org/t/openstack/build/12318c6501c345fb8d12c1ce748c2c8d : SUCCESS in 4m 39s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/efaac4dd37e84f5998a889cd77309d13 : SUCCESS in 27m 35s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f1a5c053a4724217add82d93d156e862 : SUCCESS in 44m 59s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/cb5e6ca502004cf19ac8ad23317e0027 : SUCCESS in 40m 11s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/567b81b06b654e1ba821f810d5f963ad : SUCCESS in 1h 39m 05s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/1c51a659f7834e608b5e697863bba053 : SUCCESS in 1h 29m 33s\n- watcher-tempest-prometheus-eventlet https://zuul.opendev.org/t/openstack/build/c83a425a95f6409483c0e0a611ae5111 : SUCCESS in 1h 34m 40s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/d1643fd5699c464aba91f6725359c1da : SUCCESS in 1h 45m 15s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9d15c098931349fc9b377a6c05861733 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"93a1d59a06c1313ebd90f5dd7fcaf8150040cbf7","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-10-05 10:14:59.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c33ea2d8934e86ef7e42dc2655d9e20bea517bd3","author":{"_account_id":28647,"name":"David","display_name":"morenod","email":"dsanzmor@redhat.com","username":"morenod"},"date":"2026-10-05 11:44:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"25320a5b27a2299c5e559f821df15313586e0044","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-10-05 13:21:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54cfd0a779d683cdcbcbd33b8fd38c79eb8aa12a","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2026-10-05 13:55:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8531596da7f553e39fb4a43eaa1b6cc20dc0b64b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 13:56:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb267f80477dde555c9418d4da5813b5654ce888","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 15:45:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f9ed21597344e8089d1ce53d75364c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b66c5c022c774209b75fda1c191f3678 : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95ac7139570f4e8cbfdad1a9c7008c37 : SUCCESS in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9645717237e41c1834eaa71ef072454 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbc7b2ff59e04930962c4d710727b949 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f4119a726bb4b0584aae171f011c7f8 : SUCCESS in 5m 22s\n- openstack-tox-py313-eventlet https://zuul.opendev.org/t/openstack/build/c7640f98ff1048a384df5880cbd23fb5 : SUCCESS in 6m 07s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/447e11cb87564e4a914e259c2b2e2a78 : SUCCESS in 42m 34s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/505eaa108c1d49fea44026b5dc63c7ef : SUCCESS in 58m 51s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/748a0b68621141929c6d65e328aee06b : SUCCESS in 23m 51s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/db4de8b6619745c1a3b12d0ca4c4cf19 : FAILURE in 1h 42m 48s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/53729f9b7ae64b269929bd5416cfbb87 : SUCCESS in 1h 42m 45s\n- watcher-tempest-prometheus-eventlet https://zuul.opendev.org/t/openstack/build/29d52745a06545c2beaa665f4db63256 : SUCCESS in 1h 21m 55s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/4dfff84380db4f8fa1bc9a8710ec7bf4 : SUCCESS in 1h 42m 34s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/aa41851884c1479db67cec4eeb38db23 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b3a2d1a7bb98ca9843d497f88db67cbd6d179055","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-10-05 15:49:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81dd8629c827c9a090884a452f02e1fcf6377ebc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 18:07:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3faf96b332364e46b357ff84277da9d3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7a83689adf14f54817e9a14b70e0bbe : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d080447e3c5b473f8c5606c84ba2540d : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/528f45ca79f343238eca927c216661b4 : SUCCESS in 5m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43cb24401d6f443889d1c78f4b880a97 : SUCCESS in 6m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c47e30eac8b54da9969ab964d96d378c : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11b8d97710fa479aa3261c22fb5b79e4 : SUCCESS in 9m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21a00125ce31435a8778f6509c24f02d : SUCCESS in 4m 53s\n- openstack-tox-py313-eventlet https://zuul.opendev.org/t/openstack/build/2fafec47725a4f9cb8717b3cabdf044d : SUCCESS in 5m 11s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/8b1ac5b241fb4ac5bd1e88c72a75fc90 : SUCCESS in 52m 37s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4a50e48835634cd1bfb3c3da0cbbe876 : SUCCESS in 55m 53s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/5c03dfdf7d0046e39bf99bb4d42f652e : SUCCESS in 44m 08s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/23d1cd8a5c324e3f85efb6e30ea2e42e : SUCCESS in 1h 42m 02s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/7b281100e616473a8699c89109bacd40 : SUCCESS in 1h 46m 52s\n- watcher-tempest-prometheus-eventlet https://zuul.opendev.org/t/openstack/build/f712afc328bb4d2e9593b8a9745993ed : SUCCESS in 1h 39m 09s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/a4349ab6ac394a6e80bfe3a625e9c914 : SUCCESS in 1h 48m 01s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/42618a839d89404aa8c8a4ad137e73a6 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b6f7f38c0b8fbe36356a14b3929c18f73fd746e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 18:08:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06effcf743ec1892a303586e09528b3e2a5e35c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 20:00:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8158ddf09144a739fbf9e435aaa164e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c669a32eb904afbb113aea6b122f21c : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19b97c6d870442e0b4f2587be8690280 : SUCCESS in 5m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/942f8c46922f4061a45c30385aa03e11 : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e3c9bbedab94d7283fd789a5ae3acfd : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cb935ed3c264bd295abe82d88d70085 : SUCCESS in 5m 28s\n- openstack-tox-py313-eventlet https://zuul.opendev.org/t/openstack/build/fb2d0a1e369d427a800ccdf3e19890b5 : SUCCESS in 6m 10s\n- watcher-grenade https://zuul.opendev.org/t/openstack/build/63931cbbc4b84bdeb151383f56463f6b : SUCCESS in 55m 06s\n- watcher-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1c5d186ff2a543c7a8913eba05ff4283 : SUCCESS in 56m 25s\n- watcher-tempest-api-ipv6-only https://zuul.opendev.org/t/openstack/build/17bdb41c85ed4e7fa3d5a2635632f49c : SUCCESS in 35m 20s\n- watcher-tempest-gnocchi https://zuul.opendev.org/t/openstack/build/235b4c73983c455f9ac504e96246eb19 : SUCCESS in 1h 39m 30s\n- watcher-tempest-prometheus https://zuul.opendev.org/t/openstack/build/354da2c576d64067bb96807646071082 : SUCCESS in 1h 45m 10s\n- watcher-tempest-prometheus-eventlet https://zuul.opendev.org/t/openstack/build/d29e2891b43848c082081454dfcb0969 : SUCCESS in 1h 19m 52s\n- watcher-tempest-aetos https://zuul.opendev.org/t/openstack/build/7bc849fd301345d8b9391ab6d31f7d38 : SUCCESS in 1h 23m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b5de90d178674004a57c57b61027be2d : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"12197b114359b8c4aae317f807ae4bafc128b55b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 20:00:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"66c82acea7a0593e28ec5e25f5a75de4ba0e00f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 20:02:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4583cf82270b4999840671477dc5412f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3039b3cc177445e68c1bccba044840d9 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fabe43e9d4a244d08604c3eed41fb48149bd80b3","revisions":{"fabe43e9d4a244d08604c3eed41fb48149bd80b3":{"kind":"REWORK","_number":1,"created":"2026-10-02 14:08:03.000000000","uploader":{"_account_id":28647,"name":"David","display_name":"morenod","email":"dsanzmor@redhat.com","username":"morenod"},"ref":"refs/changes/75/1008475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/75/1008475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/75/1008475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/75/1008475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/75/1008475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/75/1008475/1"}}},"commit":{"parents":[{"commit":"3ad41c9e9603a899036cca5bfdbe0ad54f49d020","subject":"Merge \"Fix baremetal collector notification_endpoints returns None\" into stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/3ad41c9e9603a899036cca5bfdbe0ad54f49d020"}]}],"author":{"name":"morenod","email":"dsanzmor@redhat.com","date":"2026-07-06 10:02:24.000000000","tz":120},"committer":{"name":"David","email":"dsanzmor@redhat.com","date":"2026-10-02 14:08:03.000000000","tz":0},"subject":"Rename multinode base job to three-node and use three-node nodeset","message":"Rename multinode base job to three-node and use three-node nodeset\n\nRename watcher-tempest-base-two-node to watcher-tempest-base-three-node\nand change its nodeset from openstack-two-node-noble to\nopenstack-three-node-noble. This provides 3 compute services for\nnon-real-data jobs, which is needed for scope tests that require 2 source\nnodes (with 1 excluded) and 1 destination node.\n\nThe openstack-three-node-noble nodeset already defines all three nodes\n(controller + compute1 + compute2) in the compute group. The actual\nnumber of active compute services per job is controlled by the n-cpu\nvariable:\n\n- Non-real-data jobs: n-cpu is enabled by default on all nodes\n  (3 compute services: controller, compute1, compute2)\n\n- Real-data jobs: set n-cpu: false on controller in devstack_services\n  (2 compute services: compute1, compute2)\n\nThis approach reuses the existing nodeset instead of creating a duplicate,\navoiding maintenance overhead and potential divergence.\n\nJobs updated to use the new parent name:\n- watcher-tempest-gnocchi\n- watcher-tempest-prometheus\n\nChange-Id: I178dfcceb998704cb635f4ce3b6616c3fb2fe3e7\nSigned-off-by: morenod \u003cdsanzmor@redhat.com\u003e\nAssisted-By: Claude Sonnet 4.5 \u003cnoreply@anthropic.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/watcher-tempest-plugin/+/996222\n(cherry picked from commit 68140367e191b7956dc26566b66053eaf828f571)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/fabe43e9d4a244d08604c3eed41fb48149bd80b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/fabe43e9d4a244d08604c3eed41fb48149bd80b3"}]},"branch":"refs/heads/stable/2026.2","conflicts":{"base":"642b193e739b78b233ccfb8ed7d75cb157c8ba56","ours":"3ad41c9e9603a899036cca5bfdbe0ad54f49d020","theirs":"68140367e191b7956dc26566b66053eaf828f571","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
