)]}'
{"id":"openstack%2Fwatcher~407881","triplet_id":"openstack%2Fwatcher~master~Ia8edc9b428c14ea35225cbbe1a54b67a33151a64","project":"openstack/watcher","branch":"master","topic":"fix_methodname_in_sample","hashtags":[],"change_id":"Ia8edc9b428c14ea35225cbbe1a54b67a33151a64","subject":"Fix method name in doc/source/dev/plugin/action-plugin.rst","status":"MERGED","created":"2016-12-07 06:53:14.000000000","updated":"2016-12-16 17:02:41.000000000","submitted":"2016-12-16 17:02:41.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c4d7fdfd8637da0e3d8f17dff431d6c6ccecff3f","_number":407881,"virtual_id_number":407881,"owner":{"_account_id":13111,"name":"Hidekazu Nakamura","email":"hidekazuna@gmail.com","username":"hid-nakamura"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2016-12-16 17:02:41.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":12394,"name":"David TARDIVEL","email":"David.TARDIVEL@b-com.com","username":"David.T"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12394,"name":"David TARDIVEL","email":"David.TARDIVEL@b-com.com","username":"David.T"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-12-16 16:56:28.000000000","_account_id":12394,"name":"David TARDIVEL","email":"David.TARDIVEL@b-com.com","username":"David.T"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12394,"name":"David TARDIVEL","email":"David.TARDIVEL@b-com.com","username":"David.T"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-12-16 16:56:28.000000000","_account_id":12394,"name":"David TARDIVEL","email":"David.TARDIVEL@b-com.com","username":"David.T"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":12394,"name":"David TARDIVEL","email":"David.TARDIVEL@b-com.com","username":"David.T"}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":12394,"name":"David TARDIVEL","email":"David.TARDIVEL@b-com.com","username":"David.T"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-12-16 16:56:28.000000000","updated_by":{"_account_id":12394,"name":"David TARDIVEL","email":"David.TARDIVEL@b-com.com","username":"David.T"},"reviewer":{"_account_id":12394,"name":"David TARDIVEL","email":"David.TARDIVEL@b-com.com","username":"David.T"},"state":"REVIEWER"},{"updated":"2016-12-16 17:02:41.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"ffa9336f5315248bf119f5fec96b920579ce8b18","author":{"_account_id":13111,"name":"Hidekazu Nakamura","email":"hidekazuna@gmail.com","username":"hid-nakamura"},"date":"2016-12-07 06:53:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47e2b6214d661b491ff73d48725aa6aabdeb1a8b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-07 07:20:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-watcher-docs-ubuntu-xenial http://docs-draft.openstack.org/81/407881/1/check/gate-watcher-docs-ubuntu-xenial/c65b0fc//doc/build/html/ : SUCCESS in 4m 39s\n- gate-watcher-pep8-ubuntu-xenial http://logs.openstack.org/81/407881/1/check/gate-watcher-pep8-ubuntu-xenial/a181f04/ : SUCCESS in 2m 36s\n- gate-watcher-python27-ubuntu-xenial http://logs.openstack.org/81/407881/1/check/gate-watcher-python27-ubuntu-xenial/88ec5d9/ : FAILURE in 2m 40s\n- gate-watcher-python34 http://logs.openstack.org/81/407881/1/check/gate-watcher-python34/785782b/ : FAILURE in 4m 59s\n- gate-watcher-python35 http://logs.openstack.org/81/407881/1/check/gate-watcher-python35/418ec0a/ : FAILURE in 2m 57s\n- watcher-coverage-ubuntu-xenial http://logs.openstack.org/81/407881/1/check/watcher-coverage-ubuntu-xenial/bc1022e/ : FAILURE in 2m 58s (non-voting)\n- gate-rally-dsvm-watcher-watcher-ubuntu-xenial-nv http://logs.openstack.org/81/407881/1/check/gate-rally-dsvm-watcher-watcher-ubuntu-xenial-nv/d28b571/ : SUCCESS in 16m 19s (non-voting)\n- gate-watcher-dsvm-multinode-ubuntu-xenial-nv http://logs.openstack.org/81/407881/1/check/gate-watcher-dsvm-multinode-ubuntu-xenial-nv/fa450a5/ : SUCCESS in 25m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d322a75c7e3fcd7f46713c717ef23fd895344ddc","author":{"_account_id":12394,"name":"David TARDIVEL","email":"David.TARDIVEL@b-com.com","username":"David.T"},"date":"2016-12-07 07:52:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c667eb4b4d051f5b32076b405e80a4999f99c090","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-07 08:21:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-watcher-docs-ubuntu-xenial http://docs-draft.openstack.org/81/407881/1/check/gate-watcher-docs-ubuntu-xenial/8f97eee//doc/build/html/ : SUCCESS in 3m 08s\n- gate-watcher-pep8-ubuntu-xenial http://logs.openstack.org/81/407881/1/check/gate-watcher-pep8-ubuntu-xenial/56cbf4f/ : SUCCESS in 2m 35s\n- gate-watcher-python27-ubuntu-xenial http://logs.openstack.org/81/407881/1/check/gate-watcher-python27-ubuntu-xenial/5560409/ : FAILURE in 2m 40s\n- gate-watcher-python34 http://logs.openstack.org/81/407881/1/check/gate-watcher-python34/127bc7e/ : FAILURE in 2m 41s\n- gate-watcher-python35 http://logs.openstack.org/81/407881/1/check/gate-watcher-python35/f2e6f88/ : FAILURE in 5m 11s\n- watcher-coverage-ubuntu-xenial http://logs.openstack.org/81/407881/1/check/watcher-coverage-ubuntu-xenial/0ef0641/ : FAILURE in 5m 00s (non-voting)\n- gate-rally-dsvm-watcher-watcher-ubuntu-xenial-nv http://logs.openstack.org/81/407881/1/check/gate-rally-dsvm-watcher-watcher-ubuntu-xenial-nv/fe45ca0/ : SUCCESS in 26m 20s (non-voting)\n- gate-watcher-dsvm-multinode-ubuntu-xenial-nv http://logs.openstack.org/81/407881/1/check/gate-watcher-dsvm-multinode-ubuntu-xenial-nv/d26d53e/ : SUCCESS in 26m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83b64cf9ee7f2a5e74752b8b7d6b8f239d3fe8d5","author":{"_account_id":12394,"name":"David TARDIVEL","email":"David.TARDIVEL@b-com.com","username":"David.T"},"date":"2016-12-07 08:54:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2eb88f6fbfa9402c47588f19af313ce108c8a753","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-07 09:24:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-watcher-docs-ubuntu-xenial http://docs-draft.openstack.org/81/407881/1/check/gate-watcher-docs-ubuntu-xenial/6dfd771//doc/build/html/ : SUCCESS in 3m 22s\n- gate-watcher-pep8-ubuntu-xenial http://logs.openstack.org/81/407881/1/check/gate-watcher-pep8-ubuntu-xenial/b31e0ef/ : SUCCESS in 2m 35s\n- gate-watcher-python27-ubuntu-xenial http://logs.openstack.org/81/407881/1/check/gate-watcher-python27-ubuntu-xenial/8563f85/ : SUCCESS in 4m 32s\n- gate-watcher-python34 http://logs.openstack.org/81/407881/1/check/gate-watcher-python34/934c517/ : SUCCESS in 3m 03s\n- gate-watcher-python35 http://logs.openstack.org/81/407881/1/check/gate-watcher-python35/6e4c235/ : SUCCESS in 4m 00s\n- watcher-coverage-ubuntu-xenial http://logs.openstack.org/81/407881/1/check/watcher-coverage-ubuntu-xenial/852b4ed/ : SUCCESS in 4m 22s (non-voting)\n- gate-rally-dsvm-watcher-watcher-ubuntu-xenial-nv http://logs.openstack.org/81/407881/1/check/gate-rally-dsvm-watcher-watcher-ubuntu-xenial-nv/3d0e237/ : SUCCESS in 25m 25s (non-voting)\n- gate-watcher-dsvm-multinode-ubuntu-xenial-nv http://logs.openstack.org/81/407881/1/check/gate-watcher-dsvm-multinode-ubuntu-xenial-nv/f59c765/ : SUCCESS in 26m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d50a8b0835cde9ed845707a01a6a9c265888d4dc","author":{"_account_id":12394,"name":"David TARDIVEL","email":"David.TARDIVEL@b-com.com","username":"David.T"},"date":"2016-12-16 16:56:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ef1b52a992c03ff67fb3c204d607f8fd44537177","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-16 16:56:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"ae7d2d8bd795304f3b9fae73f91517fe404c902b","date":"2016-12-16 17:02:41.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"afc1ecb971e46c897dd18a215ea307c8644b6dac","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-16 17:02:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-watcher-docs-ubuntu-xenial http://docs-draft.openstack.org/81/407881/1/gate/gate-watcher-docs-ubuntu-xenial/8f8d719//doc/build/html/ : SUCCESS in 3m 30s\n- gate-watcher-pep8-ubuntu-xenial http://logs.openstack.org/81/407881/1/gate/gate-watcher-pep8-ubuntu-xenial/5d8dc4c/ : SUCCESS in 5m 35s\n- gate-watcher-python27-ubuntu-xenial http://logs.openstack.org/81/407881/1/gate/gate-watcher-python27-ubuntu-xenial/2d53e6e/ : SUCCESS in 5m 18s\n- gate-watcher-python35 http://logs.openstack.org/81/407881/1/gate/gate-watcher-python35/52b21fe/ : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a323ed54f9142fa841338a03ecebdc5bcfb44a5","revisions":{"6a323ed54f9142fa841338a03ecebdc5bcfb44a5":{"kind":"REWORK","_number":1,"created":"2016-12-07 06:53:14.000000000","uploader":{"_account_id":13111,"name":"Hidekazu Nakamura","email":"hidekazuna@gmail.com","username":"hid-nakamura"},"ref":"refs/changes/81/407881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/81/407881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/81/407881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/81/407881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/81/407881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/81/407881/1"}}},"commit":{"parents":[{"commit":"4e240b945b9bae4aa75e3bdd59141a94d3d87b6b","subject":"Merge \"Fix one ref that does not work.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/4e240b945b9bae4aa75e3bdd59141a94d3d87b6b"}]}],"author":{"name":"Hidekazu Nakamura","email":"hid-nakamura@vf.jp.nec.com","date":"2016-12-07 06:50:51.000000000","tz":540},"committer":{"name":"Hidekazu Nakamura","email":"hid-nakamura@vf.jp.nec.com","date":"2016-12-07 06:50:51.000000000","tz":540},"subject":"Fix method name in doc/source/dev/plugin/action-plugin.rst","message":"Fix method name in doc/source/dev/plugin/action-plugin.rst\n\nprecondition -\u003e pre_condition\npostcondition -\u003e post_condition\n\nChange-Id: Ia8edc9b428c14ea35225cbbe1a54b67a33151a64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/6a323ed54f9142fa841338a03ecebdc5bcfb44a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/6a323ed54f9142fa841338a03ecebdc5bcfb44a5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4e240b945b9bae4aa75e3bdd59141a94d3d87b6b","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
