)]}'
{"id":"openstack%2Fwatcher~482754","triplet_id":"openstack%2Fwatcher~master~I063d7c8d16fbbab01b9bec9a0e8e4d4dbbfbfb64","project":"openstack/watcher","branch":"master","topic":"bp/cdm-scoping","hashtags":[],"change_id":"I063d7c8d16fbbab01b9bec9a0e8e4d4dbbfbfb64","subject":"[WIP] CDM Scoping","status":"ABANDONED","created":"2017-07-12 01:00:05.000000000","updated":"2017-07-21 02:36:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fae48e988f47e3eb17a548f9ebc135621a6d60f8","_number":482754,"virtual_id_number":482754,"owner":{"_account_id":13111,"name":"Hidekazu Nakamura","email":"hidekazuna@gmail.com","username":"hid-nakamura"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":-1,"date":"2017-07-12 02:05:00.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2017-07-17 13:52:46.000000000","_account_id":19055,"name":"Alexander Chadin","email":"joker946@gmail.com","username":"joker946"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19055,"name":"Alexander Chadin","email":"joker946@gmail.com","username":"joker946"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":19055,"name":"Alexander Chadin","email":"joker946@gmail.com","username":"joker946"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":19055,"name":"Alexander Chadin","email":"joker946@gmail.com","username":"joker946"}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":19055,"name":"Alexander Chadin","email":"joker946@gmail.com","username":"joker946"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-07-12 02:05:00.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-07-17 13:52:46.000000000","updated_by":{"_account_id":19055,"name":"Alexander Chadin","email":"joker946@gmail.com","username":"joker946"},"reviewer":{"_account_id":19055,"name":"Alexander Chadin","email":"joker946@gmail.com","username":"joker946"},"state":"REVIEWER"}],"messages":[{"id":"abc3c633b21f9a8886ca5b7bf21c29b8effa5692","author":{"_account_id":13111,"name":"Hidekazu Nakamura","email":"hidekazuna@gmail.com","username":"hid-nakamura"},"date":"2017-07-12 01:00:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3479d880a8323fc57b90d4fbfcf2a59eed8b9eb5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-12 02:05:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-watcher-docs-ubuntu-xenial http://docs-draft.openstack.org/54/482754/1/check/gate-watcher-docs-ubuntu-xenial/bafe125//doc/build/html/ : SUCCESS in 3m 09s\n- gate-watcher-pep8-ubuntu-xenial http://logs.openstack.org/54/482754/1/check/gate-watcher-pep8-ubuntu-xenial/4e0823a/ : FAILURE in 6m 41s\n- gate-watcher-python27-ubuntu-xenial http://logs.openstack.org/54/482754/1/check/gate-watcher-python27-ubuntu-xenial/e22556e/ : SUCCESS in 3m 21s\n- gate-watcher-python35 http://logs.openstack.org/54/482754/1/check/gate-watcher-python35/907acbf/ : SUCCESS in 5m 35s\n- watcher-coverage-ubuntu-xenial-nv http://logs.openstack.org/54/482754/1/check/watcher-coverage-ubuntu-xenial-nv/a0863f4/ : SUCCESS in 5m 12s (non-voting)\n- gate-rally-dsvm-watcher-watcher-ubuntu-xenial-nv http://logs.openstack.org/54/482754/1/check/gate-rally-dsvm-watcher-watcher-ubuntu-xenial-nv/8de236f/ : SUCCESS in 27m 40s (non-voting)\n- gate-watcher-dsvm-multinode-ubuntu-xenial-nv http://logs.openstack.org/54/482754/1/check/gate-watcher-dsvm-multinode-ubuntu-xenial-nv/930c41d/ : FAILURE in 50m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56ffb80180470bd0c61523b799605dfd71a0daff","author":{"_account_id":19055,"name":"Alexander Chadin","email":"joker946@gmail.com","username":"joker946"},"date":"2017-07-17 13:52:46.000000000","message":"Patch Set 1:\n\nPlease, don\u0027t forget to submit spec","accounts_in_message":[],"_revision_number":1},{"id":"0b9c0dd0ecc7a0ffc9f8cadf71de16f1a7421a27","author":{"_account_id":13111,"name":"Hidekazu Nakamura","email":"hidekazuna@gmail.com","username":"hid-nakamura"},"date":"2017-07-21 02:36:59.000000000","message":"Abandoned\n\nThis is WIP","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a883019b83c1a568d7109326799cdb07386e90a5","revisions":{"a883019b83c1a568d7109326799cdb07386e90a5":{"kind":"REWORK","_number":1,"created":"2017-07-12 01:00:05.000000000","uploader":{"_account_id":13111,"name":"Hidekazu Nakamura","email":"hidekazuna@gmail.com","username":"hid-nakamura"},"ref":"refs/changes/54/482754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/54/482754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/54/482754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/54/482754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/54/482754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/54/482754/1"}}},"commit":{"parents":[{"commit":"63c1aabdd29ffbb9e0204337a0e69241e5aa3588","subject":"bug fix: Can\u0027t get sample through CeilometerHelper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/63c1aabdd29ffbb9e0204337a0e69241e5aa3588"}]}],"author":{"name":"Hidekazu Nakamura","email":"hid-nakamura@vf.jp.nec.com","date":"2017-07-12 00:58:05.000000000","tz":540},"committer":{"name":"Hidekazu Nakamura","email":"hid-nakamura@vf.jp.nec.com","date":"2017-07-12 00:58:05.000000000","tz":540},"subject":"[WIP] CDM Scoping","message":"[WIP] CDM Scoping\n\nThis patch moves audit_scope_handler from BaseStrategy to\nBaseClusterDataModelCollector.\n\nChange-Id: I063d7c8d16fbbab01b9bec9a0e8e4d4dbbfbfb64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/a883019b83c1a568d7109326799cdb07386e90a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/a883019b83c1a568d7109326799cdb07386e90a5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"63c1aabdd29ffbb9e0204337a0e69241e5aa3588","is_merged_in_target_branch":true,"change_id":"I293b130c3f709dc93cd4b1b7382ae8895a54765d","change_number":482016,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
