)]}'
{"id":"openstack%2Fwatcher~663332","triplet_id":"openstack%2Fwatcher~master~I7e441608b38338eecd80e663ed3abe66a89e504f","project":"openstack/watcher","branch":"master","topic":"bug/1831917","hashtags":[],"change_id":"I7e441608b38338eecd80e663ed3abe66a89e504f","subject":"Configure nova notification format in non-grenade CI jobs","status":"MERGED","created":"2019-06-05 13:34:38.000000000","updated":"2019-06-28 02:29:14.000000000","submitted":"2019-06-28 02:29:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663332-1561688954679-4aedcafd","meta_rev_id":"5212002b5b051723fe10b5eeceeb1d5d3e1f0baf","_number":663332,"virtual_id_number":663332,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"date":"2019-06-25 15:00:03.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-06-28 02:29:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"recommended":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":1,"date":"2019-06-25 15:18:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2019-06-26 09:07:33.000000000","_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2019-06-28 01:48:49.000000000","_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 15:13:06.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-06-10 10:16:52.000000000","updated_by":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"reviewer":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"state":"REVIEWER"},{"updated":"2019-06-25 15:18:43.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-06-28 01:48:49.000000000","updated_by":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"reviewer":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"state":"REVIEWER"},{"updated":"2019-06-28 02:29:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9dad3cda69011322aeb5d6f31df2316863442db6","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-05 13:34:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1c2c7f7e0a3759433cee3af308ddd06deb3e046","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 14:58:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/32/663332/1/check/openstack-tox-cover/688d633/cover/ : SUCCESS in 5m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/663332/1/check/openstack-tox-lower-constraints/1174e93/ : SUCCESS in 5m 33s\n- openstack-tox-pep8 http://logs.openstack.org/32/663332/1/check/openstack-tox-pep8/ff7d441/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/32/663332/1/check/openstack-tox-py27/9fa5574/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/32/663332/1/check/openstack-tox-py36/ec0be2a/ : SUCCESS in 5m 34s\n- openstack-tox-py37 http://logs.openstack.org/32/663332/1/check/openstack-tox-py37/f7962fd/ : SUCCESS in 6m 47s\n- openstack-tox-docs http://logs.openstack.org/32/663332/1/check/openstack-tox-docs/ecf5b4b/html/ : SUCCESS in 4m 52s\n- watcher-tempest-functional http://logs.openstack.org/32/663332/1/check/watcher-tempest-functional/9fa1092/ : SUCCESS in 40m 41s\n- watcher-grenade http://logs.openstack.org/32/663332/1/check/watcher-grenade/ed2534c/ : SUCCESS in 48m 39s\n- watcher-tempest-dummy_optim http://logs.openstack.org/32/663332/1/check/watcher-tempest-dummy_optim/6ef334b/ : SUCCESS in 46m 01s\n- watcher-tempest-actuator http://logs.openstack.org/32/663332/1/check/watcher-tempest-actuator/a69cc96/ : SUCCESS in 51m 06s\n- watcher-tempest-basic_optim http://logs.openstack.org/32/663332/1/check/watcher-tempest-basic_optim/d66ef1d/ : SUCCESS in 49m 48s\n- watcher-tempest-vm_workload_consolidation http://logs.openstack.org/32/663332/1/check/watcher-tempest-vm_workload_consolidation/67d143a/ : SUCCESS in 49m 36s\n- watcher-tempest-workload_balancing http://logs.openstack.org/32/663332/1/check/watcher-tempest-workload_balancing/a367cb6/ : SUCCESS in 49m 39s\n- watcherclient-tempest-functional http://logs.openstack.org/32/663332/1/check/watcherclient-tempest-functional/59632f1/ : SUCCESS in 35m 07s\n- watcher-tempest-zone_migration http://logs.openstack.org/32/663332/1/check/watcher-tempest-zone_migration/52f0c60/ : SUCCESS in 1h 00m 54s\n- watcher-tempest-host_maintenance http://logs.openstack.org/32/663332/1/check/watcher-tempest-host_maintenance/cc0aef9/ : SUCCESS in 47m 59s\n- watcher-tempest-storage_balance http://logs.openstack.org/32/663332/1/check/watcher-tempest-storage_balance/89e5c7e/ : SUCCESS in 48m 22s\n- watcher-tls-test http://logs.openstack.org/32/663332/1/check/watcher-tls-test/17c6f83/ : SUCCESS in 48m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d3e8a5b55c433d33854ac18c75984c9cba95d7ad","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-05 15:15:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e6a04152e68c2ff5d77a30dd9a9582510f94e40d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-05 17:40:04.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e32b034ae28ddc2785550756fed36d46d4d01c5d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-05 17:41:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"15d9e06eabf3416d739de0630f0baaab26c4fe0a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-05 18:47:10.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"2af02a46e6afcc025f7aaec17b05b104b3994041","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-06 15:48:53.000000000","message":"Patch Set 1:\n\nOK it looks like these files are just used for contributor docs and samples:\n\nhttps://docs.openstack.org/watcher/latest/contributor/devstack.html\n\nThe watcher devstack plugin code used to configure nova for notifications I1c9c141d98d858c36ad8bb7be0b95c38ff1d5752 but removed that when watcher changed over to versioned notifications because nova just emitted those by default (until it didn\u0027t). So I likely need to modify either the devstack plugin code like I1c9c141d98d858c36ad8bb7be0b95c38ff1d5752 used to do or set the configuration from the parent .zuul.yaml job definition - not sure which is cleaner (maybe the former).","accounts_in_message":[],"_revision_number":1},{"id":"244c25f5c69b884b1171b0c1e960f93912a9df8b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-06 15:55:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"22e9ad265399e69743d11e3dd21d60c32e5e23e7","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-06 15:56:15.000000000","message":"Patch Set 2: Workflow-1\n\nI just want to see if this works in the non-grenade jobs. Since the grenade job is using a legacy setup it won\u0027t pick up the watcher-tempest-functional change, so we might need to just configure nova from the devstack/lib/watcher script like it did before I1c9c141d98d858c36ad8bb7be0b95c38ff1d5752 - the only thing I\u0027m not sure about that in that one is if nova gets restarted after changing its configuration.","accounts_in_message":[],"_revision_number":2},{"id":"c6475fbfe7cc91808805f3f5ae8a9c362c2b4d9d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-06 15:58:58.000000000","message":"Patch Set 2:\n\nReported bug 1831917 for this.","accounts_in_message":[],"_revision_number":2},{"id":"3954e9f2f04c7fd788e19d4df25ffb8d4b762511","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 16:56:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/32/663332/2/check/openstack-tox-cover/9cb935f/cover/ : SUCCESS in 5m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/663332/2/check/openstack-tox-lower-constraints/72aabbc/ : SUCCESS in 5m 08s\n- openstack-tox-pep8 http://logs.openstack.org/32/663332/2/check/openstack-tox-pep8/7e586cd/ : SUCCESS in 5m 21s\n- openstack-tox-py27 http://logs.openstack.org/32/663332/2/check/openstack-tox-py27/329fe5d/ : SUCCESS in 5m 05s\n- openstack-tox-py36 http://logs.openstack.org/32/663332/2/check/openstack-tox-py36/8ab92f7/ : SUCCESS in 5m 20s\n- openstack-tox-py37 http://logs.openstack.org/32/663332/2/check/openstack-tox-py37/64fc4ba/ : SUCCESS in 6m 26s\n- openstack-tox-docs http://logs.openstack.org/32/663332/2/check/openstack-tox-docs/a4e5f41/html/ : SUCCESS in 5m 09s\n- watcher-tempest-functional http://logs.openstack.org/32/663332/2/check/watcher-tempest-functional/b895286/ : SUCCESS in 31m 16s\n- watcher-grenade http://logs.openstack.org/32/663332/2/check/watcher-grenade/dc130e3/ : SUCCESS in 49m 33s\n- watcher-tempest-dummy_optim http://logs.openstack.org/32/663332/2/check/watcher-tempest-dummy_optim/130f590/ : SUCCESS in 46m 11s\n- watcher-tempest-actuator http://logs.openstack.org/32/663332/2/check/watcher-tempest-actuator/9cf91fc/ : SUCCESS in 50m 06s\n- watcher-tempest-basic_optim http://logs.openstack.org/32/663332/2/check/watcher-tempest-basic_optim/89700bd/ : SUCCESS in 52m 16s\n- watcher-tempest-vm_workload_consolidation http://logs.openstack.org/32/663332/2/check/watcher-tempest-vm_workload_consolidation/d2b1627/ : SUCCESS in 57m 50s\n- watcher-tempest-workload_balancing http://logs.openstack.org/32/663332/2/check/watcher-tempest-workload_balancing/b575c9f/ : SUCCESS in 50m 46s\n- watcherclient-tempest-functional http://logs.openstack.org/32/663332/2/check/watcherclient-tempest-functional/eccd317/ : SUCCESS in 36m 45s\n- watcher-tempest-zone_migration http://logs.openstack.org/32/663332/2/check/watcher-tempest-zone_migration/34ff00c/ : SUCCESS in 45m 33s\n- watcher-tempest-host_maintenance http://logs.openstack.org/32/663332/2/check/watcher-tempest-host_maintenance/0fff7a6/ : SUCCESS in 50m 12s\n- watcher-tempest-storage_balance http://logs.openstack.org/32/663332/2/check/watcher-tempest-storage_balance/4bf83b9/ : SUCCESS in 48m 59s\n- watcher-tls-test http://logs.openstack.org/32/663332/2/check/watcher-tls-test/bc2cd12/ : SUCCESS in 49m 35s","accounts_in_message":[],"_revision_number":2},{"id":"c40fbb474e5e800ead4bda3ddebde94bfbdd970a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-07 17:11:29.000000000","message":"Patch Set 2:\n\nOK this works for the non-grenade jobs:\n\nhttp://logs.openstack.org/32/663332/2/check/watcher-tempest-vm_workload_consolidation/d2b1627/controller/logs/screen-n-api.txt.gz#_Jun_06_16_32_34_453551\n\nhttp://logs.openstack.org/32/663332/2/check/watcher-tempest-vm_workload_consolidation/d2b1627/controller/logs/screen-n-cpu.txt.gz#_Jun_06_16_34_12_129117\n\nNow I just need to fix the grenade job, which likely means moving this into the devstack plugin lib script itself.","accounts_in_message":[],"_revision_number":2},{"id":"92ab95ee8d9f06c8ea8e330f69a84220384355a9","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-07 18:10:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a164d4ba51169b45407de27fee444e626bb2d5d7","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-07 18:10:49.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"08dd5e6ea4208fdef072581386662b9375a134b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 19:23:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/32/663332/3/check/openstack-tox-cover/a09de16/cover/ : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/663332/3/check/openstack-tox-lower-constraints/94a57a9/ : SUCCESS in 6m 45s\n- openstack-tox-pep8 http://logs.openstack.org/32/663332/3/check/openstack-tox-pep8/d1744f6/ : SUCCESS in 5m 19s\n- openstack-tox-py27 http://logs.openstack.org/32/663332/3/check/openstack-tox-py27/8ac5243/ : SUCCESS in 5m 42s\n- openstack-tox-py36 http://logs.openstack.org/32/663332/3/check/openstack-tox-py36/e93a0e2/ : SUCCESS in 5m 59s\n- openstack-tox-py37 http://logs.openstack.org/32/663332/3/check/openstack-tox-py37/3fe8bc0/ : SUCCESS in 7m 33s\n- openstack-tox-docs http://logs.openstack.org/32/663332/3/check/openstack-tox-docs/d282fcf/html/ : SUCCESS in 6m 38s\n- watcher-tempest-functional http://logs.openstack.org/32/663332/3/check/watcher-tempest-functional/c917f55/ : SUCCESS in 51m 15s\n- watcher-grenade http://logs.openstack.org/32/663332/3/check/watcher-grenade/cc7881f/ : SUCCESS in 46m 29s\n- watcher-tempest-dummy_optim http://logs.openstack.org/32/663332/3/check/watcher-tempest-dummy_optim/1438f3a/ : SUCCESS in 49m 46s\n- watcher-tempest-actuator http://logs.openstack.org/32/663332/3/check/watcher-tempest-actuator/7f05105/ : SUCCESS in 59m 39s\n- watcher-tempest-basic_optim http://logs.openstack.org/32/663332/3/check/watcher-tempest-basic_optim/8587c6f/ : SUCCESS in 50m 56s\n- watcher-tempest-vm_workload_consolidation http://logs.openstack.org/32/663332/3/check/watcher-tempest-vm_workload_consolidation/131adf8/ : SUCCESS in 49m 28s\n- watcher-tempest-workload_balancing http://logs.openstack.org/32/663332/3/check/watcher-tempest-workload_balancing/08162d0/ : SUCCESS in 47m 41s\n- watcherclient-tempest-functional http://logs.openstack.org/32/663332/3/check/watcherclient-tempest-functional/1620500/ : SUCCESS in 33m 29s\n- watcher-tempest-zone_migration http://logs.openstack.org/32/663332/3/check/watcher-tempest-zone_migration/429a8ef/ : SUCCESS in 53m 03s\n- watcher-tempest-host_maintenance http://logs.openstack.org/32/663332/3/check/watcher-tempest-host_maintenance/c6af045/ : SUCCESS in 44m 37s\n- watcher-tempest-storage_balance http://logs.openstack.org/32/663332/3/check/watcher-tempest-storage_balance/43561ad/ : SUCCESS in 50m 55s\n- watcher-tls-test http://logs.openstack.org/32/663332/3/check/watcher-tls-test/beac82a/ : SUCCESS in 54m 25s","accounts_in_message":[],"_revision_number":3},{"id":"a5e410313f71afe741c721be7a3cde104c1c3778","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-08 01:47:34.000000000","message":"Patch Set 3:\n\nNon-grenade job looks OK:\n\nhttp://logs.openstack.org/32/663332/3/check/watcher-tempest-vm_workload_consolidation/131adf8/controller/logs/screen-n-api.txt.gz#_Jun_07_18_51_42_911467\n\nhttp://logs.openstack.org/32/663332/3/check/watcher-tempest-vm_workload_consolidation/131adf8/compute1/logs/screen-n-cpu.txt.gz#_Jun_07_19_04_53_981909\n\nBut the grenade job does not:\n\nhttp://logs.openstack.org/32/663332/3/check/watcher-grenade/cc7881f/logs/screen-n-api.txt.gz#_Jun_07_18_58_06_716807","accounts_in_message":[],"_revision_number":3},{"id":"a95b15df9a08a49d496602cb783a7616d16a11e0","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-11 14:48:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7167bac44f059581225ab3910d2be7c5f3c78526","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-11 14:49:04.000000000","message":"Patch Set 4: Workflow-1\n\nI\u0027m not sure that the grenade part is going to work.","accounts_in_message":[],"_revision_number":4},{"id":"66281cd9d64478f1e13475d62c6ce93386270c59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 16:09:20.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/32/663332/4/check/openstack-tox-cover/e786801/cover/ : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/663332/4/check/openstack-tox-lower-constraints/7158984/ : SUCCESS in 6m 25s\n- openstack-tox-pep8 http://logs.openstack.org/32/663332/4/check/openstack-tox-pep8/c59263e/ : SUCCESS in 5m 48s\n- openstack-tox-py27 http://logs.openstack.org/32/663332/4/check/openstack-tox-py27/db2b22b/ : SUCCESS in 5m 21s\n- openstack-tox-py36 http://logs.openstack.org/32/663332/4/check/openstack-tox-py36/6bedeee/ : SUCCESS in 6m 03s\n- openstack-tox-py37 http://logs.openstack.org/32/663332/4/check/openstack-tox-py37/998adab/ : SUCCESS in 6m 46s\n- openstack-tox-docs http://logs.openstack.org/32/663332/4/check/openstack-tox-docs/d2edc26/html/ : SUCCESS in 4m 43s\n- watcher-tempest-functional http://logs.openstack.org/32/663332/4/check/watcher-tempest-functional/2244d34/ : FAILURE in 37m 32s\n- watcher-grenade http://logs.openstack.org/32/663332/4/check/watcher-grenade/f563468/ : FAILURE in 41m 43s\n- watcher-tempest-dummy_optim http://logs.openstack.org/32/663332/4/check/watcher-tempest-dummy_optim/dd20202/ : SUCCESS in 1h 14m 59s\n- watcher-tempest-actuator http://logs.openstack.org/32/663332/4/check/watcher-tempest-actuator/b258b29/ : SUCCESS in 51m 15s\n- watcher-tempest-basic_optim http://logs.openstack.org/32/663332/4/check/watcher-tempest-basic_optim/ed6235c/ : SUCCESS in 55m 05s\n- watcher-tempest-vm_workload_consolidation http://logs.openstack.org/32/663332/4/check/watcher-tempest-vm_workload_consolidation/c26d2a0/ : SUCCESS in 51m 20s\n- watcher-tempest-workload_balancing http://logs.openstack.org/32/663332/4/check/watcher-tempest-workload_balancing/d0cbb24/ : SUCCESS in 59m 22s\n- watcherclient-tempest-functional http://logs.openstack.org/32/663332/4/check/watcherclient-tempest-functional/55722d5/ : SUCCESS in 35m 12s\n- watcher-tempest-zone_migration http://logs.openstack.org/32/663332/4/check/watcher-tempest-zone_migration/91a7a36/ : SUCCESS in 1h 05m 50s\n- watcher-tempest-host_maintenance http://logs.openstack.org/32/663332/4/check/watcher-tempest-host_maintenance/1b7486a/ : SUCCESS in 49m 55s\n- watcher-tempest-storage_balance http://logs.openstack.org/32/663332/4/check/watcher-tempest-storage_balance/9f8cee1/ : SUCCESS in 51m 04s\n- watcher-tls-test http://logs.openstack.org/32/663332/4/check/watcher-tls-test/41c0afd/ : SUCCESS in 41m 49s","accounts_in_message":[],"_revision_number":4},{"id":"3f53fb0acb7de36005864cd657e2305aa0d29332","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-11 16:18:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a3756faa0e198d765257e615a8d5fb1df883b8aa","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-11 17:59:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f6e6db84d5aad95747cb45ecf94b16ccfd2d9cfb","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-11 17:59:52.000000000","message":"Patch Set 5: Workflow-1\n\nLet\u0027s see if this works.","accounts_in_message":[],"_revision_number":5},{"id":"b2f5afaffc857bb3796caf4aa96a3b3415ae7e87","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-11 18:00:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"30e20f2d1aa161b28e62ff05d53748f355ae49bd","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-11 18:01:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"39b45ad9794bbc8a1b9115a958f26936b78356cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 19:38:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/32/663332/6/check/openstack-tox-cover/3fe5b71/cover/ : SUCCESS in 6m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/663332/6/check/openstack-tox-lower-constraints/8e02095/ : SUCCESS in 6m 22s\n- openstack-tox-pep8 http://logs.openstack.org/32/663332/6/check/openstack-tox-pep8/6392f22/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/32/663332/6/check/openstack-tox-py27/2726cf1/ : SUCCESS in 6m 16s\n- openstack-tox-py36 http://logs.openstack.org/32/663332/6/check/openstack-tox-py36/5a41fdb/ : SUCCESS in 5m 24s\n- openstack-tox-py37 http://logs.openstack.org/32/663332/6/check/openstack-tox-py37/b2f9476/ : SUCCESS in 7m 39s\n- openstack-tox-docs http://logs.openstack.org/32/663332/6/check/openstack-tox-docs/bea7084/html/ : SUCCESS in 4m 53s\n- watcher-tempest-functional http://logs.openstack.org/32/663332/6/check/watcher-tempest-functional/451541e/ : SUCCESS in 34m 38s\n- watcher-grenade http://logs.openstack.org/32/663332/6/check/watcher-grenade/e85ebab/ : SUCCESS in 45m 19s\n- watcher-tempest-dummy_optim http://logs.openstack.org/32/663332/6/check/watcher-tempest-dummy_optim/a607be7/ : SUCCESS in 49m 29s\n- watcher-tempest-actuator http://logs.openstack.org/32/663332/6/check/watcher-tempest-actuator/cdfcda1/ : SUCCESS in 1h 00m 01s\n- watcher-tempest-basic_optim http://logs.openstack.org/32/663332/6/check/watcher-tempest-basic_optim/9fcc84c/ : SUCCESS in 55m 10s\n- watcher-tempest-vm_workload_consolidation http://logs.openstack.org/32/663332/6/check/watcher-tempest-vm_workload_consolidation/474941d/ : SUCCESS in 50m 13s\n- watcher-tempest-workload_balancing http://logs.openstack.org/32/663332/6/check/watcher-tempest-workload_balancing/70622a9/ : SUCCESS in 49m 17s\n- watcherclient-tempest-functional http://logs.openstack.org/32/663332/6/check/watcherclient-tempest-functional/5db8bc5/ : SUCCESS in 36m 53s\n- watcher-tempest-zone_migration http://logs.openstack.org/32/663332/6/check/watcher-tempest-zone_migration/f7e47cd/ : SUCCESS in 46m 58s\n- watcher-tempest-host_maintenance http://logs.openstack.org/32/663332/6/check/watcher-tempest-host_maintenance/e0f39cc/ : SUCCESS in 50m 28s\n- watcher-tempest-storage_balance http://logs.openstack.org/32/663332/6/check/watcher-tempest-storage_balance/973da5c/ : SUCCESS in 57m 02s\n- watcher-tls-test http://logs.openstack.org/32/663332/6/check/watcher-tls-test/aea014f/ : SUCCESS in 49m 38s","accounts_in_message":[],"_revision_number":6},{"id":"6b0871e0d028f9555c937b2f20ea198f88d36904","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-11 19:39:50.000000000","message":"Patch Set 6: Workflow-1\n\nThe grenade part is getting into the local.conf on the new side:\n\nhttp://logs.openstack.org/32/663332/6/check/watcher-grenade/e85ebab/logs/new/local.conf.txt.gz\n\nbut n-api is still showing \"unversioned\" after upgrading nova to train:\n\nhttp://logs.openstack.org/32/663332/6/check/watcher-grenade/e85ebab/logs/screen-n-api.txt.gz#_Jun_11_18_52_04_667925\n\nI wonder if the dependency isn\u0027t being pulled in for devstack on the new side?","accounts_in_message":[],"_revision_number":6},{"id":"4a6bdca7316bc300ca1358829ab15467b347c202","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-11 19:40:18.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d442c1cb20e2b364f68e93c38b0ccc7f2ab04fa5","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-11 19:47:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5d4556e0c50b155a6852d8a1284c3ecaec49581f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-11 19:55:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"619603000dfe5639e50c36027dff74b4d658e17b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-11 19:55:48.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"75097ac34d788c2be7f813dbdda7488af10024e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:20:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/32/663332/7/check/openstack-tox-cover/409fd07/cover/ : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/663332/7/check/openstack-tox-lower-constraints/65e8dd6/ : SUCCESS in 5m 38s\n- openstack-tox-pep8 http://logs.openstack.org/32/663332/7/check/openstack-tox-pep8/8025f22/ : SUCCESS in 8m 49s\n- openstack-tox-py27 http://logs.openstack.org/32/663332/7/check/openstack-tox-py27/925876b/ : SUCCESS in 7m 03s\n- openstack-tox-py36 http://logs.openstack.org/32/663332/7/check/openstack-tox-py36/f8e67e8/ : SUCCESS in 6m 01s\n- openstack-tox-py37 http://logs.openstack.org/32/663332/7/check/openstack-tox-py37/bbc6a73/ : SUCCESS in 8m 42s\n- openstack-tox-docs http://logs.openstack.org/32/663332/7/check/openstack-tox-docs/421866b/html/ : SUCCESS in 5m 33s\n- watcher-tempest-functional http://logs.openstack.org/32/663332/7/check/watcher-tempest-functional/c12254a/ : SUCCESS in 33m 37s\n- watcher-grenade http://logs.openstack.org/32/663332/7/check/watcher-grenade/03dcb3a/ : SUCCESS in 52m 49s\n- watcher-tempest-dummy_optim http://logs.openstack.org/32/663332/7/check/watcher-tempest-dummy_optim/8842795/ : SUCCESS in 49m 49s\n- watcher-tempest-actuator http://logs.openstack.org/32/663332/7/check/watcher-tempest-actuator/0ae6fe3/ : SUCCESS in 53m 47s\n- watcher-tempest-basic_optim http://logs.openstack.org/32/663332/7/check/watcher-tempest-basic_optim/cd76f26/ : SUCCESS in 55m 50s\n- watcher-tempest-vm_workload_consolidation http://logs.openstack.org/32/663332/7/check/watcher-tempest-vm_workload_consolidation/e13b79d/ : SUCCESS in 46m 19s\n- watcher-tempest-workload_balancing http://logs.openstack.org/32/663332/7/check/watcher-tempest-workload_balancing/39adfc3/ : SUCCESS in 54m 17s\n- watcherclient-tempest-functional http://logs.openstack.org/32/663332/7/check/watcherclient-tempest-functional/c6b9dc6/ : SUCCESS in 40m 00s\n- watcher-tempest-zone_migration http://logs.openstack.org/32/663332/7/check/watcher-tempest-zone_migration/07ce06a/ : SUCCESS in 57m 40s\n- watcher-tempest-host_maintenance http://logs.openstack.org/32/663332/7/check/watcher-tempest-host_maintenance/15992ba/ : SUCCESS in 50m 32s\n- watcher-tempest-storage_balance http://logs.openstack.org/32/663332/7/check/watcher-tempest-storage_balance/9e4f3fa/ : SUCCESS in 55m 46s\n- watcher-tls-test http://logs.openstack.org/32/663332/7/check/watcher-tls-test/6a4ef58/ : SUCCESS in 54m 04s","accounts_in_message":[],"_revision_number":7},{"id":"4b83f3df3709ee0b732bd843f54fb579a8356296","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-12 00:24:22.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nStill not working for grenade and I\u0027m not sure why. Even having the option in local.conf doesn\u0027t seem to be getting picked up when grenade runs, but maybe that\u0027s because it\u0027s getting loaded from the watcher grenade plugin *after* nova has already been upgraded?","accounts_in_message":[],"_revision_number":7},{"id":"662b48363715d36e6d4a2238ece2494c3c164c16","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-12 00:26:46.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9345f259592a7578a98a5824bcc526ea3d14fa7b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-12 00:28:08.000000000","message":"Patch Set 8: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"9ecf76927533766fb1e5d84c81c7607036fb71fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 02:05:03.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/32/663332/8/check/openstack-tox-cover/3c1a723/cover/ : SUCCESS in 6m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/663332/8/check/openstack-tox-lower-constraints/6ccda84/ : SUCCESS in 5m 28s\n- openstack-tox-pep8 http://logs.openstack.org/32/663332/8/check/openstack-tox-pep8/570b0b8/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/32/663332/8/check/openstack-tox-py27/6472cf9/ : SUCCESS in 5m 06s\n- openstack-tox-py36 http://logs.openstack.org/32/663332/8/check/openstack-tox-py36/a94675a/ : SUCCESS in 5m 01s\n- openstack-tox-py37 http://logs.openstack.org/32/663332/8/check/openstack-tox-py37/46f61ad/ : SUCCESS in 6m 18s\n- openstack-tox-docs http://logs.openstack.org/32/663332/8/check/openstack-tox-docs/f6a2ba5/html/ : SUCCESS in 5m 01s\n- watcher-tempest-functional http://logs.openstack.org/32/663332/8/check/watcher-tempest-functional/52c93ce/ : SUCCESS in 33m 30s\n- watcher-grenade http://logs.openstack.org/32/663332/8/check/watcher-grenade/32f6696/ : SUCCESS in 45m 08s\n- watcher-tempest-dummy_optim http://logs.openstack.org/32/663332/8/check/watcher-tempest-dummy_optim/e23fab4/ : SUCCESS in 50m 54s\n- watcher-tempest-actuator http://logs.openstack.org/32/663332/8/check/watcher-tempest-actuator/0400aa9/ : SUCCESS in 50m 12s\n- watcher-tempest-basic_optim http://logs.openstack.org/32/663332/8/check/watcher-tempest-basic_optim/e993f55/ : SUCCESS in 49m 03s\n- watcher-tempest-vm_workload_consolidation http://logs.openstack.org/32/663332/8/check/watcher-tempest-vm_workload_consolidation/2cfa9a0/ : SUCCESS in 50m 02s\n- watcher-tempest-workload_balancing http://logs.openstack.org/32/663332/8/check/watcher-tempest-workload_balancing/c9a3e21/ : SUCCESS in 48m 37s\n- watcherclient-tempest-functional http://logs.openstack.org/32/663332/8/check/watcherclient-tempest-functional/d8bf836/ : SUCCESS in 30m 31s\n- watcher-tempest-zone_migration http://logs.openstack.org/32/663332/8/check/watcher-tempest-zone_migration/c1b4cd9/ : SUCCESS in 46m 27s\n- watcher-tempest-host_maintenance http://logs.openstack.org/32/663332/8/check/watcher-tempest-host_maintenance/6cad635/ : SUCCESS in 52m 17s\n- watcher-tempest-storage_balance http://logs.openstack.org/32/663332/8/check/watcher-tempest-storage_balance/48f519c/ : SUCCESS in 51m 14s\n- watcher-tls-test http://logs.openstack.org/32/663332/8/check/watcher-tls-test/aea4457/ : SUCCESS in 50m 34s","accounts_in_message":[],"_revision_number":8},{"id":"2fea1c6ed6a675b0c6c291d778cab9a57654de11","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-12 14:21:07.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"659b8868ba85d1924149a0f270e3dd651460099f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-12 18:48:46.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3b0100929e9460d200d4045048600345fb9bb792","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-12 19:01:30.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f633bdb1f2b968778bb81b594a1b98df5c8e1d15","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-12 19:02:24.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"f5f8b6b26b92cd7d865015339be0208644799aa8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 20:25:04.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/32/663332/9/check/openstack-tox-cover/02f24ce/cover/ : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/663332/9/check/openstack-tox-lower-constraints/4ee4155/ : SUCCESS in 5m 46s\n- openstack-tox-pep8 http://logs.openstack.org/32/663332/9/check/openstack-tox-pep8/743f460/ : SUCCESS in 5m 38s\n- openstack-tox-py27 http://logs.openstack.org/32/663332/9/check/openstack-tox-py27/8d5a7ec/ : SUCCESS in 5m 54s\n- openstack-tox-py36 http://logs.openstack.org/32/663332/9/check/openstack-tox-py36/bdfcd75/ : SUCCESS in 5m 42s\n- openstack-tox-py37 http://logs.openstack.org/32/663332/9/check/openstack-tox-py37/fc6e202/ : SUCCESS in 6m 31s\n- openstack-tox-docs http://logs.openstack.org/32/663332/9/check/openstack-tox-docs/e1a5de8/html/ : SUCCESS in 4m 59s\n- watcher-tempest-functional http://logs.openstack.org/32/663332/9/check/watcher-tempest-functional/3273168/ : SUCCESS in 31m 46s\n- watcher-grenade http://logs.openstack.org/32/663332/9/check/watcher-grenade/4a8893d/ : SUCCESS in 48m 30s\n- watcher-tempest-dummy_optim http://logs.openstack.org/32/663332/9/check/watcher-tempest-dummy_optim/73f3489/ : SUCCESS in 47m 14s\n- watcher-tempest-actuator http://logs.openstack.org/32/663332/9/check/watcher-tempest-actuator/d67d814/ : SUCCESS in 1h 01m 11s\n- watcher-tempest-basic_optim http://logs.openstack.org/32/663332/9/check/watcher-tempest-basic_optim/8813b5c/ : SUCCESS in 50m 46s\n- watcher-tempest-vm_workload_consolidation http://logs.openstack.org/32/663332/9/check/watcher-tempest-vm_workload_consolidation/2c41f53/ : SUCCESS in 52m 43s\n- watcher-tempest-workload_balancing http://logs.openstack.org/32/663332/9/check/watcher-tempest-workload_balancing/7c99de5/ : SUCCESS in 48m 54s\n- watcherclient-tempest-functional http://logs.openstack.org/32/663332/9/check/watcherclient-tempest-functional/b812a1f/ : SUCCESS in 34m 08s\n- watcher-tempest-zone_migration http://logs.openstack.org/32/663332/9/check/watcher-tempest-zone_migration/f8ed30e/ : SUCCESS in 50m 28s\n- watcher-tempest-host_maintenance http://logs.openstack.org/32/663332/9/check/watcher-tempest-host_maintenance/63f99e4/ : SUCCESS in 49m 00s\n- watcher-tempest-storage_balance http://logs.openstack.org/32/663332/9/check/watcher-tempest-storage_balance/0d408a1/ : SUCCESS in 54m 10s\n- watcher-tls-test http://logs.openstack.org/32/663332/9/check/watcher-tls-test/5cb1fef/ : SUCCESS in 54m 30s","accounts_in_message":[],"_revision_number":9},{"id":"39310c7a5ee6ee5081c3e35b810c38e0c816a868","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-13 13:27:51.000000000","message":"Patch Set 9:\n\nStill not working - it seems the override-defaults isn\u0027t getting run on the new side of the grenade run and I\u0027m not sure why.","accounts_in_message":[],"_revision_number":9},{"id":"606e6f499513e07fdaa4532877bcbe462a2f6c48","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-13 13:53:34.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"1b110d564a36fb1ed9f03420b63a70336f6198b7","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-13 14:04:49.000000000","message":"Patch Set 10: Workflow-1\n\nIt seems the settings aren\u0027t getting loaded on the target side because they only get loaded up on the base side when the old devstack runs for stable/stein:\n\nhttps://github.com/openstack/grenade/blob/b646be00f226cc0d7946c5564332b825c322af31/grenade.sh#L216\n\nThe new side doesn\u0027t actually run devstack stack.sh again so it doesn\u0027t load our new settings:\n\nhttps://github.com/openstack/grenade/blob/b646be00f226cc0d7946c5564332b825c322af31/grenade.sh#L325","accounts_in_message":[],"_revision_number":10},{"id":"cf81cc5b9c9de14f0e208837ee9274de90539680","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 15:36:56.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/32/663332/10/check/openstack-tox-cover/aecc5c8/cover/ : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/663332/10/check/openstack-tox-lower-constraints/f6851fb/ : SUCCESS in 7m 09s\n- openstack-tox-pep8 http://logs.openstack.org/32/663332/10/check/openstack-tox-pep8/9c1544b/ : SUCCESS in 5m 07s\n- openstack-tox-py27 http://logs.openstack.org/32/663332/10/check/openstack-tox-py27/b0dd106/ : SUCCESS in 6m 43s\n- openstack-tox-py36 http://logs.openstack.org/32/663332/10/check/openstack-tox-py36/9ba858f/ : SUCCESS in 5m 47s\n- openstack-tox-py37 http://logs.openstack.org/32/663332/10/check/openstack-tox-py37/92a7e14/ : SUCCESS in 5m 49s\n- openstack-tox-docs http://logs.openstack.org/32/663332/10/check/openstack-tox-docs/952ca14/html/ : SUCCESS in 5m 02s\n- watcher-tempest-functional http://logs.openstack.org/32/663332/10/check/watcher-tempest-functional/4887e20/ : SUCCESS in 33m 15s\n- watcher-grenade http://logs.openstack.org/32/663332/10/check/watcher-grenade/e050a96/ : SUCCESS in 47m 06s\n- watcher-tempest-dummy_optim http://logs.openstack.org/32/663332/10/check/watcher-tempest-dummy_optim/d1dc7cc/ : SUCCESS in 52m 24s\n- watcher-tempest-actuator http://logs.openstack.org/32/663332/10/check/watcher-tempest-actuator/9830133/ : SUCCESS in 52m 47s\n- watcher-tempest-basic_optim http://logs.openstack.org/32/663332/10/check/watcher-tempest-basic_optim/7bb3ce2/ : SUCCESS in 49m 08s\n- watcher-tempest-vm_workload_consolidation http://logs.openstack.org/32/663332/10/check/watcher-tempest-vm_workload_consolidation/0709ae5/ : SUCCESS in 48m 26s\n- watcher-tempest-workload_balancing http://logs.openstack.org/32/663332/10/check/watcher-tempest-workload_balancing/3d917bf/ : SUCCESS in 47m 57s\n- watcherclient-tempest-functional http://logs.openstack.org/32/663332/10/check/watcherclient-tempest-functional/7aff7d0/ : SUCCESS in 35m 14s\n- watcher-tempest-zone_migration http://logs.openstack.org/32/663332/10/check/watcher-tempest-zone_migration/d5e8e1a/ : SUCCESS in 59m 49s\n- watcher-tempest-host_maintenance http://logs.openstack.org/32/663332/10/check/watcher-tempest-host_maintenance/57d7fd8/ : SUCCESS in 47m 35s\n- watcher-tempest-storage_balance http://logs.openstack.org/32/663332/10/check/watcher-tempest-storage_balance/a76d718/ : SUCCESS in 47m 32s\n- watcher-tls-test http://logs.openstack.org/32/663332/10/check/watcher-tls-test/ce3db27/ : SUCCESS in 56m 29s","accounts_in_message":[],"_revision_number":10},{"id":"4f2a0dbf17fa833ee59e8e69469b27fabe053451","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-17 16:42:35.000000000","message":"Uploaded patch set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"09d8e20bd47568e969f7fe5098f3329bfbefb75d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-17 16:42:48.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"17046ad8662e9c490c72d8811d0d90be145b624a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 17:52:57.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/32/663332/11/check/openstack-tox-cover/a1c8ed4/cover/ : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/663332/11/check/openstack-tox-lower-constraints/6e2235a/ : SUCCESS in 5m 20s\n- openstack-tox-pep8 http://logs.openstack.org/32/663332/11/check/openstack-tox-pep8/51a821b/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/32/663332/11/check/openstack-tox-py27/3624970/ : SUCCESS in 7m 47s\n- openstack-tox-py36 http://logs.openstack.org/32/663332/11/check/openstack-tox-py36/68076c2/ : SUCCESS in 5m 58s\n- openstack-tox-py37 http://logs.openstack.org/32/663332/11/check/openstack-tox-py37/95bd4b7/ : SUCCESS in 5m 58s\n- openstack-tox-docs http://logs.openstack.org/32/663332/11/check/openstack-tox-docs/6505e89/html/ : SUCCESS in 5m 05s\n- watcher-tempest-functional http://logs.openstack.org/32/663332/11/check/watcher-tempest-functional/2c78ed7/ : SUCCESS in 33m 48s\n- watcher-grenade http://logs.openstack.org/32/663332/11/check/watcher-grenade/ab5f1bc/ : SUCCESS in 46m 18s\n- watcher-tempest-dummy_optim http://logs.openstack.org/32/663332/11/check/watcher-tempest-dummy_optim/fb45bcc/ : SUCCESS in 47m 59s\n- watcher-tempest-actuator http://logs.openstack.org/32/663332/11/check/watcher-tempest-actuator/9e4dbf7/ : SUCCESS in 48m 27s\n- watcher-tempest-basic_optim http://logs.openstack.org/32/663332/11/check/watcher-tempest-basic_optim/2bf3071/ : SUCCESS in 49m 52s\n- watcher-tempest-vm_workload_consolidation http://logs.openstack.org/32/663332/11/check/watcher-tempest-vm_workload_consolidation/3d1db52/ : SUCCESS in 50m 41s\n- watcher-tempest-workload_balancing http://logs.openstack.org/32/663332/11/check/watcher-tempest-workload_balancing/0e7bc9b/ : SUCCESS in 54m 27s\n- watcherclient-tempest-functional http://logs.openstack.org/32/663332/11/check/watcherclient-tempest-functional/94ef4f4/ : SUCCESS in 40m 10s\n- watcher-tempest-zone_migration http://logs.openstack.org/32/663332/11/check/watcher-tempest-zone_migration/9f0520f/ : SUCCESS in 47m 48s\n- watcher-tempest-host_maintenance http://logs.openstack.org/32/663332/11/check/watcher-tempest-host_maintenance/a28a329/ : SUCCESS in 51m 39s\n- watcher-tempest-storage_balance http://logs.openstack.org/32/663332/11/check/watcher-tempest-storage_balance/077bfb3/ : SUCCESS in 49m 00s\n- watcher-tls-test http://logs.openstack.org/32/663332/11/check/watcher-tls-test/130dfdf/ : SUCCESS in 52m 33s","accounts_in_message":[],"_revision_number":11},{"id":"6e2790e1a0985bacd48c9c9550ca04269054fd70","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-17 19:32:27.000000000","message":"Patch Set 11:\n\nThis isn\u0027t working in the grenade job either. Maybe need to move back to devstack/upgrade/settings again.","accounts_in_message":[],"_revision_number":11},{"id":"e6f0789c39063e8c7a399829b26ad01d0a53afe9","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-24 17:10:09.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"0ae306869dcf7d513b636286927230949c7ad0d7","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-24 17:10:24.000000000","message":"Patch Set 12:\n\nThis should be good to go.","accounts_in_message":[],"_revision_number":12},{"id":"592cb1d53f46083d32f9c26ee1b1ebf8f780a0e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 18:53:28.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/32/663332/12/check/openstack-tox-cover/5db2f55/cover/ : SUCCESS in 7m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/663332/12/check/openstack-tox-lower-constraints/3e699b3/ : SUCCESS in 5m 24s\n- openstack-tox-pep8 http://logs.openstack.org/32/663332/12/check/openstack-tox-pep8/3074341/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/32/663332/12/check/openstack-tox-py27/9f9473d/ : SUCCESS in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/32/663332/12/check/openstack-tox-py36/6838dfe/ : SUCCESS in 6m 52s\n- openstack-tox-py37 http://logs.openstack.org/32/663332/12/check/openstack-tox-py37/3b8fa61/ : SUCCESS in 6m 00s\n- openstack-tox-docs http://logs.openstack.org/32/663332/12/check/openstack-tox-docs/39b2797/html/ : SUCCESS in 4m 57s\n- watcher-tempest-functional http://logs.openstack.org/32/663332/12/check/watcher-tempest-functional/6815ce4/ : SUCCESS in 34m 21s\n- watcher-grenade http://logs.openstack.org/32/663332/12/check/watcher-grenade/5ca4699/ : SUCCESS in 48m 31s\n- watcher-tempest-dummy_optim http://logs.openstack.org/32/663332/12/check/watcher-tempest-dummy_optim/6185ed8/ : SUCCESS in 50m 18s\n- watcher-tempest-actuator http://logs.openstack.org/32/663332/12/check/watcher-tempest-actuator/5ac0423/ : SUCCESS in 50m 11s\n- watcher-tempest-basic_optim http://logs.openstack.org/32/663332/12/check/watcher-tempest-basic_optim/12ca883/ : SUCCESS in 49m 22s\n- watcher-tempest-vm_workload_consolidation http://logs.openstack.org/32/663332/12/check/watcher-tempest-vm_workload_consolidation/e9fe7aa/ : SUCCESS in 50m 20s\n- watcher-tempest-workload_balancing http://logs.openstack.org/32/663332/12/check/watcher-tempest-workload_balancing/9db9936/ : SUCCESS in 46m 52s\n- watcherclient-tempest-functional http://logs.openstack.org/32/663332/12/check/watcherclient-tempest-functional/ad44b16/ : SUCCESS in 33m 38s\n- watcher-tempest-zone_migration http://logs.openstack.org/32/663332/12/check/watcher-tempest-zone_migration/879c406/ : SUCCESS in 49m 35s\n- watcher-tempest-host_maintenance http://logs.openstack.org/32/663332/12/check/watcher-tempest-host_maintenance/2661863/ : SUCCESS in 50m 24s\n- watcher-tempest-storage_balance http://logs.openstack.org/32/663332/12/check/watcher-tempest-storage_balance/aed2171/ : SUCCESS in 55m 00s\n- watcher-tls-test http://logs.openstack.org/32/663332/12/check/watcher-tls-test/ec54be1/ : SUCCESS in 51m 02s","accounts_in_message":[],"_revision_number":12},{"id":"246ccf7d8eb7bd9d670b71726a455d4c98b1ad16","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-24 19:18:17.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"4141adf7de451499606730ab7418bb4174e21515","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-25 15:00:03.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"259853351ae3db0d79f1950582df135a2d1d5ae4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-25 15:18:43.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"99b7c6f1e19f3b564210fcabce56f890499c6f37","author":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"date":"2019-06-26 09:07:33.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"026447e1482105d1f1fbd14a6dff7dc6d9cab45a","author":{"_account_id":21692,"name":"licanwei","email":"li.canwei2@zte.com.cn","username":"licanwei"},"date":"2019-06-28 01:48:49.000000000","message":"Patch Set 12: Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"2d5baed0334a67ee6a1a1acd162cf513c60d61ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 01:49:04.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"2e2f3ba45fc874de5e6f5a5e6601718fbc86f6f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 02:29:14.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/32/663332/12/gate/openstack-tox-lower-constraints/bec3138/ : SUCCESS in 4m 21s\n- openstack-tox-pep8 http://logs.openstack.org/32/663332/12/gate/openstack-tox-pep8/ec5dd99/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/32/663332/12/gate/openstack-tox-py27/7a7a292/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/32/663332/12/gate/openstack-tox-py36/be0d2f6/ : SUCCESS in 4m 54s\n- openstack-tox-py37 http://logs.openstack.org/32/663332/12/gate/openstack-tox-py37/ceaf807/ : SUCCESS in 5m 19s\n- openstack-tox-docs http://logs.openstack.org/32/663332/12/gate/openstack-tox-docs/2880821/html/ : SUCCESS in 3m 10s\n- watcher-tempest-functional http://logs.openstack.org/32/663332/12/gate/watcher-tempest-functional/7a3bf77/ : SUCCESS in 33m 07s","accounts_in_message":[],"_revision_number":12},{"id":"f001c73079d5fc1c9a0f987d47f9586fa7359f93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 02:29:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"966a4dfa5f030d4b88e4902c9f762a2d6213a670","revisions":{"487643d0bca7d35fa557013f9560edd9d1798e0b":{"kind":"REWORK","_number":1,"created":"2019-06-05 13:34:38.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/32/663332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/32/663332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/32/663332/1"}}},"commit":{"parents":[{"commit":"7f37f7b92a096405ce24485589ae450baa654e2d","subject":"Remove apidoc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/7f37f7b92a096405ce24485589ae450baa654e2d"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-05 13:33:11.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-05 13:33:11.000000000","tz":-240},"subject":"Configure nova notification format","message":"Configure nova notification format\n\nNova used to emit versioned and unversioned notiifcations\nby default but that changed in https://review.opendev.org/603079/\nso now nova emits only unversioned notifications by default.\nWatcher listens for versioned notifications so we need to configure\nnova in the devstack plugin to emit both versioned (for Watcher)\nand unversioned (for Ceilometer) notifications explicitly.\n\nChange-Id: I7e441608b38338eecd80e663ed3abe66a89e504f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/487643d0bca7d35fa557013f9560edd9d1798e0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/487643d0bca7d35fa557013f9560edd9d1798e0b"}]},"branch":"refs/heads/master"},"1c29d598f59c3116e74ac738a8a6b0e68f03b253":{"kind":"REWORK","_number":2,"created":"2019-06-06 15:55:00.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/32/663332/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/32/663332/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/32/663332/2"}}},"commit":{"parents":[{"commit":"7f37f7b92a096405ce24485589ae450baa654e2d","subject":"Remove apidoc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/7f37f7b92a096405ce24485589ae450baa654e2d"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-05 13:33:11.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-06 15:52:18.000000000","tz":-240},"subject":"Configure nova notification format","message":"Configure nova notification format\n\nNova used to emit versioned and unversioned notiifcations\nby default but that changed in https://review.opendev.org/603079/\nso now nova emits only unversioned notifications by default.\nWatcher listens for versioned notifications so we need to configure\nnova to emit both versioned (for Watcher) and unversioned\n(for Ceilometer) notifications explicitly.\n\nThe local.conf.* samples are just referenced from the contributor\ndocs but will not fix the devstack-based CI jobs. For that the\nwatcher-tempest-functional job definition is updated to configure\nnova.\n\nChange-Id: I7e441608b38338eecd80e663ed3abe66a89e504f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/1c29d598f59c3116e74ac738a8a6b0e68f03b253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/1c29d598f59c3116e74ac738a8a6b0e68f03b253"}]},"branch":"refs/heads/master"},"72f9286ccd8a030fdd8b091b3106e7f3eab6cdb4":{"kind":"REWORK","_number":3,"created":"2019-06-07 18:10:36.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/32/663332/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/32/663332/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/32/663332/3"}}},"commit":{"parents":[{"commit":"8000dd650f332d8425e976f65a4e6fa3fb6d7530","subject":"Merge \"add strategy tempest job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/8000dd650f332d8425e976f65a4e6fa3fb6d7530"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-05 13:33:11.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-07 18:09:01.000000000","tz":-240},"subject":"Configure nova notification format","message":"Configure nova notification format\n\nNova used to emit versioned and unversioned notiifcations\nby default but that changed in https://review.opendev.org/603079/\nso now nova emits only unversioned notifications by default.\nWatcher listens for versioned notifications so we need to configure\nnova to emit both versioned (for Watcher) and unversioned\n(for Ceilometer) notifications explicitly.\n\nCloses-Bug: #1831917\n\nChange-Id: I7e441608b38338eecd80e663ed3abe66a89e504f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/72f9286ccd8a030fdd8b091b3106e7f3eab6cdb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/72f9286ccd8a030fdd8b091b3106e7f3eab6cdb4"}]},"branch":"refs/heads/master"},"f9ac25a354d20ecb31bf0ed720c272397a2b21f3":{"kind":"REWORK","_number":4,"created":"2019-06-11 14:48:30.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/32/663332/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/32/663332/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/32/663332/4"}}},"commit":{"parents":[{"commit":"5cb08dd17f18a8a9f9c30705a6da98bfbd120cc7","subject":"Fix base enable_plugin branch for grenade run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/5cb08dd17f18a8a9f9c30705a6da98bfbd120cc7"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-05 13:33:11.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-11 14:48:24.000000000","tz":-240},"subject":"Configure nova notification format in CI jobs","message":"Configure nova notification format in CI jobs\n\nNova used to emit versioned and unversioned notiifcations\nby default but that changed in https://review.opendev.org/603079/\nso now nova emits only unversioned notifications by default.\nWatcher listens for versioned notifications so we need to configure\nnova to emit both versioned (for Watcher) and unversioned\n(for Ceilometer) notifications explicitly.\n\nFor the grenade run we have to hack the settings file for the\n\"new\" devstack local.conf (the master version) which we can remove\nin the U release when the \"old\" side is using stable/stein and is\ndeployed from our devstack/lib/watcher script.\n\nCloses-Bug: #1831917\n\nChange-Id: I7e441608b38338eecd80e663ed3abe66a89e504f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/f9ac25a354d20ecb31bf0ed720c272397a2b21f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/f9ac25a354d20ecb31bf0ed720c272397a2b21f3"}]},"branch":"refs/heads/master"},"1eba61638a34c47d644721e15ac09445038336c7":{"kind":"REWORK","_number":5,"created":"2019-06-11 17:59:21.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/32/663332/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/32/663332/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/32/663332/5"}}},"commit":{"parents":[{"commit":"5cb08dd17f18a8a9f9c30705a6da98bfbd120cc7","subject":"Fix base enable_plugin branch for grenade run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/5cb08dd17f18a8a9f9c30705a6da98bfbd120cc7"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-05 13:33:11.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-11 17:57:40.000000000","tz":-240},"subject":"Configure nova notification format in CI jobs","message":"Configure nova notification format in CI jobs\n\nNova used to emit versioned and unversioned notiifcations\nby default but that changed in https://review.opendev.org/603079/\nso now nova emits only unversioned notifications by default.\nWatcher listens for versioned notifications so we need to configure\nnova to emit both versioned (for Watcher) and unversioned\n(for Ceilometer) notifications explicitly.\n\nSince the non-grenade job definitions are zuul v3 native we can\nconfigure nova in the parent job. The grenade job is still a legacy\ndefinition using devstack-gate so we have to pass a variable through\nthat way.\n\nCloses-Bug: #1831917\n\nDepends-On: https://review.opendev.org/662849\n\nChange-Id: I7e441608b38338eecd80e663ed3abe66a89e504f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/1eba61638a34c47d644721e15ac09445038336c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/1eba61638a34c47d644721e15ac09445038336c7"}]},"branch":"refs/heads/master"},"537d23910ccb2d2121f5c174233851f0fa682005":{"kind":"REWORK","_number":6,"created":"2019-06-11 18:01:53.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/32/663332/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/32/663332/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/32/663332/6"}}},"commit":{"parents":[{"commit":"5cb08dd17f18a8a9f9c30705a6da98bfbd120cc7","subject":"Fix base enable_plugin branch for grenade run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/5cb08dd17f18a8a9f9c30705a6da98bfbd120cc7"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-05 13:33:11.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-11 18:01:47.000000000","tz":-240},"subject":"Configure nova notification format in CI jobs","message":"Configure nova notification format in CI jobs\n\nNova used to emit versioned and unversioned notiifcations\nby default but that changed in https://review.opendev.org/603079/\nso now nova emits only unversioned notifications by default.\nWatcher listens for versioned notifications so we need to configure\nnova to emit both versioned (for Watcher) and unversioned\n(for Ceilometer) notifications explicitly.\n\nSince the non-grenade job definitions are zuul v3 native we can\nconfigure nova in the parent job. The grenade job is still a legacy\ndefinition using devstack-gate so we have to pass a variable through\nthat way.\n\nCloses-Bug: #1831917\n\nDepends-On: https://review.opendev.org/662849\n\nChange-Id: I7e441608b38338eecd80e663ed3abe66a89e504f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/537d23910ccb2d2121f5c174233851f0fa682005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/537d23910ccb2d2121f5c174233851f0fa682005"}]},"branch":"refs/heads/master"},"e1b7e0a30a2d6af976720676dd8d0fd75f92eb56":{"kind":"REWORK","_number":7,"created":"2019-06-11 19:55:28.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/32/663332/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/32/663332/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/32/663332/7"}}},"commit":{"parents":[{"commit":"5cb08dd17f18a8a9f9c30705a6da98bfbd120cc7","subject":"Fix base enable_plugin branch for grenade run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/5cb08dd17f18a8a9f9c30705a6da98bfbd120cc7"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-05 13:33:11.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-11 19:55:11.000000000","tz":-240},"subject":"Configure nova notification format in CI jobs","message":"Configure nova notification format in CI jobs\n\nNova used to emit versioned and unversioned notiifcations\nby default but that changed in https://review.opendev.org/603079/\nso now nova emits only unversioned notifications by default.\nWatcher listens for versioned notifications so we need to configure\nnova to emit both versioned (for Watcher) and unversioned\n(for Ceilometer) notifications explicitly.\n\nSince the non-grenade job definitions are zuul v3 native we can\nconfigure nova in the parent job. The grenade job is still a legacy\ndefinition using devstack-gate so we have to pass a variable through\nthat way.\n\nCloses-Bug: #1831917\n\nDepends-On: https://review.opendev.org/662849\n\nChange-Id: I7e441608b38338eecd80e663ed3abe66a89e504f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/e1b7e0a30a2d6af976720676dd8d0fd75f92eb56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/e1b7e0a30a2d6af976720676dd8d0fd75f92eb56"}]},"branch":"refs/heads/master"},"6cde6cb15855919c927bbc55b5d9ddd72d33c052":{"kind":"REWORK","_number":8,"created":"2019-06-12 00:26:46.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/32/663332/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/32/663332/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/32/663332/8"}}},"commit":{"parents":[{"commit":"5cb08dd17f18a8a9f9c30705a6da98bfbd120cc7","subject":"Fix base enable_plugin branch for grenade run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/5cb08dd17f18a8a9f9c30705a6da98bfbd120cc7"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-05 13:33:11.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-12 00:26:36.000000000","tz":-240},"subject":"WIP: Configure nova notification format in CI jobs","message":"WIP: Configure nova notification format in CI jobs\n\nNova used to emit versioned and unversioned notiifcations\nby default but that changed in https://review.opendev.org/603079/\nso now nova emits only unversioned notifications by default.\nWatcher listens for versioned notifications so we need to configure\nnova to emit both versioned (for Watcher) and unversioned\n(for Ceilometer) notifications explicitly.\n\nSince the non-grenade job definitions are zuul v3 native we can\nconfigure nova in the parent job. The grenade job is still a legacy\ndefinition using devstack-gate so we have to pass a variable through\nthat way.\n\nCloses-Bug: #1831917\n\nDepends-On: https://review.opendev.org/662849\n\nChange-Id: I7e441608b38338eecd80e663ed3abe66a89e504f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/6cde6cb15855919c927bbc55b5d9ddd72d33c052"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/6cde6cb15855919c927bbc55b5d9ddd72d33c052"}]},"branch":"refs/heads/master"},"2e28a9cd07f575c7d2f1d4f166e4aad5431a1126":{"kind":"REWORK","_number":9,"created":"2019-06-12 19:01:30.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/32/663332/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/32/663332/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/32/663332/9"}}},"commit":{"parents":[{"commit":"e2b1547d3c1bb25a2f47b8acddbd1b259144ae00","subject":"Fix base enable_plugin branch for grenade run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/e2b1547d3c1bb25a2f47b8acddbd1b259144ae00"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-05 13:33:11.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-12 19:00:25.000000000","tz":-240},"subject":"WIP: Configure nova notification format in CI jobs","message":"WIP: Configure nova notification format in CI jobs\n\nNova used to emit versioned and unversioned notiifcations\nby default but that changed in https://review.opendev.org/603079/\nso now nova emits only unversioned notifications by default.\nWatcher listens for versioned notifications so we need to configure\nnova to emit both versioned (for Watcher) and unversioned\n(for Ceilometer) notifications explicitly.\n\nThis adds an override-defaults file so devstack will load up\nthe nova devstack variable to set the notification_format before\nimporting and stacking the nova lib script.\n\nCloses-Bug: #1831917\n\nDepends-On: https://review.opendev.org/662849\n\nChange-Id: I7e441608b38338eecd80e663ed3abe66a89e504f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/2e28a9cd07f575c7d2f1d4f166e4aad5431a1126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/2e28a9cd07f575c7d2f1d4f166e4aad5431a1126"}]},"branch":"refs/heads/master"},"2e39363fb557710e315c5008283995276d4ebcd9":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-06-13 13:53:34.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/32/663332/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/32/663332/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/32/663332/10"}}},"commit":{"parents":[{"commit":"28df60e275d44c6e07e963b7ed139093e79a0045","subject":"Fix base enable_plugin branch for grenade run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/28df60e275d44c6e07e963b7ed139093e79a0045"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-05 13:33:11.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-13 13:53:23.000000000","tz":-240},"subject":"WIP: Configure nova notification format in CI jobs","message":"WIP: Configure nova notification format in CI jobs\n\nNova used to emit versioned and unversioned notiifcations\nby default but that changed in https://review.opendev.org/603079/\nso now nova emits only unversioned notifications by default.\nWatcher listens for versioned notifications so we need to configure\nnova to emit both versioned (for Watcher) and unversioned\n(for Ceilometer) notifications explicitly.\n\nThis adds an override-defaults file so devstack will load up\nthe nova devstack variable to set the notification_format before\nimporting and stacking the nova lib script.\n\nCloses-Bug: #1831917\n\nDepends-On: https://review.opendev.org/662849\n\nChange-Id: I7e441608b38338eecd80e663ed3abe66a89e504f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/2e39363fb557710e315c5008283995276d4ebcd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/2e39363fb557710e315c5008283995276d4ebcd9"}]},"branch":"refs/heads/master"},"c10eedfc241a7e2be1b67623627b3866684d659e":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2019-06-17 16:42:35.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/32/663332/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/32/663332/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/32/663332/11"}}},"commit":{"parents":[{"commit":"28df60e275d44c6e07e963b7ed139093e79a0045","subject":"Fix base enable_plugin branch for grenade run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/28df60e275d44c6e07e963b7ed139093e79a0045"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-05 13:33:11.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-17 16:41:23.000000000","tz":-240},"subject":"Configure nova notification format in CI jobs","message":"Configure nova notification format in CI jobs\n\nNova used to emit versioned and unversioned notiifcations\nby default but that changed in https://review.opendev.org/603079/\nso now nova emits only unversioned notifications by default.\nWatcher listens for versioned notifications so we need to configure\nnova to emit both versioned (for Watcher) and unversioned\n(for Ceilometer) notifications explicitly.\n\nThis adds an override-defaults file so devstack will load up\nthe nova devstack variable to set the notification_format before\nimporting and stacking the nova lib script.\n\nNote that grenade only runs stack.sh on the base (old) side\nso this change has to depend on a devstack stable/stein change\nto add the NOVA_NOTIFICATION_FORMAT variable that we override.\n\nCloses-Bug: #1831917\n\nDepends-On: Ied9d50b07c368d5c2be658c744f340a8d1ee41e0\n\nChange-Id: I7e441608b38338eecd80e663ed3abe66a89e504f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/c10eedfc241a7e2be1b67623627b3866684d659e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/c10eedfc241a7e2be1b67623627b3866684d659e"}]},"branch":"refs/heads/master"},"966a4dfa5f030d4b88e4902c9f762a2d6213a670":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":12,"created":"2019-06-24 17:10:09.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/32/663332/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/32/663332/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/32/663332/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/32/663332/12"}}},"commit":{"parents":[{"commit":"899e53476148bfb2bacfe2af91e667c7d1b2522e","subject":"Merge \"check instance state for instance.update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/899e53476148bfb2bacfe2af91e667c7d1b2522e"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-05 13:33:11.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-06-24 17:09:34.000000000","tz":-240},"subject":"Configure nova notification format in non-grenade CI jobs","message":"Configure nova notification format in non-grenade CI jobs\n\nNova used to emit versioned and unversioned notiifcations\nby default but that changed in https://review.opendev.org/603079/\nso now nova emits only unversioned notifications by default.\nWatcher listens for versioned notifications so we need to configure\nnova to emit both versioned (for Watcher) and unversioned\n(for Ceilometer) notifications explicitly.\n\nThis adds an override-defaults file so devstack will load up\nthe nova devstack variable to set the notification_format before\nimporting and stacking the nova lib script.\n\nNote that this only fixes the non-grenade CI jobs since grenade\nrequires separate handling for overriding defaults which is proving\nhard to do and will be addressed in a separate change.\n\nPartial-Bug: #1831917\n\nChange-Id: I7e441608b38338eecd80e663ed3abe66a89e504f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/966a4dfa5f030d4b88e4902c9f762a2d6213a670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/966a4dfa5f030d4b88e4902c9f762a2d6213a670"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
