)]}'
{"id":"openstack%2Fwatcher~900872","triplet_id":"openstack%2Fwatcher~master~Ifeb38b98f1d702908531de5fc5c846bd1c53de4b","project":"openstack/watcher","branch":"master","attention_set":{},"removed_from_attention_set":{"28748":{"account":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"last_update":"2023-11-14 10:58:47.000000000","reason":"\u003cGERRIT_ACCOUNT_28748\u003e replied on the change","reason_account":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}},"8543":{"account":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"last_update":"2023-11-14 12:41:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifeb38b98f1d702908531de5fc5c846bd1c53de4b","subject":"Fix object tests","status":"MERGED","created":"2023-11-14 10:41:55.000000000","updated":"2023-11-14 12:42:37.000000000","submitted":"2023-11-14 12:41:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900872","meta_rev_id":"1e04d0883d71df2d446095ecdd80ef6d900b8a8a","_number":900872,"virtual_id_number":900872,"owner":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-14 12:41:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"all":[{"value":2,"date":"2023-11-14 12:00:58.000000000","_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"all":[{"value":1,"date":"2023-11-14 12:00:58.000000000","_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Standard Change","+1":"Important Change","+2":"High Priority Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-14 10:44:48.000000000","updated_by":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2023-11-14 11:51:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35f6d4fc4581bce254cab14e02b75c602b1592e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2023-11-14 10:41:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a733221cfb83253f9a325f14adbf7948502a9df","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2023-11-14 10:58:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c980d8bd99ec618417ef9db6f0275979c66af625","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 11:51:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c008ecaa26a4a8388e169a118e0c21a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8db27fda889142738b041fc1162ae472 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b76c0c167d444d8d8f296b7655997337 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/429d6e2f5cc5491fa8876d360dcab198 : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a8fea1b4fa1484492997281f02e94b0 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f3c3c383d464c8cadc5e9618af93197 : SUCCESS in 7m 13s\n- watcher-tempest-functional https://zuul.opendev.org/t/openstack/build/2f89307098724564a1f72c66a629bba8 : SUCCESS in 38m 18s\n- watcher-tempest-strategies https://zuul.opendev.org/t/openstack/build/34e9af5bcde2431cbc69763fa49c3ec4 : SUCCESS in 1h 01m 17s\n- watcher-tempest-actuator https://zuul.opendev.org/t/openstack/build/5c47c8c576764d4e8c981c81ba2bf412 : SUCCESS in 33m 28s\n- watcherclient-tempest-functional https://zuul.opendev.org/t/openstack/build/e432514362dd499689294d348295f694 : SUCCESS in 27m 14s\n- watcher-tempest-functional-ipv6-only https://zuul.opendev.org/t/openstack/build/3a755570aa92487581b497ecfd7c6790 : SUCCESS in 26m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e8ea598370384b92a8a098134f443841aa062b7b","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2023-11-14 12:00:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5427ec6db59f2fa5b8dbfbea5b5b7542c472a9ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 12:01:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70cea1d99910d7e757c0b7147b1ef043963cb430","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 12:41:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5400df1c03354f8c910110ce95508df1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/469aaee3e29d4d37bf6ac63f2f8907d3 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc44088b0de34c99830ad6b4cab4d5e3 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5cba769e7d604db7afa78fbd764eb75b : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74170217f2d84616bd40037873c1aca8 : SUCCESS in 7m 17s\n- watcher-tempest-functional https://zuul.opendev.org/t/openstack/build/e3af14fcc683429089b165a21ad0116a : SUCCESS in 35m 11s\n- watcher-tempest-functional-ipv6-only https://zuul.opendev.org/t/openstack/build/b33b20ab805746c6963ca5b86f4c8fd0 : SUCCESS in 26m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0b2f71ae42e3874151bd95cbeb6af1b2d7959764","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 12:41:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1e04d0883d71df2d446095ecdd80ef6d900b8a8a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 12:42:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6ac7a541e0349a4b777b7867a22ddf7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7078e9a8deb44945b5590098a38e844c : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b224b562972f4ab4f127aafec80fb76b7952eaf","revisions":{"3b224b562972f4ab4f127aafec80fb76b7952eaf":{"kind":"REWORK","_number":1,"created":"2023-11-14 10:41:55.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/72/900872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/watcher","ref":"refs/changes/72/900872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/watcher refs/changes/72/900872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/watcher refs/changes/72/900872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/watcher refs/changes/72/900872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/watcher refs/changes/72/900872/1"}}},"commit":{"parents":[{"commit":"40e93407c77c227b5c9159462d75b7015116aa9f","subject":"Merge \"Handle deprecated \"cpu_util\" metric\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/40e93407c77c227b5c9159462d75b7015116aa9f"}]}],"author":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2023-11-14 10:38:40.000000000","tz":0},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2023-11-14 10:38:40.000000000","tz":0},"subject":"Fix object tests","message":"Fix object tests\n\nA couple of object tests are failing, probably after a dependency\nbump.\n\nwatcher.objects.base.objects is mocked, so the registered object\nversion isn\u0027t properly retrieved, leading to a type error:\n\n    File \"/mnt/data/workspace/watcher/watcher/tests/objects/test_objects.py\",\n    line 535, in test_hook_chooses_newer_properly\n      reg.registration_hook(MyObj, 0)\n    File \"/mnt/data/workspace/watcher/watcher/objects/base.py\",\n    line 46, in registration_hook\n      cur_version \u003d versionutils.convert_version_to_tuple(\n    File \"/home/ubuntu/openstack_venv/lib/python3.10/site-packages/oslo_utils/versionutils.py\",\n    line 91, in convert_version_to_tuple\n      version_str \u003d re.sub(r\u0027(\\d+)(a|alpha|b|beta|rc)\\d+$\u0027, \u0027\\\\1\u0027, version_str)\n    File \"/usr/lib/python3.10/re.py\", line 209, in sub\n      return _compile(pattern, flags).sub(repl, string, count)\n  TypeError: expected string or bytes-like object\n\nWe\u0027ll solve the issue by setting the VERSION attribute against\nthe mock object.\n\nChange-Id: Ifeb38b98f1d702908531de5fc5c846bd1c53de4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/3b224b562972f4ab4f127aafec80fb76b7952eaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/watcher/commit/3b224b562972f4ab4f127aafec80fb76b7952eaf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
